)]}'
{"id":"openstack%2Fdesignate~870680","triplet_id":"openstack%2Fdesignate~master~I7a847bef217140c3c1aa84bdd7b33a51884bad02","project":"openstack/designate","branch":"master","topic":"DNM-dnspython-2.3.0","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-01-18 19:07:03.000000000","reason":"Change was abandoned"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-01-17 19:10:30.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I7a847bef217140c3c1aa84bdd7b33a51884bad02","subject":"DNM: Testing dnspython 2.3.0 with tempest","status":"ABANDONED","created":"2023-01-16 23:11:46.000000000","updated":"2023-01-18 19:07:03.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ed403ba6afc3f0851ef10d7d7f20e9998774b618","_number":870680,"virtual_id_number":870680,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-01-18 18:07:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-16 23:39:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-17 05:55:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"eb214ab22316d90f823981ba436721dbeea35646","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-16 23:11:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d585385fa47e7bceec971c0d7abad17992fe8c9c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-16 23:11:54.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b48d6202a9b9703442da1915641732b950a1577e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 23:39:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c577eb744ff464c98167bfbda3bf5a5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4016e838d020452c91151179b14fcdf5 : RETRY_LIMIT in 4m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7b9a5f129b6a4601a53ed0dd13b53b7b : SUCCESS in 4m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05dbb036d1ce526ea0177cc3d31f4886991f830e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-16 23:54:06.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"53a1f24f9c79668ca8c5b33634ffca5938bc46e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 00:24:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/928b5ea3c1fe4eedbbee5c0c3b8bca80\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e051e2264f454d0db30e6d195c75a3f2 : RETRY_LIMIT in 4m 27s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0e295ea6df10483fb34becdeb55c6b11 : SUCCESS in 3m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"195857016afb8d8bc37c2941c180194cb1bb5803","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-17 01:11:42.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a8c29e316610c5e9bfc9191618a1c281697f5cb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 01:59:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d06c9ced5afb432ea47805f25063b912\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f4a0992524dd4a10978d96cbfe7d6a6a : RETRY_LIMIT in 20m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5039366595974315bd7e32f552bca608 : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"708fbe7ec3446f135d0f7083c6ceb151fe9c1dc2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-01-17 05:55:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"14c573636d52d7019ca30cc44a46c0cacbf0eade","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-17 17:48:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0557ba3dc92b9fff4329cb6db24ee9d758e2c6aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-17 18:10:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e0ee53fa66d2f94ad2a5e69c49bc7cec6af9fd6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 18:10:48.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/designate for 870680,4","accounts_in_message":[],"_revision_number":4},{"id":"a219de3d080c6199badd8d4329db074763054bce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-17 18:12:06.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"4608fe09330764adcba0f329a7b9de5c00ea3f48","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-17 18:20:01.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"b4ffd452422994a89c51f8fc96add72b37190284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 18:37:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5f37e2c274d4437b1999364eb6c1093\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6195525e61fd42eca27b5a0f8bd05bc1 : FAILURE in 23m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/759dcb1d640147da8abc0514bcb98943 : SUCCESS in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cf5da5d8dbedbdac16f549b6bfda14cf1cd6bc53","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-17 19:10:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d0ee3efe7d314a3b506073393fe1e82d65b2fa2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-18 17:26:45.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"333515303e90af0a1fa3c25438da44e0fc74a812","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 18:07:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bed157fc7e2e4298af0b8ff3c9d865ed\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2c8723173e11463a9c2fe54deda8058d : SUCCESS in 38m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/87c450fd7bc84f00b592b91d89d9ac52 : SUCCESS in 5m 34s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ed403ba6afc3f0851ef10d7d7f20e9998774b618","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-18 19:07:03.000000000","message":"Abandoned\n\nThis test patch has passed, so no longer needed.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7bd20a51ebfdb5f874373d9fb020ba704557ae23","revisions":{"b9d8235a092064eafa799f525ed63e4685e06ce2":{"kind":"REWORK","_number":1,"created":"2023-01-16 23:11:46.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/80/870680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/80/870680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/80/870680/1"}}},"commit":{"parents":[{"commit":"ada1455232da6d813afb3742671ea387d99db304","subject":"Workaround issue with dnspython \u003e\u003d 2.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ada1455232da6d813afb3742671ea387d99db304"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-01-16 23:10:54.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-01-16 23:11:37.000000000","tz":0},"subject":"DNM: Testing dnspython 2.3.0 with tempest","message":"DNM: Testing dnspython 2.3.0 with tempest\n\nChange-Id: I7a847bef217140c3c1aa84bdd7b33a51884bad02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9d8235a092064eafa799f525ed63e4685e06ce2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9d8235a092064eafa799f525ed63e4685e06ce2"}]},"branch":"refs/heads/master"},"cf80e3cbc23089e6fe366f862c4f82abf388b078":{"kind":"REWORK","_number":2,"created":"2023-01-16 23:54:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/80/870680/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/80/870680/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/80/870680/2"}}},"commit":{"parents":[{"commit":"ada1455232da6d813afb3742671ea387d99db304","subject":"Workaround issue with dnspython \u003e\u003d 2.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ada1455232da6d813afb3742671ea387d99db304"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-01-16 23:10:54.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-01-16 23:54:06.000000000","tz":0},"subject":"DNM: Testing dnspython 2.3.0 with tempest","message":"DNM: Testing dnspython 2.3.0 with tempest\n\nChange-Id: I7a847bef217140c3c1aa84bdd7b33a51884bad02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cf80e3cbc23089e6fe366f862c4f82abf388b078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cf80e3cbc23089e6fe366f862c4f82abf388b078"}]},"branch":"refs/heads/master"},"acfacd3cce1d22a1adff12f7b1ba5133bb5a1474":{"kind":"REWORK","_number":3,"created":"2023-01-17 01:11:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/80/870680/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/80/870680/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/80/870680/3"}}},"commit":{"parents":[{"commit":"ada1455232da6d813afb3742671ea387d99db304","subject":"Workaround issue with dnspython \u003e\u003d 2.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ada1455232da6d813afb3742671ea387d99db304"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-01-16 23:10:54.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-01-17 01:11:42.000000000","tz":0},"subject":"DNM: Testing dnspython 2.3.0 with tempest","message":"DNM: Testing dnspython 2.3.0 with tempest\n\nChange-Id: I7a847bef217140c3c1aa84bdd7b33a51884bad02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/acfacd3cce1d22a1adff12f7b1ba5133bb5a1474"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/acfacd3cce1d22a1adff12f7b1ba5133bb5a1474"}]},"branch":"refs/heads/master"},"d5012cc597182ed66d744f833cfb80518e1de00a":{"kind":"REWORK","_number":4,"created":"2023-01-17 18:10:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/80/870680/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/80/870680/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/80/870680/4"}}},"commit":{"parents":[{"commit":"ada1455232da6d813afb3742671ea387d99db304","subject":"Workaround issue with dnspython \u003e\u003d 2.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ada1455232da6d813afb3742671ea387d99db304"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-01-16 23:10:54.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-01-17 18:10:06.000000000","tz":0},"subject":"DNM: Testing dnspython 2.3.0 with tempest","message":"DNM: Testing dnspython 2.3.0 with tempest\n\nChange-Id: I7a847bef217140c3c1aa84bdd7b33a51884bad02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d5012cc597182ed66d744f833cfb80518e1de00a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d5012cc597182ed66d744f833cfb80518e1de00a"}]},"branch":"refs/heads/master"},"cd35a7baa73b13d37cbfc02c194278075aa92aa9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-01-17 18:12:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/80/870680/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/80/870680/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/80/870680/5"}}},"commit":{"parents":[{"commit":"41bf7dd319966e3fa0182b88b0869ab13b52092e","subject":"Workaround issue with dnspython \u003e\u003d 2.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/41bf7dd319966e3fa0182b88b0869ab13b52092e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-01-16 23:10:54.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-01-17 18:12:06.000000000","tz":0},"subject":"DNM: Testing dnspython 2.3.0 with tempest","message":"DNM: Testing dnspython 2.3.0 with tempest\n\nChange-Id: I7a847bef217140c3c1aa84bdd7b33a51884bad02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cd35a7baa73b13d37cbfc02c194278075aa92aa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cd35a7baa73b13d37cbfc02c194278075aa92aa9"}]},"branch":"refs/heads/master","description":"Rebase"},"7bd20a51ebfdb5f874373d9fb020ba704557ae23":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-01-18 17:26:45.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/80/870680/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/80/870680/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/870680/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/80/870680/6"}}},"commit":{"parents":[{"commit":"41bf7dd319966e3fa0182b88b0869ab13b52092e","subject":"Workaround issue with dnspython \u003e\u003d 2.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/41bf7dd319966e3fa0182b88b0869ab13b52092e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-01-16 23:10:54.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-01-18 17:26:45.000000000","tz":0},"subject":"DNM: Testing dnspython 2.3.0 with tempest","message":"DNM: Testing dnspython 2.3.0 with tempest\n\nAdding depends on to fix nova eventlet issue in devstack.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/870988\n\nChange-Id: I7a847bef217140c3c1aa84bdd7b33a51884bad02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7bd20a51ebfdb5f874373d9fb020ba704557ae23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7bd20a51ebfdb5f874373d9fb020ba704557ae23"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
