)]}'
{"id":"openstack%2Fdesignate~871164","triplet_id":"openstack%2Fdesignate~master~Ibc2147fb08beae7a2190b9860e3342150c43dfab","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-01-22 00:00:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibc2147fb08beae7a2190b9860e3342150c43dfab","subject":"Use new get_rpc_client API from oslo.messaging","status":"MERGED","created":"2023-01-19 20:36:44.000000000","updated":"2023-01-22 00:01:58.000000000","submitted":"2023-01-22 00:00:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871164","meta_rev_id":"218e348c9c623d13cc49becb342618129fc254b5","_number":871164,"virtual_id_number":871164,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-22 00:00:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-01-20 21:40:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2023-01-20 18:17:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-21 08:25:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-21 08:25:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2023-01-21 21:26:51.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-19 21:58:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-19 22:54:46.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2023-01-20 17:12:31.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2023-01-20 18:17:32.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-01-20 21:40:25.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2023-01-21 08:25:21.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"43b62c3f478e9367630c5fa634f1a19084a8f29c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-19 20:36:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"acce6fdf22e83daa30ba8053e27b864064797371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 21:58:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5420e66253344ae2b9df2482cd6c2de2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/aff876a92ed74ae0bad7fab3d682bbff : SUCCESS in 38m 51s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/66d04963f5694825bae428281543fa48 : SUCCESS in 53m 42s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/418891ec09d8473aa3dd6943ab5ec653 : SUCCESS in 39m 51s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/70c3b44835e248cb858886b24610e849 : SUCCESS in 44m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/25875c06d453471fbcb00d3100ff2cb3 : FAILURE in 38m 07s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/17ad2dc103704ab19e73231d4abc768e : SUCCESS in 1h 09m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9f608653b22f47f993081c369aa5fead : SUCCESS in 59m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a0c0df758c0548d9b76c31ae7f45d6ec : SUCCESS in 42m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e3b36102836a46c19a780a3992f80340 : SUCCESS in 42m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5973da76e06046598cb598a98f61c57c : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d19430616d53407c92fd2d4582af9d7f : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f58be298a46b46f4a09554b8ebf4ece5 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0476bada51434297a3dc15009190e95b : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a90907197f934f2aa9034b2a31b75898 : SUCCESS in 8m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/901459b94348419aa8069e56e8e457a5 : SUCCESS in 1m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1e34f35e530b4284ace5a2d145da00ea : SUCCESS in 27m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cc52894dbbd6489e8c32ce7312a4198a : SUCCESS in 5m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53001a70ff0a8c4cb510737521bc2d6db08ba40d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-19 22:54:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55cf9f2405a4652e9d2ddf153f713b050948a7bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 00:11:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3a39e2398664fb99cb227ad809e430b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d71cff4bdeed4fc39d2f5ef698224df6 : SUCCESS in 47m 46s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/72d4f7b7c7b44021ba24ee9736a76ea8 : TIMED_OUT in 1h 15m 27s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2cc2796d2d34eccbf3a11afc005e752 : SUCCESS in 52m 16s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4b00e424b1964ccdba428194c38924a6 : SUCCESS in 54m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/eea53f5808f2452ea526eccd1b3698ce : SUCCESS in 48m 01s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d01d5df2bc8b4cb29fc06b3e68db13a1 : SUCCESS in 1h 10m 01s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1297c8ba75fb496c8c496e5f6c6362cd : FAILURE in 11m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7a1f984860fb4ad1bbdf76c1dac2e1f9 : SUCCESS in 45m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ce25a49629264d6c885a8fd4eb1c9650 : SUCCESS in 45m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e13bf395c13447f9c7821a20ce38fcf : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7963ff0f4e494a90bd893757dbf2382a : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e3d8d0588244e11a1277c62aeefde5b : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efc96ea1b1b6420282b14b58926070e5 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0d024a4131e49c48d6b139ff4696c16 : SUCCESS in 8m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e018d68462764ee4a91bcb0994c2d2e6 : SUCCESS in 1m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/92abb0a80eb44a44aa5a52b44e527ffd : FAILURE in 9m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ec16e82b9867412399f49d478d9b33ad : SUCCESS in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6113952c27346c5ab54cc8314a613b52adf6bdf9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-20 17:12:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84deb79433c996004d240badf58d5b854312e8b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 18:16:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af2def017b914bdd8c7b5f94c53a5d1a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7946c7609451466f808a4c0b45516753 : SUCCESS in 38m 50s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f6ff965a808f4d528a98a8dc8ff8699a : SUCCESS in 59m 03s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/893ca7838b3542aba30d8150ce91db2c : SUCCESS in 52m 59s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6e05892fe8a44ae69b90dfa7f043e9fe : SUCCESS in 47m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cee8e151d25849749e6e443803660265 : SUCCESS in 41m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9251999b855f4599a94e25e2ec97a317 : SUCCESS in 1h 00m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e46f71ff8b1a44459dbf36c2023fb905 : SUCCESS in 58m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/56d5615c146c4ca2a7cfeee498d61ff9 : SUCCESS in 36m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ebf8586f71e743e28f401daceff3b791 : SUCCESS in 42m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69a5ec1013fc47bc804658951982cf7f : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24298687ac6044fd99248f8b5cbf7bb0 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec6f203c4c4240679a539d16abfa7db2 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8babdb195ffb4503854f800dc654d785 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcd907f6848f4be39468bb52f5a54ea9 : SUCCESS in 8m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/46a320ceaad344728c69f6478f498917 : SUCCESS in 2m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fbd12a4e20bd4583bb7695f5f943d0db : SUCCESS in 26m 45s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4712f25d935b4d21830a5f0c876d406a : SUCCESS in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a1dc98fbfd5179dd1cdf5b56533270f4710ac34","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-20 18:17:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df2e573bebc8672b33df35f6770f0c8fb53a1754","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-01-20 21:40:25.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f00aa5ce6c85aa732570a3736a203ad7371b96f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-21 08:25:21.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"e8271ab002cc24770882893bf7e371d6c2422af4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-21 08:25:26.000000000","message":"Patch Set 1: Workflow+1 -Review-Priority","accounts_in_message":[],"_revision_number":1},{"id":"ce0d5b02ff7ebc140dea071531061d3bc2b13606","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-21 08:26:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d7ec0d549fed5558c1825a799d96c1dd28076db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-21 09:25:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0580d384d57849b8a69403ae62804322\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2b4555c79f4b4b879cc6e0f85c6d5aa1 : SUCCESS in 38m 15s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/86e4b2fdc3eb4a73a8383da8ebb5cf13 : SUCCESS in 40m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3d45552cb6bd487ea24f8d66bded9d2a : SUCCESS in 37m 33s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/31a375981fac4094ae118fccb3bc9f78 : SUCCESS in 54m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dd7ede9e9a5d4c1291b77e7a68da49c6 : SUCCESS in 38m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7bea6b99ab3c4bb8bbad03dddf93d048 : SUCCESS in 43m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea0044dedb4c4502939dd3e4483962af : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1d1faf0b23f41c1932d064eb5791993 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76be995e925244efbbb06c3eca0f7bae : FAILURE in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b24f4eafb3c6484096533fb7439a2dfa : SUCCESS in 8m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4371454085dc48be905bf1d9f177368b : SUCCESS in 1m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/66e5e360163a435fbf91f344a731552a : SUCCESS in 34m 33s","accounts_in_message":[],"_revision_number":1},{"id":"d2b69f08b44c96b92e11cc6f29e6078313276944","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-21 20:16:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a75f58605f8fcf513219848bf8e1e7e48f25eb72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-21 21:24:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a1b17e795b94d61bb1215f1c4805375\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/085b36a744274e99a755352444f81ab0 : SUCCESS in 34m 36s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9ee72d9ac6b847e0aab1bbece099a84b : SUCCESS in 1h 03m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8531b79662904bd8949b952aaccd82ea : SUCCESS in 34m 34s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9cf4186d90464fcd913f9156cfd6d0ef : SUCCESS in 32m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/40d182f70b2d4df2914d17aaa10e1a47 : SUCCESS in 29m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/328bf447cce94cec9434c54b02b06009 : SUCCESS in 50m 05s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8dd2a823b60f4c719ed284b0328f77aa : SUCCESS in 43m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/484cd69cc12b45e0b7de40b600cca041 : SUCCESS in 46m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8ca183c5b7fe46b99bd3f3af51faec1d : SUCCESS in 41m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc9c659daf0243848596ec7fde251141 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/141a2fbcee4e47b29dd68fb3aa33b812 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4c19d3f648c4ac49f70638e324e963a : FAILURE in 8m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05e492489fd84b59bfa6021e0f71a3f0 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b900eec460604ec8bc5c55e6bfbe8e1f : SUCCESS in 6m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/612cbd5eb15a46ea942df32294fcdaf2 : SUCCESS in 1m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b8abaa6cb85348679865083faae5a8c2 : SUCCESS in 29m 21s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5f82fa7de7dd46e3b79e305f0752cd1a : SUCCESS in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf3613452360b6e892a595d8606bd460fc665803","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-21 21:26:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa829b746f872cfd2dc8a5b6ee62f857a5d0ea7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-21 22:41:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97a499858f834576b06a4de615b95e52\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/33f78af1a8f341a28bfe3c59af50ab3d : SUCCESS in 42m 48s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a9ac3de1a78249ceb25f1dea1e9bdd4f : SUCCESS in 1h 03m 07s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8804d3eba45e4799b51e4386b11ea226 : SUCCESS in 50m 59s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/af9134a5770943a3b10ecb25452140b5 : SUCCESS in 44m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b67b018847a04b26b18e2dc813045648 : SUCCESS in 46m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8bc8b4a6a8734489962275115ed3a8cc : SUCCESS in 1h 09m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6b8b1079bec14413b1e9ae3f5fbd2973 : SUCCESS in 1h 06m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b80bc6faca8e464ea7d047598f678d02 : SUCCESS in 48m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2c322824b5404822a87e89bb43407846 : SUCCESS in 53m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbedfd4d44c14ffaa3f005159df9acfc : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ad5e4cfbc544f17afbc840a18e9d891 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f74eb25f80b43628b9b394191267c56 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90fd6af52b7c4dbfae17f0cda701a3b2 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8573afce14e64209af3e336eb12a1059 : SUCCESS in 7m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0d2819a07a194b8caa44972b3b8fbe8d : SUCCESS in 2m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/81606725e496427ebbc59fd561c9d028 : SUCCESS in 25m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cfd4b4db73f84a52a0dc83ef7d64624d : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a0d8819f457514cf5d21cf3be0b553c1189ee8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-21 22:42:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4810c0b85365ab2ddb9daf434fce5a1243eba00f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-22 00:00:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08be40b3aafc4d52bad7e901ff7debc8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6d84246bd819458b93d5c84757f809ac : SUCCESS in 43m 46s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0606e4f461d94b3da797c8a658d8e20b : SUCCESS in 50m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dd860efeb82f4079acefc27a61c87edf : SUCCESS in 37m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/12807bf73a414ecfaa888d7ad9eec286 : SUCCESS in 1h 09m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/44551b4c9b3744c5aef3f5d7f95c4ca1 : SUCCESS in 36m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5e9e00cf35624ef2a8efe41224158877 : SUCCESS in 48m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e873bbe1e7f94b749e67d64bda46df12 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c701d170551474f93b6939bd1c75569 : SUCCESS in 7m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb6da2d9606d400d9972d4ebcc6d5d45 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e90d8b42e64422e89ddb1234fe127b1 : SUCCESS in 8m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/24cc9778e1ce4053a32846b390ba3946 : SUCCESS in 1m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a5713b8b55e04105bde980e7d7891f5a : SUCCESS in 32m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a30ae4f8bf39620369f7f10ac1cfddfbbe6586ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-22 00:00:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"218e348c9c623d13cc49becb342618129fc254b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-22 00:01:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcc161a078ab4e149794d14df168391a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63bf5619bf9b4e92890a5144346454fe : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31b1959620b335467700ed6c33014837e49a1248","revisions":{"31b1959620b335467700ed6c33014837e49a1248":{"kind":"REWORK","_number":1,"created":"2023-01-19 20:36:44.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/64/871164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/64/871164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/871164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/871164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/871164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/64/871164/1"}}},"commit":{"parents":[{"commit":"41bf7dd319966e3fa0182b88b0869ab13b52092e","subject":"Workaround issue with dnspython \u003e\u003d 2.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/41bf7dd319966e3fa0182b88b0869ab13b52092e"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-19 20:36:37.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-19 20:36:43.000000000","tz":0},"subject":"Use new get_rpc_client API from oslo.messaging","message":"Use new get_rpc_client API from oslo.messaging\n\nUse the new API that is consistent with\nthe existing API instead of instantiating the client\nclass directly.\n\nThis was introduced in release 14.1.0 here [1] and\nadded into oslo.messaging here [2]\n\n[1] https://review.opendev.org/c/openstack/requirements/+/869340\n[2] https://review.opendev.org/c/openstack/oslo.messaging/+/862419\n\nChange-Id: Ibc2147fb08beae7a2190b9860e3342150c43dfab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/31b1959620b335467700ed6c33014837e49a1248"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/31b1959620b335467700ed6c33014837e49a1248"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
