)]}'
{"id":"openstack%2Fdesignate~871255","triplet_id":"openstack%2Fdesignate~master~I5e9733abaaa40c874e1d80d7b57e563df0f12cee","project":"openstack/designate","branch":"master","topic":"batch","attention_set":{},"removed_from_attention_set":{"32755":{"account":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"last_update":"2023-04-06 22:51:09.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-04-06 21:07:27.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-04-06 21:24:23.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I5e9733abaaa40c874e1d80d7b57e563df0f12cee","subject":"Move to a batch model for incrementing serial","status":"MERGED","created":"2023-01-22 02:14:12.000000000","updated":"2023-04-06 22:52:18.000000000","submitted":"2023-04-06 22:51:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871255-batch","meta_rev_id":"51d15d861005cb6002dede63569b0e66625ad573","_number":871255,"virtual_id_number":871255,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-06 22:51:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-04-04 03:48:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2023-04-06 21:07:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-06 21:24:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-22 02:15:02.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-01-22 02:15:02.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"REVIEWER"},{"updated":"2023-01-22 05:29:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-03 20:19:49.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"bd6e41b385b653b13fcae7f9d92a2c20c03018af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-22 02:14:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"491d6ebfe42adf14117fb28a32fbd5d4c343010e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-22 02:18:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"be8d9013e3ed44cdbfff1269ce8c474b48314e4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-22 05:29:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8ee4b00edc84863afbb4b4a816c04d0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/00a6edcbdb7e4675bdcf2d8c14b1141f : SUCCESS in 2m 35s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4856bd457d7d43d2893fda73cd0bf007 : TIMED_OUT in 1h 12m 58s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8fd701ffb8e545339b5370276c2d352e : TIMED_OUT in 1h 15m 13s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/80ecba43243043aaaf76b273036ef5ef : TIMED_OUT in 1h 13m 18s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c6eff8c2d3784bea9db3ec62729c47c7 : TIMED_OUT in 1h 13m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/41866485e7704758a1827e2b7eae09d5 : TIMED_OUT in 1h 13m 52s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d5373c818b154b3998410fa9b46573fd : TIMED_OUT in 3h 03m 08s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4ab4006bea374da79563215773607e0a : TIMED_OUT in 3h 02m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/faed7e44576f439f848d11f769d6a725 : TIMED_OUT in 2h 02m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/34310bcd7fdc4b29b7d7686cb4eabaa5 : TIMED_OUT in 2h 02m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b2eaf67c6374f69af81c2338f62734b : FAILURE in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2039176e26a148b0a9fb3b81ebaa9926 : FAILURE in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1cd88825a8d446586339577dda00b02 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/247139e94000427d831c59ab4cf7661c : FAILURE in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe7919ce4a4c44b7bd84fe1bbdccff4c : SUCCESS in 7m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e9d5e79298bf4da89aa263913d9606b2 : FAILURE in 27m 37s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f4c3024539104f779126979ccf31110c : FAILURE in 4m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"76234b7d564e65eacbb67bdbd98c260c696b1697","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-22 06:41:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"87d7a74aed8c572743bf8b94c5a27c3e0832a74f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-22 09:50:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bc535cfd8ee46648bfe8f8491ec28cd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3f41971fc0474cd9b03f8d6669fa6af0 : SUCCESS in 1m 37s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5536464480cf4851b17d05c4d203914a : TIMED_OUT in 1h 12m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/645511f6314c430b9775ed95c2847939 : TIMED_OUT in 1h 15m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbb360d126144e5cb5bb3e9aca44f8aa : TIMED_OUT in 1h 13m 56s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/fcf9adda1e0746ca939206497b2b2a25 : TIMED_OUT in 1h 13m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dc5a9ff57d5747c1bafa5371b5ff366b : TIMED_OUT in 1h 13m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ee60576dd23f47b0897009199f155d70 : TIMED_OUT in 3h 02m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5610bd11f4314f29b6f18f1392f84a6b : TIMED_OUT in 3h 02m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/147577c233824e61a5b002013c8841bc : TIMED_OUT in 2h 03m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fd8bc88e72744886966c54b0970ea8ab : TIMED_OUT in 2h 02m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/389157bfe6b746cf84598ef29598a3d8 : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/925599d0a07c4f17914b163a192462fa : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbea70a7efea498d90dd6281f04b6e98 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f44079caea1f4235a3c8fd242a3b13dc : FAILURE in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69f0a2acb3d240c69d7efe81fdb99d9f : SUCCESS in 6m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0542f4d530954941ba61a839ac67383e : FAILURE in 37m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/545fec7786c5477687b949a6f4a3c92d : SUCCESS in 4m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"675d64d48438a173d51620ef4dd54974bda3d746","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-28 03:24:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9763c16c953adf4ef52b797eb6c31568d9eda560","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-28 04:12:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e5e4beaa212a11750403a54453c8c71299ff35e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-28 05:42:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"63be33c7e29c7145296ed86c6b16e6c783662029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-28 08:51:27.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2d6499f4c294be8884bdafb6b8fe02f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/12a01025c8e44b0689e359df9522c50a : SUCCESS in 1m 38s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/db8607d8b24e4f6d8cf7eaf925b06e63 : TIMED_OUT in 1h 12m 40s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5cc13335310d4bfdb657d5ef5e370760 : TIMED_OUT in 1h 14m 34s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2dab57899be548898692ab0cf446c2b4 : TIMED_OUT in 1h 13m 32s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3d567c24c03940b2977bd5ad16b963f1 : TIMED_OUT in 1h 12m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/16fac26246774ff0ada7a467f1b47cd6 : TIMED_OUT in 1h 12m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d249a1ed724f4f33a8d05451a13fde83 : TIMED_OUT in 3h 02m 58s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f310cbf4a999418b9dab986ab6a5de00 : TIMED_OUT in 3h 02m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/31b54a3cad1945df89a22be8f97ad0b9 : TIMED_OUT in 2h 02m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/28796b85731a4f16b24d0dd08ddaaed5 : TIMED_OUT in 2h 03m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6c3c83b1113497792eff8292e37e4f1 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90448f93dede4da5a2803cba621be9be : FAILURE in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/831aab129d634219a996f41531d9f2f8 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a602ea375bf8436c942174623b63c9b1 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2e461ad90714d8dabac391be3bb59f1 : SUCCESS in 9m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/feda9118a3c64464872073204ef60c92 : FAILURE in 29m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d6282decfbd149c2b684e02561970832 : SUCCESS in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"35e02a6df07c7c7f7ebce6444437ad7be089f01d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-29 04:11:45.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a47d3033c734419264dedd2ca3f4d81450f4d6ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-29 05:22:30.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"406e434f16b0eb68d9d0c318612eaa8f9b642feb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-29 06:15:37.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"9cf35046b773445b3bf20d21b5e6c3f47b68c4d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-29 06:51:23.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"12ff080cae7370913a7f00dcd201fea3f03fd715","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-29 06:52:04.000000000","message":"Uploaded patch set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"29d35e12290f25a349048184e86e99dcd8b79cbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 08:14:40.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1842bf975f04e4f9405c15fd720c91a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b1adff3c14444b5080b3026edbb0dcfd : SUCCESS in 2m 15s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/17e1b7f224074a3ab5d46771da843ab8 : SUCCESS in 53m 32s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4ccd29dd817e4053a60dface6fe42522 : SUCCESS in 1h 06m 31s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b93c2a2061814378a356653a88441189 : SUCCESS in 56m 50s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/36ae9b911f2343558d8c1de66ee07515 : SUCCESS in 53m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/42fa5d997b964ab0801752571f7aba68 : SUCCESS in 54m 48s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f4d806f842b2464daa638273a7d4382e : SUCCESS in 1h 04m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/76f7ccec0407449aaea2a210d073fd2e : SUCCESS in 1h 16m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5604a90e307b4b2ea92d69dd0758f3ec : SUCCESS in 51m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/31dc9a57a2d443c285512dbda8b51e9b : SUCCESS in 48m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f97847eb6f3145d080929f7b34a8e49a : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b66eb66327524edca5f2e0f885025dbb : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1296b03fee594ec99a26d79acbb4d6c9 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d6152946a2a4cb28c5cb374632ac145 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf673a1b636f4693bd78faf5e9be6944 : SUCCESS in 8m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1da9f0d4008140b1b96a1bba4ec28e8e : SUCCESS in 24m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1563339675eb4a90b320e00ac5e81a27 : SUCCESS in 4m 06s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"441d78eeaed7a69d002425801e739ee5b6b7892f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-01 04:21:24.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"6ccd7c845ed78336ee0bdbd1e828f39052115903","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-01 04:45:09.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"e9bed6c3a351740c8ac7c0e334d3668b61395d66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 06:00:43.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b318a02d7ef4ee6b1a74b4c8db9fe99\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4fd39ec87a1c4f71b24f0087c40ae7ea : SUCCESS in 2m 00s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e6001f6683ea44c4877ff18671af488a : SUCCESS in 54m 16s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/794fd82b62bd4657a14922f9e3017f61 : SUCCESS in 1h 10m 12s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/be9ef8faac5348aca4fb56298e44fa03 : SUCCESS in 58m 11s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/dd5d86abdd5b48f3b1cab6bc10350920 : SUCCESS in 49m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/28a3fd5416e947f8a3185e7419305f7f : SUCCESS in 44m 56s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2a8b997dc7494c8d88fdb2f1e45c9240 : SUCCESS in 1h 07m 36s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/60bbaf6aae314b54bf44c7e55d9a5ce1 : SUCCESS in 1h 08m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/14043671dfdc44f2842bf9e519e557da : SUCCESS in 47m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bd7aef29628544ffab386aaa79a4f0fe : SUCCESS in 52m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4f6e3c3d8b944ab8a3e4c37fe2c4ac5 : FAILURE in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9964d63fb0a743b8aa17ebff52fcc9e7 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f355f6bb086043b480d6533687046fec : FAILURE in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5fe3062b849d45daad47e4d80804f199 : FAILURE in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86fe4ae714304119b5447579af0d07ad : SUCCESS in 6m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/15431094489446cb8a3991f3dc0fc8b4 : SUCCESS in 26m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cfcbfdc0d4da4fd180e18cda82a99c8a : FAILURE in 3m 19s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"999b5bc7368ca7219c13ce167ebcfa313945d650","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-02 01:36:09.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"9045aecf8eb4a7d45b960ee1010bc10817dce010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 03:02:09.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/49cc88b064524fbd9d7e871b8f2f0530\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3da364fe3af14b4fbb4e3399cb7ed3f9 : SUCCESS in 2m 03s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ef82a4d6223943289c6ef42b57f71c5c : SUCCESS in 54m 31s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/517e5d68dae843d695f290fd3c8f0ee3 : SUCCESS in 1h 05m 59s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3c8ef00053f44f9953d85326ceaaf44 : SUCCESS in 1h 04m 45s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/cb3055e98de44fb48fb594da5905940e : SUCCESS in 57m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2e851725453f44cabf6eaa880ebbf0e5 : SUCCESS in 51m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1ac90a7a66af49738cfb7f66ef86c20c : SUCCESS in 1h 12m 37s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c4434a6c2ec34efa800d68ae37a0c2fb : SUCCESS in 1h 19m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ae981d707df840c591e83ab06ad51109 : SUCCESS in 50m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e6638214abf545a3946d49d671c52eca : SUCCESS in 52m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac5a4f761f47481598a525c638d11eb0 : FAILURE in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22d27ff5cacd474b88180bc0c7f28cf7 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0c13f49752a4c42bb40aea8a85efe1f : FAILURE in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a3e13682eb84a78b52dcf2fa19cdd2d : FAILURE in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b0a3618b2694863849446a7ba050c1f : SUCCESS in 7m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a83895977b3b47caa077bf2470d01274 : SUCCESS in 27m 10s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e57ff14e0ada43189d1980acfe96ef18 : FAILURE in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"81edabb94ef2e9fa565efe44622567fee6a02bb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-03 00:12:27.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"d0af69be3a06c0326c7175dc9dfd0b7daa8ab194","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 01:12:13.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39196ed0d5ae44d0bd752d3891f5cdcc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cea57c7910464f54a66d699e2565d58d : SUCCESS in 2m 43s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/eb001e4af751477ab0d638e24f893956 : SUCCESS in 37m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0c7b3f67b88246c989c1652c2d7c7c73 : SUCCESS in 50m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c01b27197714ab58c3dc801e59c56d9 : SUCCESS in 48m 39s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/575a5380e0ee4473b46f634223ce8023 : SUCCESS in 38m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/88435e29b46c411f8929f3c13d7dd907 : SUCCESS in 42m 14s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/30dcd33970504b1fb40616500a24a2a8 : SUCCESS in 52m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/288c897ed96a4eb7af411c6c2b1b30b4 : SUCCESS in 57m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/223cfb7c0de546b5b479726b735e0fcc : SUCCESS in 38m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5b6b0364924142989510c66997474004 : SUCCESS in 48m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7a4f0af185640f3b7a9c6706cd28e4a : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcace7cb7ca74d24a79ba43c47ca4a96 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83bcd6ee80dd482eb185cb8232fbcd65 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ae533c2a40547a995c173646467c3f4 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b2ebe4c0bd742cdaf46abe1f16ea058 : SUCCESS in 7m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9eb80c617c164ddc9a0aa7996ef3fa40 : SUCCESS in 28m 46s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4d22750798b04d01b72429c463e4ab80 : SUCCESS in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"b48a1d272c408a816faa8318178f89fffc47328d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-05 09:24:24.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"cefe75df38f527c837026c1c710577346d91f256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-05 11:22:53.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4799cde2fd864793997a3a3398708759\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7170164cda5946dd913d5804e5a8f722 : SUCCESS in 2m 34s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f15df1c0c4724611a883d5c91b155db3 : FAILURE in 32m 49s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/78fa42b7d51f4aa3a98961d74346ef1d : FAILURE in 45m 41s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d33a31ac4644867b10df78d70368878 : FAILURE in 38m 18s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/52d9e0471b5e41139f9d4aafb7343bdf : FAILURE in 33m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9e878bac92c74eb7b8825f641c85061a : FAILURE in 36m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6cd003cba81f45558b1118fa92317765 : FAILURE in 46m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e38890c534c84f979d7c527df1e763ad : FAILURE in 50m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3c18a551b5ff4a18b54af1b77ab6fcf4 : FAILURE in 32m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ed19d82e844d4b4c933bcaa273f7631d : FAILURE in 1h 57m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00e0dc28e7d04582a008aa5eb39d5f2d : FAILURE in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66cec7f5aa7e498ebdbc8a39a8815697 : FAILURE in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7437467b2e544d57a942436bfb6116c3 : FAILURE in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3796f39a9ef4aad88c25574eede9168 : FAILURE in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f753772a0ee54d37bd942653246ba242 : SUCCESS in 7m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f70bae9aedc5448a83dc622e416f8b75 : SUCCESS in 30m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ce24d06d623d45758af5cd5535545454 : FAILURE in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"0cf2cb15b0567341f11041a50e2c95ac37d6de54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-05 21:43:40.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"e9bfc37619ba0db915ed5c6f2daf6222d4623abb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-05 22:45:03.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8db098987586464184e1d6dab0149885\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bd52bd9d37684495a57b4d199ae6184b : SUCCESS in 2m 07s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1be4cba01dd04e2d93ea5df76bd89af3 : FAILURE in 43m 34s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4b0c8dd66a0f4d83a4140c222e2b4786 : FAILURE in 57m 16s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/abf9ad5148344f94899ec44849b7b836 : FAILURE in 44m 11s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6f426dfe1a8c4440af22fb114cadf376 : FAILURE in 39m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e83dd131d9cc4f4f9c8371d86fe0ca2e : FAILURE in 41m 25s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4be0621e3dee44338fd80c6a902de887 : FAILURE in 59m 12s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/900d8b4afb78425f847c7e464dc0fbc8 : FAILURE in 59m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6addbd2afdb94cd1a8c910000ad55327 : FAILURE in 40m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a75555624ff548c99d12d89fd84ad2d6 : FAILURE in 45m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1561429bfea94c2f828e4eaa09af53be : FAILURE in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/878cb57cc2bd4c97b5bf5ca85bbb1b32 : FAILURE in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8abc0611667e4386b9ccba1bc18babe0 : FAILURE in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9eda9ce8900246efa46bc05c6016b500 : FAILURE in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e3a24ad517846018b23f953bdbaaa5b : SUCCESS in 8m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f0d65a074e274f198e3256239c3429c3 : SUCCESS in 25m 31s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2fad55fcb70a43988f084967eced98fb : FAILURE in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"623601933cf5edbce9aa12364aa3f4430e2fbd4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-07 02:56:20.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"e9fa1e742f5327731a51e5d454bf1137e23b5274","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-07 04:02:34.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"60ade613f472017129e7e58227df63f635dc8016","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-07 04:45:36.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"42d5b969bfa62ac01868cd0824ae1ab170032ed5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 06:05:46.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34a7f99ace454741997e7ad19c60a1b3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/03ce48f83ac84a0d99f1080a0c446475 : SUCCESS in 2m 27s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3244ad35d33444ed829e03c45c0131b2 : SUCCESS in 46m 28s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1ebd65f21c764c3695dc87faa757c3db : SUCCESS in 1h 05m 52s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d94042e0b60c4776833d02e1f30cc1f2 : SUCCESS in 1h 03m 02s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9e024b151b8a42cab06ecebbb525ac60 : SUCCESS in 45m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/be8ca184752948038e36951f3b6931e1 : SUCCESS in 46m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b9fb224a43e14673a6aed62bfe37a6f7 : SUCCESS in 1h 13m 50s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bfbec839c9c64d6fb8cdc56b2332a2da : SUCCESS in 1h 07m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/673700b0c1814d2186311825c81adfac : SUCCESS in 44m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ad72953b7e564d21a06590f6041fce29 : SUCCESS in 44m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04b7a921f01149d79b32464d05e66a54 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75bdc13ef1944e4c80db80ecec020e25 : FAILURE in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1fd6bd1409e4781a5c6bd58dfffe83b : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95448e05c35a430fa2d4c2de7587561e : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/673311ff5d5147d28e5fe862339c2669 : SUCCESS in 6m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fb45759e5b854dab897560836298d67d : SUCCESS in 35m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/114c633014b241c686d153992431bdd3 : SUCCESS in 3m 54s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"c403b6ab57c5545cc00e873021185231d65bc59b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-07 06:32:09.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"21cd91d8f339fc6ee9635893e6489b9e31e71625","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 07:56:13.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06b39e7f63524fe18ceb010ab461e9e0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/52cb6877b033445d90ed64f2767e93c6 : SUCCESS in 2m 41s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a7e6fae8019948b0ae68d42f3fb0644e : SUCCESS in 39m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e11fdd94eff44e5095b380fdb3e61918 : SUCCESS in 1h 10m 28s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d5886e51449492590a7bd72da3c8270 : SUCCESS in 40m 20s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/715bc478318b49cfb64f44f287bab6b4 : SUCCESS in 34m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1ea3e74092ef42e7b8abad2e50d18169 : SUCCESS in 55m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4fa6887e3eb0438b81c9a72c08849b90 : FAILURE in 11m 03s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a6fe17aa9582498fb2c3bc8170d46632 : SUCCESS in 51m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8852ad07084e485099be9e87f0c22b6e : SUCCESS in 47m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7441e830ddea469eb086246067c39793 : SUCCESS in 35m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8a64e613a6f43dc95fdef8c2a95b211 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10f5819360244a3faac4450d5f0db4f9 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d70fc0d28f3b4a4ba23f3a08a829ec56 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87f8e4b2f78d42848071179d4cb764e0 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7371c8aa604e44dcb94b65377c97f393 : SUCCESS in 6m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/abfce6b478fc47cb8cdf79211e923807 : SUCCESS in 27m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/73d80cf95bd54ed790f5b7a387919a26 : SUCCESS in 6m 52s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"4e2b16065d53418806d98714a648eb7e6d244a61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-16 06:07:52.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"19cc6f3a1a1d71c055ba64b01ddad53a013edbf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 07:18:09.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a58ab9bf6a444408c494b6a74e0e1d0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fe6691ce29714078b7dbdf000909b33f : SUCCESS in 2m 09s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/da5bde5d9ccf4c88b31a27e8fb7b5cb4 : SUCCESS in 37m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2ad6311982d64da4a9b979c973d45162 : SUCCESS in 54m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/81b482c2c5ed4d8b9c5ee6de4860e8f0 : SUCCESS in 57m 07s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c9bfacbb49164077ad7fc52f78799a40 : SUCCESS in 43m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c21a9fdb76874690aae04ec33fc3f8be : SUCCESS in 44m 09s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4c745ae7923f4aaf947a99fe5dff01c3 : SUCCESS in 1h 04m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/12caa92a6b9c497aaf7cad4645000a2c : SUCCESS in 1h 04m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e0d2fc39c8194c27afe77be608d26256 : SUCCESS in 35m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9b83893e4b2e4ac489d9cf43f7a17269 : SUCCESS in 40m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df0ef5841688401b93ecd6aa4b254b9f : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2a6cdf856654b39badff2334b196118 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7188c0682eff487f88072ea1094890cf : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9dcdff434eee4217b95c1ca6f6c6a03b : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cf63a87712e4d048853ccefbd020f7c : SUCCESS in 6m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/69e397e573b542d6b7436197f2d40bab : SUCCESS in 26m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d32c2eb69a494fb794200a0f5c6b9dc6 : SUCCESS in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"0c4d773908b0c0d6670323dddd45275afd6a04b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-17 06:57:15.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"69b5856948810e295e1c2fb9522480e858837873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 08:01:22.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84d4890aae5e42e183fa22f36e2c6276\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7ae4d3276dfc475a9f5dda2bcbf65bee : SUCCESS in 2m 05s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5dc67d9bb13442cf822bc1c57809fafd : SUCCESS in 44m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d4534888755d492ab54b9fe73f510c89 : SUCCESS in 57m 02s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0508b97b41a94b7f9549f1cd7c3e4798 : SUCCESS in 48m 11s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/50b0507972c74b1792f656fd034a6321 : SUCCESS in 42m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/db07049c8bb44a078584538586096d0e : SUCCESS in 49m 01s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9a2c7ddea43d4b989f2ce005b03efd97 : SUCCESS in 1h 00m 23s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/05cfacf2a3864c3593557c64c97a8592 : SUCCESS in 1h 02m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/48fe8c4c73454f9d80dcd7a8d8e2b111 : SUCCESS in 49m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5ffd69aa4cf54aa2a00004b498160c81 : SUCCESS in 52m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/536a4435c20d458eac04095af78d6c75 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4f62a64447d4efe859b22b3f42f55d8 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abe8e5b1933f4003a8fdfb628c03bdc1 : SUCCESS in 6m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff704b36d6594a7eb60dcc1a5908d980 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53206365f6594e44ae331e629143e931 : SUCCESS in 9m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d55c9edf0ad441ffbffd9e7109b0841e : SUCCESS in 38m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b7804882f3bd452ab3d9481269171dcd : SUCCESS in 6m 40s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"aca451065bbeff913be23b539bb209fa8e695342","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-17 16:47:43.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"9855abec495dc1381c62b23ccaa6b65ef44c8b95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-17 17:20:38.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"3c4396e773d34a2d0a11a2ff15bb90371f5eb646","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-17 17:40:53.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"ee59101e73a3cdedb193789b9f198792fe4d718f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 18:31:51.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f90e7baca574e03a75ad92a260f0c42\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e94c13e5a2fb4c83b27aacc562651117 : SUCCESS in 1m 58s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fbbd79f1bc7746fca744d1ea41c6df24 : SUCCESS in 40m 37s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2f6c627dca7a464494639332ed66164c : FAILURE in 49m 31s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3173f92a24654eef834eae117de21aae : SUCCESS in 52m 13s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/bc967b67ddab4f6cb0795bbeb28abbee : SUCCESS in 43m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/44703bd3beda4cc9baabaf5015c87162 : SUCCESS in 44m 49s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/49a0a523bf534a07a602d06f6ae7540e : SUCCESS in 1h 03m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/365e3bd2f698480c868a0e25b041cb79 : SUCCESS in 1h 03m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8a478df075364aae89ae6a1389e61844 : SUCCESS in 49m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/89067fb3ec6c4e0cb354c0a0f54e7265 : SUCCESS in 41m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e302d1e5fbaf4b05af9718a18a218108 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6314a6e6bec4f0bae9e12b77dee2f10 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79ee95a85b1141ab84e6ee9fe8d2bb66 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c8236e13d173419bbef038b720392f67 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/061a9e43f5a54072bfcebc224e519b69 : SUCCESS in 9m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a72f64e140ca46429cfb00764d70e6fa : SUCCESS in 27m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cbad290bf2134378a9b0a12f26685dd5 : SUCCESS in 6m 27s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"e1c2d9ad15b7cd46a9195b3a064ceb4fdde2ad0e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-17 18:38:15.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"8827bfc854c7f6b07e5f04ccdbea7c759a62d6e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 19:52:10.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dda6be8b61f0406da567c2fa464ff3be\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/31e08dc92ce546d98aeef3fb191180df : SUCCESS in 2m 13s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ca61be2265ac4a7098ca4ae55ef786a4 : SUCCESS in 46m 44s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/cbd7479f95414b4fbf1b5a610c87d5fc : SUCCESS in 1h 05m 40s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ea7d919fa3044969cd7328b8d7974a9 : SUCCESS in 48m 49s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/78a690d68fa740bcb378fd7b86209d37 : SUCCESS in 48m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/43b1c6aa54d04034adf501dd66d8e831 : SUCCESS in 48m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3805862c238e4829b0cf6f2cf148dc92 : SUCCESS in 1h 07m 22s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8fc9577e17404555be7ff5292e658854 : SUCCESS in 1h 05m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d2db6ba6ee054263b0037a1c36065044 : SUCCESS in 41m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4ce8df35da3d48fe9b72f58f20e033b3 : SUCCESS in 51m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77d999315c164adba41b4b8a2f3ce2e3 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cea0b6b04154396918db7941a5426d4 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3876d3ef947a470e860a4cee750efbb6 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ecefef1c7aa24907a23723c7958e8f35 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5bbaaed245a40f9a9f49702dd067669 : SUCCESS in 9m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/01bb47993b56410ebbf6b6d094d8e6b7 : SUCCESS in 25m 47s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/fb8a55c9761c4a34b7764a8414f6a46e : SUCCESS in 4m 16s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"b4946542b12ba3e480bc296b935ef1fb37d997db","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-17 20:22:35.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"0f9c2774a777da65fc296be7b6eb51e9fcea2f2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 21:54:24.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/458db1a8a48941a9a59dd9fecd8abd06\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b853de630ab64f78948d27371712f73e : SUCCESS in 2m 11s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/027e8d1be4524204a64cad35d4ab33d6 : SUCCESS in 51m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/96ac7e416ec04b5cadb901209daa24a8 : SUCCESS in 1h 15m 16s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/50d7fb371c4448a6a6b2347df3b004f7 : SUCCESS in 54m 46s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3a31e3bd5bc2423192450607f15ef339 : SUCCESS in 42m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/447935717c544dd5808bd29b4ad83248 : SUCCESS in 40m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d22e93f0a2254fdc952382d498a98dc3 : SUCCESS in 1h 04m 03s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ec0095233fec4328b21642517d50801f : SUCCESS in 56m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/207dd1897c66409f98eb519b45cc4879 : SUCCESS in 1h 07m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bfa9fef3bcd747ca8e01f39b753b6105 : SUCCESS in 40m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c49cc93e25be413ca7029e74c291ef89 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83e0670351974201bf9d533f6ab23cb7 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/010e8d7910a1441ca37a761e8088e002 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc3f18febd2947b1b5a4495e9e4fb5ce : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5848e7fceb004421b4399f851c5a14c3 : SUCCESS in 6m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0405567b023c4924b60e32055095b09f : SUCCESS in 26m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/96e4878e82894e89a8e6c2b055ec3d7f : SUCCESS in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"233f87f287fba864b4e56334184a1ecee5b11dd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-17 21:56:40.000000000","message":"Patch Set 25: Patch Set 24 was rebased","accounts_in_message":[],"_revision_number":25},{"id":"484d3403dc1041af215a1e290ace4e51299aab75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 22:50:31.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e93ce66b0a3c4d16be3bf6e8db8dee38\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a983e459afc845689d509a26ce137397 : SUCCESS in 1m 52s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8a67524024a54ad58fea8dcd2e763f3b : SUCCESS in 36m 07s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0353d52fb37242a0ade5004157b9fbca : SUCCESS in 51m 31s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cb6ce1da1e44212b10877f7589bb9cd : SUCCESS in 43m 19s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/5bb58a1a245047829b932c9050805a2d : SUCCESS in 41m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1a5c5b82bb5641338648dd9fe0358c47 : SUCCESS in 38m 05s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b49686beccde45cab54c52591bf76245 : SUCCESS in 46m 11s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ea8f1f3ca8fc465da64d281aa3d0d703 : SUCCESS in 43m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1514f72cf59547208c715ca064c0c92f : SUCCESS in 34m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5a52d174f7e6409b962b9d4b6da2db98 : SUCCESS in 40m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed1663bec12e4eefb67288e145a6bae9 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c42c55ae8872446eabdd78cee46bc250 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa270198ba7e4e5fa3cbcd4f26bab35f : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57217e61f8a84a01bfa71d957fa26a62 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22159df65e084fc0a6b395fd14e77cf2 : SUCCESS in 7m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ea8882e67b1045f599a3992039bc1ce7 : SUCCESS in 33m 04s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d1e3eb3331e54e2eac41dfd6cfa3edd7 : SUCCESS in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"e3f35a268ccc5f9e0c274551fe0c4515248cf856","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-17 23:30:35.000000000","message":"Patch Set 26: Patch Set 25 was rebased","accounts_in_message":[],"_revision_number":26},{"id":"1dcae03211b508948658b1590cb579e106435846","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 00:40:06.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/641432a523104f0ea6a631cef0316344\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e6e42501c42642868fbc0c8af81fcdc7 : SUCCESS in 1m 35s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5a8fc79a9c534c6183b386a700d1f8b2 : SUCCESS in 50m 18s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/13a6963ad1d247748d12152efc1e6ddf : SUCCESS in 1h 02m 33s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c27b2932cc14aefbbbe65212e96619e : SUCCESS in 1h 01m 21s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ef81c57f22974cbeab7bb81945d9a522 : SUCCESS in 48m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a3e2c3610dfa45f6baf66497215aeea3 : SUCCESS in 45m 09s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/305645ed85974366a4a5b566438d91d5 : SUCCESS in 1h 02m 57s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8e2789f59c2044f7bf568a1cd4c55bf7 : SUCCESS in 1h 04m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/aa6bcfb88769486891e9d4d251b122ff : SUCCESS in 44m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9c45842c99964f4189838cccbe0927b1 : SUCCESS in 46m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9782fb7c21d4bb19513d9e719fea6a1 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb4212750914447cbe742a4004f96458 : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08c668350ba14983b038768cf4a422c0 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8094fe413fc149d1ba254335ce2c07eb : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/632c67afc97e4930a6bf0aa6cf1dd21d : SUCCESS in 6m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/151c94aa107244fa9ee492f3744f028e : SUCCESS in 35m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c9b81a315c1b44f79b654fcd3b2ce929 : SUCCESS in 3m 54s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"4f54c864ec6e2a91d62c9460926496325b76c43b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-18 02:28:51.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"7824126af3709ae7a4540f8141444524107442f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 03:47:27.000000000","message":"Patch Set 26:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f3d73d0550548f6ac1e663f7587c9a6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aac2f18be0ce42418bfc2d006ea783ca : SUCCESS in 2m 45s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ed395566508b4c8f929cc1f9b373fd14 : SUCCESS in 38m 34s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/56b69a5a705d4b43961caec50f991f8a : SUCCESS in 59m 29s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/744a1c4d63fd4518b2c14f1c321c2fcf : SUCCESS in 54m 06s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ebbe0814487f45a588ff6bdcb70b9519 : SUCCESS in 51m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1017f2db6b3945ad8431ab57959df250 : SUCCESS in 53m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e8c1d3e81af5489cae631f9ac4720fed : SUCCESS in 1h 09m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3c8dd866c58649b9bb7f59837d65ff95 : SUCCESS in 1h 08m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d6be3e4e969948648db82c7bdcd955d8 : SUCCESS in 50m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/58638d05b409458ea1e59ee235c6f3e4 : SUCCESS in 47m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8763c97b25042eab58d54dd8baf8c80 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eec3f5a400bd4166b86856891c68dbc5 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fa9960d79d34830a955fdbc2772d480 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35f5ccaeac564339b181d07c0c1d7d61 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/903b57519ee342fa99a9c5b6ce9b6fa4 : SUCCESS in 7m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/74410a84004946e6940cf12e23f3c94f : SUCCESS in 27m 03s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/63a92e4f082649de9cc1eb7eec111984 : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"1d6638d05b21e6437bf7fdd2d32bfb49dec12d0b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-18 06:28:23.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"f20d30ac12f30745e62ae5302192c1ee0249972e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 07:51:30.000000000","message":"Patch Set 26:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f55b7d1665654748bae42a4511c75402\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b4840c045e5d4166b748486511e9ecb8 : SUCCESS in 2m 11s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ed21a12912984559b2c262323c59c62c : SUCCESS in 50m 18s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0685190b53e34d89957b6de799a2b9b1 : SUCCESS in 1h 09m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d28516a26db1495f8d0a498922e83cd8 : SUCCESS in 1h 01m 17s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8e5152786ec34c35a32eb5f1419b28e8 : SUCCESS in 51m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c373d8db39534e5dacc7495c82091ce8 : SUCCESS in 48m 26s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2b87c3cda2e647bca7ccb489a2a422fa : SUCCESS in 1h 14m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7983216e8aee4ce4af4c5d05486213d5 : SUCCESS in 1h 10m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/90bc04efd3954bdbadfd8af103a5bfba : SUCCESS in 49m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c0136177379147559e9072af1c57d8cc : SUCCESS in 52m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71d11125b99c4eb7af8d7148e67b92c5 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f891bce95e5f49fab8d94bde6de0d815 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3ba6dcc10504053b2d9ed82afb41e90 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ddb7fb500da44e492e36af5ebca4a4b : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fcaad47fda545efbacf35b248c24ac9 : SUCCESS in 7m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/73abe40455a243b5b535c94c7fadf97b : SUCCESS in 35m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f20de560513e470b8e09a0fe6185843e : SUCCESS in 6m 27s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"f036dbc90ce425ab66ba6e064ec59a61d7715467","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-19 02:55:24.000000000","message":"Patch Set 27: Patch Set 26 was rebased","accounts_in_message":[],"_revision_number":27},{"id":"e00fe97f0cc43286aadf410ea1b7f3ddb571eade","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-19 04:14:22.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15a08d0b9d9f4d4bb7cf275e8be8ad2f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aa1931abe4d14db1bf90196789ab71ab : SUCCESS in 1m 57s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cf1e801cfeff4c7c83bddcd19b803d8c : SUCCESS in 43m 55s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5ae55f14a1164532b4055e2097c0319b : SUCCESS in 48m 34s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/60ae90973b154695b89a4c66ca595deb : SUCCESS in 41m 03s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/98e6c79d2aad4c84bb30a3e1e7fff861 : SUCCESS in 43m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9d881e7a46d946babd22b6e12b1a8ce9 : SUCCESS in 46m 38s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e34e957c1c9845679ba2d4ef030ea224 : SUCCESS in 1h 05m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/10dc3ab909a5485b8b08170d26b307bb : SUCCESS in 1h 12m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/04beba597f4449ebaa2cbec83a2801e8 : SUCCESS in 49m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1c85e816c25749d8b4498f10b4988cc4 : SUCCESS in 38m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62218fb4f5ce4a579cf076fa8c84df64 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c898b9a5f5e4ae083b91c797e731f19 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8aff501a71eb45c9ae26fe908644470c : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ea4d4e98f1045f8951e61b2a9fe79e7 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5f962668e1342b0ae002deda4df4865 : SUCCESS in 8m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1414ff654fbc4b4e8497f0080801cf95 : SUCCESS in 26m 31s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/eacafbc8ae65474b8b2b56cc80573a30 : SUCCESS in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"d287694ad5cf133a56472b746eaccac095fbf139","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-20 06:33:51.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"c395b8e1401e272751b0e455f7ac0e892866c6db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-20 06:49:42.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"8497542af226d0085a291481342228728eef4d91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 08:00:32.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/641773b7add74f0282753bca905b5807\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dd2ade549b5948d2a3132de5e3798bf8 : SUCCESS in 2m 08s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9eebb4c3ed5046e7a9f6e96eb6c68cee : SUCCESS in 45m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3a37b1cb79904d478f316f850e6ef520 : SUCCESS in 1h 08m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f8c99addea4437189926af816397a08 : SUCCESS in 51m 25s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/be138aa2382d42ce95180369212d65f2 : SUCCESS in 59m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d46120a007d1421a975f7d82b9b84b6b : SUCCESS in 44m 07s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7776726247e7440dba775819b0dcd1af : SUCCESS in 1h 01m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/736218e2bbac4058afc65d6147efa57c : SUCCESS in 56m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/144436b1ad7e4e6c909a0c6719283ad3 : SUCCESS in 1h 04m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3d71530c641149fcb27a0ee32a8b1444 : SUCCESS in 32m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eac5c90f3e3f4faab93e242607fee497 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a768ae40872143f0b61ebb4b2ed4262f : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3231f96b4e5e4e88abe5afe58bbb58bb : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f91b5edfd3a14392b5c5a95b561743a6 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eb5e1e8496a4aaab12b64a407c8e1c3 : SUCCESS in 9m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/13ff4f9b834045a1bfe12f265060b570 : SUCCESS in 25m 57s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4744c832b0cb486599a69054373eaf93 : SUCCESS in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"1d5f00b06457ddd90ab505664da528052d500734","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-23 05:25:19.000000000","message":"Patch Set 30: Patch Set 29 was rebased","accounts_in_message":[],"_revision_number":30},{"id":"45d31380e6784d454e2d366ac5f8ef48885c5675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 06:21:19.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7437e66e86cd42a490c5d0c7420c25fe\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/58afe964c7204ec18de9ddb33876cd24 : SUCCESS in 2m 00s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/20b8d71873344468aeb5e289f2f38ed6 : SUCCESS in 40m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/670e33ac7e054e49b4b1270228bc4012 : SUCCESS in 46m 43s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/21b352be69544f4e8d5302300bb27850 : SUCCESS in 34m 36s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/180a5b778152427a81f0bff4644c6368 : SUCCESS in 32m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/079ff9a3552a4ea99bea0c5e9233336d : SUCCESS in 35m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a5200c6f1d3941e190baa848eb35e8ee : SUCCESS in 55m 07s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3a85693d89534ace8d5db64403208baf : SUCCESS in 50m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/aa70f0f8807f4ac49504efb4d3f8992d : SUCCESS in 40m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2ceef330940d4c8da82db679987de429 : SUCCESS in 41m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5d6875f85f444598748dd15bb58192d : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fb71a0308bc4ca4994aa36d2f11d047 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37edb1b7fc50494a8c8b15fc61ec9e79 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd9c6a9b7a294e708f7dabcb9e64d468 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79be188cad224f788cd01254d128bd7a : SUCCESS in 7m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b7715bb959d349de952b0fbc7c7fc7de : SUCCESS in 36m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/63bcf8f9c7d8417fbb9bdb4783618a4c : SUCCESS in 3m 56s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"b71a05578902befd35cee2d72e536d13ffb54381","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-23 18:00:31.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"a2d9d24edb2f0ba13b8c3784e4b3395c316bde6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 19:34:15.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91b88edcc701422099955d0d77c9ef74\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9a3a88d4964d4224a8d5d0dea9f65fd0 : SUCCESS in 2m 34s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/089055095d9145e4b9e206f4635d84c2 : SUCCESS in 39m 01s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/75a2532ad0c34891a923a4a71d453bb6 : SUCCESS in 49m 57s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a683ca7e7eae4422acfa81471fb5d697 : SUCCESS in 49m 51s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ba97360459914763ab1f718756363582 : SUCCESS in 34m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3394eb1a59bf48acb63b5442a87cbbd4 : SUCCESS in 41m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5faeddfea97c4723b53eddf24cb09619 : SUCCESS in 1h 17m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7ec292b111a344988992e75495f97b93 : SUCCESS in 1h 01m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/699f6247fce44cf387b03dbc70bbdf56 : SUCCESS in 39m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d0325a17b3294ad988c9b4c6c8313b8e : SUCCESS in 37m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a88637c424ac414fbd9f981c42204f0b : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bccfa080d38463cae07a3866dbdd002 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f55b9ea3d6a344d9be824127ef590bcf : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de14f145329544899c1531d108c7fbc8 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de84a9a5a2ab48c3a5a2d035e1aeabce : SUCCESS in 7m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fffdd2a549e14100b78744afdadef1fb : SUCCESS in 25m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/19d87c1c359e4c5cb81b8a08f7c452aa : SUCCESS in 5m 10s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"398cbeee676711539669f73625417d319426b603","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-26 03:45:06.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"b203f089e45e7c58689151d51ecb5c18096dd634","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-26 03:46:17.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"5fce54e1041858c3e742b06abdf4ce8b72945f8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-26 04:04:42.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"6e25718b90f1fe6d8be0c051f57a03528a905e78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-26 04:58:04.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5726b97f99394b14b0de1fbc843b254c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/39013095822b4c34bd19dc0900cadf51 : SUCCESS in 1m 52s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/40219e613f554cd8b818c2bec328ad4b : FAILURE in 16m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/780af1dcf32f4205ac935cc819982c97 : FAILURE in 30m 54s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a65f78e583e4b18950d6c05c985596e : FAILURE in 25m 06s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/293077c851af4bf6b923f1ee31ed71f4 : FAILURE in 19m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0240f89ae91a4d10b876fd0aaf3ee427 : FAILURE in 37m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/978bbfd3fabe41f998f25da6379aa6ea : FAILURE in 34m 13s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8734589c790c42b9b3e60156c2cb9dd8 : FAILURE in 48m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bc274c64c1e448a8bf1582854b879365 : FAILURE in 30m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2030837b69dc4ef185f1107af462f62d : FAILURE in 29m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/533fd876ce9b4da5b4fa7e5365837d77 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ec37581e67847baa0003da9bdfceb9f : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49f83438857e47df9c4a54c580858c4d : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/941b25e8fd8248bd89c7ae3f4aee8957 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ff5083d8b5341889bf5dab0324f8d31 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d04ee13a9c3a4d828e834f060a89f996 : SUCCESS in 4m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e2a83ca0f6e74c9bb3a60ddf9ce4bb91 : FAILURE in 30m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2848d9d3089f470596393b579bb42a25 : SUCCESS in 5m 44s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"5e277a752b1a977920498581c41589c5b6795e70","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-26 17:24:41.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"7708144dbd47997b41c822a6f5d2fb0ad9ed6f09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-26 18:36:47.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c760ef7aec9436a9b5c094106a9f665\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7706379931bf4896baf795a37523204c : SUCCESS in 1m 55s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0a877719e0b44f16a1e86e96e85c774e : SUCCESS in 50m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4bfd81933d4749f08ee546117d9a402a : SUCCESS in 1h 07m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0286e3e0b764465bb72ebd0e30841d8a : SUCCESS in 50m 41s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/cfac1edb6142410683a7c4eb969038b9 : SUCCESS in 34m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fd844aa8270e490faa693dcf1d39bb68 : SUCCESS in 36m 07s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ba6cc0468b90468a8a23e56c555e2b61 : SUCCESS in 1h 02m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1d13d0e40e7c4b94822661dbc9291ea0 : SUCCESS in 43m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/53c13f00c5354920b256c1108c3f8360 : SUCCESS in 37m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8646ad163e4248f98e24549e36ba0506 : SUCCESS in 31m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b0b56478fd64270aaf3aa945ab9910f : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62974057fa9842aaaa9665553007369b : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/528f1cd43a204adaa6e7a48c4c2aa04a : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/168a702891ca4546a7a72760bc1c10fd : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab16b967a7fa437690a817a82e1e4dee : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6340a31898d848319efbe2c7057e95f4 : SUCCESS in 4m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/007f565d7cf749a7a5c60c942c30dd39 : SUCCESS in 25m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/07d26a12fe27457b96538f94a90d42c6 : SUCCESS in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"357d64fb80827c3b18fc9d31943335b651b6a274","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-12 05:45:19.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"e6bb62238035b22e5908a9711c518c721e513c1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-12 07:05:29.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afb9bb71676445f3bf1ddb58c43352e5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/419fd5a60da442fabfeeed8a87a36f76 : SUCCESS in 2m 12s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fd22475597e24f68af6714f1581e5050 : SUCCESS in 47m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6ec0218c8ba24b4f9ed3720ae62f791c : SUCCESS in 1h 11m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc5cb0cc26154ce194ca7f062606683b : SUCCESS in 56m 41s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8d3431fc77fc483282197cd9a390474d : SUCCESS in 46m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/35126348eee844a8a720bafa5441ce52 : SUCCESS in 46m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/db9fa2c8c543427fba5bd050f965364b : SUCCESS in 1h 10m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4433222e696643ac9b45a2201ddb9c8f : SUCCESS in 1h 09m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f01f53d9af57431a830be9173bc8020a : SUCCESS in 47m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/19d972e73dc74210ae8308affb21bcc3 : SUCCESS in 53m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f64d189f1ad9496d8a054f9ade336095 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/442d6f98983e4f6b98d43716e7ff756b : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73813351f06747af93b87adf77e318e2 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99fc896d5acf42c6bf70afdb127f24cf : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38336387dfc145789a7de7b22b7829f0 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbf6091a95654b7aa4e3a783344a6ea3 : SUCCESS in 5m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f47bffe9245544cf9f30be38a78160b5 : SUCCESS in 27m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1752f904f8074979a39159ee992e2772 : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"ae6e8da1ac8f2d58bbd92d7790cb4976fe7c7237","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-20 23:00:22.000000000","message":"Patch Set 36: Patch Set 35 was rebased","accounts_in_message":[],"_revision_number":36},{"id":"365f7bc11b9f91c6d934b20fc0d9d03742c7d697","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 00:25:31.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f0daacd75a74ac2b9273301d2454117\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9387a55f6a7343eea93b824e4ab4f74e : SUCCESS in 2m 15s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/59f23beb7e024e30b1d3eccf444f3d1a : SUCCESS in 40m 43s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/29cb7d71648646b69bc71a1e288a1c3e : SUCCESS in 1h 09m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd8fc4795f1843f7b25cd248a6700aa0 : SUCCESS in 55m 59s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/b86bfa3f60624faab6ef15ea05009c50 : SUCCESS in 50m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5311ae636bb1486d8dea74f105daefd8 : SUCCESS in 47m 44s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a4952466391044e6924f6b26cdf1e5b3 : SUCCESS in 1h 09m 36s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c81d9637b41946a48c6f51e57bf4bbb7 : SUCCESS in 1h 07m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e0bbc9c221034ef6b5364484e38ec862 : SUCCESS in 41m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a5003809d50542278e0bc562650683f6 : SUCCESS in 47m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f90f6bbd9f4a44b8b136a488360f9d46 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e17dfb5304154121b54cb095b5eba4c8 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef6874685c36499aa530c96b171e8c57 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5398f3f74a20418a8b21511e5f73dd8a : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6facf113694a48a4be6364b02d6de7f5 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8650b9683019492e9f1775f0a2f440d1 : SUCCESS in 4m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dba4e71418b3465994619d94ce7e7698 : SUCCESS in 26m 04s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/90b29d14099d41fda955a14eddb1cefa : SUCCESS in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"b1a8b99bd02736ceca73c350eed51e02737abcb6","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-04-03 20:19:49.000000000","message":"Patch Set 36: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"244dcf2370d31edf136702d5539ec1379b380296","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-04 03:48:34.000000000","message":"Patch Set 37: Patch Set 36 was rebased","accounts_in_message":[],"_revision_number":37},{"id":"a30edda4389a14973615c1a5ab2d4874610129c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 05:03:38.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc2b82e47025471594d376cfa2ef9bad\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8335cf2e6de84d359e67636dfe07d20c : SUCCESS in 1m 58s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f2b5a313a61b4f5aa5f1774bcd2db547 : SUCCESS in 45m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/35350d0a13734eb790c6ac31055cfaf5 : SUCCESS in 55m 09s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/66966bf1e3cd424995cb433acbcd16f6 : SUCCESS in 50m 28s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/880ec10714464e20be47d645228f43ba : SUCCESS in 49m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b4a9329b862c42d3b783d6f99b214568 : SUCCESS in 51m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1bec83aa9cef48358ef905bcfda59b05 : SUCCESS in 1h 08m 52s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3ba5d4036d434ee387f24a736703b69d : SUCCESS in 1h 02m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/89228e5ca97d4f04a11a2710ae96f51e : SUCCESS in 49m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/16d55bb6bec94b0ebcb2b128ddddf2d8 : SUCCESS in 48m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbcfc5d41212423aa566635b45176169 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4660582286e24af487581213276b19ce : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9ba683687f14e22a0a8315bf9aab4a9 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/423d9f5e4db14b79909e543a9e0cc2f7 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c312b77ed5d4e23b4061d9af2a7288f : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c17d2f599adc409f84f94db8a3e20bab : SUCCESS in 4m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e5947a9eb7684937b0c954c20139bc55 : SUCCESS in 39m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/25ddbcbfa0e0465fadde1b8977cdef14 : SUCCESS in 4m 19s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"0c68719ec94328dca2d91fd8ecc56eceb99f78af","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-06 21:07:27.000000000","message":"Patch Set 37: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":37},{"id":"cba0d25ac384090c0d229526dd3713a19e7665ed","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-06 21:24:23.000000000","message":"Patch Set 37: Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":37},{"id":"76cf5a5118699b4194f8e5e4a6b8a9db9399acd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 21:24:56.000000000","message":"Patch Set 37: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":37},{"id":"3739f801d200c1db5764147e79c8e877a313d1ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 22:51:08.000000000","message":"Patch Set 37: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/442b216e2ec74fbd858e8ee85fee66c5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/74d10a9ac5d04c63b8e89b371c3e48a3 : SUCCESS in 1m 39s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0645a90eec8646dc8eaadee93b5202c0 : SUCCESS in 44m 18s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/eebf8959ee4641a581fc3a60002d3bde : SUCCESS in 51m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ec8f3dc014c341569bb03e6af1cc2d92 : SUCCESS in 34m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/30b9e3aa637346b384632979f5adcedc : SUCCESS in 1h 05m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/38636415c3b04fb3ad482564dffad278 : SUCCESS in 49m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0423f28793c34f84994604ae076cb65d : SUCCESS in 41m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dd796d329b64a228e955f467f16b46d : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc7898987e27436b8904dd24dc4f48ee : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0dd9902304e47fe91ac2ff88712b918 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1d70e504fcc44e29a08a9406293bd9d : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73b7108b8fcd45abbb4bd935cd878602 : SUCCESS in 3m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/73b7d4e91fd2451e9617c9102cf23b92 : SUCCESS in 29m 20s","accounts_in_message":[],"_revision_number":37},{"id":"196fc3d6a6da0ad106ad14bfd996940cf64a4efc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 22:51:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":37},{"id":"51d15d861005cb6002dede63569b0e66625ad573","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 22:52:18.000000000","message":"Patch Set 37:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f2aac1ae7324082b9180f9375cf37da\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdcec32e7b5e46a295a9c02c5ca325e4 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/011a2f9eca6649018d69d69824e5a741 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":37}],"current_revision_number":37,"current_revision":"fb9c2da977a36afc695b1e75a81aa85d8e56ef0a","revisions":{"c3291d34abbc36a2841d0a942ab6f7e05ffe8290":{"kind":"REWORK","_number":1,"created":"2023-01-22 02:14:12.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/1"}}},"commit":{"parents":[{"commit":"ab804672101ebebdd7fe7ab11d2d33c7b95914c2","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab804672101ebebdd7fe7ab11d2d33c7b95914c2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3291d34abbc36a2841d0a942ab6f7e05ffe8290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3291d34abbc36a2841d0a942ab6f7e05ffe8290"}]},"branch":"refs/heads/master"},"b991f382b21a543a416a1233a3a80da81f1214b2":{"kind":"REWORK","_number":2,"created":"2023-01-22 02:18:58.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/2"}}},"commit":{"parents":[{"commit":"ab804672101ebebdd7fe7ab11d2d33c7b95914c2","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab804672101ebebdd7fe7ab11d2d33c7b95914c2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:18:50.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b991f382b21a543a416a1233a3a80da81f1214b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b991f382b21a543a416a1233a3a80da81f1214b2"}]},"branch":"refs/heads/master"},"fca2b421d4bfe37574438653ea5032be407d3dbd":{"kind":"REWORK","_number":3,"created":"2023-01-22 06:41:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/3"}}},"commit":{"parents":[{"commit":"ab804672101ebebdd7fe7ab11d2d33c7b95914c2","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab804672101ebebdd7fe7ab11d2d33c7b95914c2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 06:41:38.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fca2b421d4bfe37574438653ea5032be407d3dbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fca2b421d4bfe37574438653ea5032be407d3dbd"}]},"branch":"refs/heads/master"},"e1250bed46ae16c246f1b65e3f1dc610ce01573a":{"kind":"REWORK","_number":4,"created":"2023-01-28 03:24:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/4"}}},"commit":{"parents":[{"commit":"ab804672101ebebdd7fe7ab11d2d33c7b95914c2","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab804672101ebebdd7fe7ab11d2d33c7b95914c2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-28 03:24:12.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e1250bed46ae16c246f1b65e3f1dc610ce01573a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e1250bed46ae16c246f1b65e3f1dc610ce01573a"}]},"branch":"refs/heads/master"},"062eb22dac7e7a84164bc20a79157382c73ac585":{"kind":"REWORK","_number":5,"created":"2023-01-28 04:12:30.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/5"}}},"commit":{"parents":[{"commit":"ab804672101ebebdd7fe7ab11d2d33c7b95914c2","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab804672101ebebdd7fe7ab11d2d33c7b95914c2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-28 04:12:24.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/062eb22dac7e7a84164bc20a79157382c73ac585"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/062eb22dac7e7a84164bc20a79157382c73ac585"}]},"branch":"refs/heads/master"},"ede0569e614b11702c1f6b1276c0d6937e601ea9":{"kind":"REWORK","_number":6,"created":"2023-01-28 05:42:53.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/6"}}},"commit":{"parents":[{"commit":"ab804672101ebebdd7fe7ab11d2d33c7b95914c2","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab804672101ebebdd7fe7ab11d2d33c7b95914c2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-28 05:42:44.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ede0569e614b11702c1f6b1276c0d6937e601ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ede0569e614b11702c1f6b1276c0d6937e601ea9"}]},"branch":"refs/heads/master"},"48b0e8bdd9ccaf17b967bebfb82accc9c1694e75":{"kind":"REWORK","_number":7,"created":"2023-01-29 04:11:45.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/7"}}},"commit":{"parents":[{"commit":"ab804672101ebebdd7fe7ab11d2d33c7b95914c2","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab804672101ebebdd7fe7ab11d2d33c7b95914c2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-29 04:11:29.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/48b0e8bdd9ccaf17b967bebfb82accc9c1694e75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/48b0e8bdd9ccaf17b967bebfb82accc9c1694e75"}]},"branch":"refs/heads/master"},"10f1002ec93c30454f0b77dcd8559335398bcae8":{"kind":"REWORK","_number":8,"created":"2023-01-29 05:22:30.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/8"}}},"commit":{"parents":[{"commit":"ab804672101ebebdd7fe7ab11d2d33c7b95914c2","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab804672101ebebdd7fe7ab11d2d33c7b95914c2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-29 05:22:23.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/10f1002ec93c30454f0b77dcd8559335398bcae8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/10f1002ec93c30454f0b77dcd8559335398bcae8"}]},"branch":"refs/heads/master"},"753fd393d04ca21f6907ef0b3ce94063601ce6d2":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2023-01-29 06:15:37.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/9"}}},"commit":{"parents":[{"commit":"ab804672101ebebdd7fe7ab11d2d33c7b95914c2","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab804672101ebebdd7fe7ab11d2d33c7b95914c2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-29 06:15:37.000000000","tz":0},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/753fd393d04ca21f6907ef0b3ce94063601ce6d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/753fd393d04ca21f6907ef0b3ce94063601ce6d2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"81655312ceca1540530428733e25ff4f1428999e":{"kind":"REWORK","_number":10,"created":"2023-01-29 06:51:23.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/10"}}},"commit":{"parents":[{"commit":"ab804672101ebebdd7fe7ab11d2d33c7b95914c2","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab804672101ebebdd7fe7ab11d2d33c7b95914c2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-29 06:51:13.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81655312ceca1540530428733e25ff4f1428999e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81655312ceca1540530428733e25ff4f1428999e"}]},"branch":"refs/heads/master"},"b35dcfdb272746c8aa503694d7f1c5bafd6951bf":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2023-01-29 06:52:04.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/11"}}},"commit":{"parents":[{"commit":"ab804672101ebebdd7fe7ab11d2d33c7b95914c2","subject":"Merge \"Use new get_rpc_client API from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab804672101ebebdd7fe7ab11d2d33c7b95914c2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-29 06:51:53.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b35dcfdb272746c8aa503694d7f1c5bafd6951bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b35dcfdb272746c8aa503694d7f1c5bafd6951bf"}]},"branch":"refs/heads/master"},"f7a34e216c94a7733a7878426356e301d39b573c":{"kind":"REWORK","_number":12,"created":"2023-02-01 04:21:24.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/12"}}},"commit":{"parents":[{"commit":"96aae1eeb7a0c00001ad6153d3feb5766a1cc160","subject":"Implement sharing of zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/96aae1eeb7a0c00001ad6153d3feb5766a1cc160"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-01 04:21:05.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f7a34e216c94a7733a7878426356e301d39b573c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f7a34e216c94a7733a7878426356e301d39b573c"}]},"branch":"refs/heads/master"},"a6173797bd789dc59b2b34b9bc51d2c427014b55":{"kind":"REWORK","_number":13,"created":"2023-02-01 04:45:09.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/13"}}},"commit":{"parents":[{"commit":"96aae1eeb7a0c00001ad6153d3feb5766a1cc160","subject":"Implement sharing of zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/96aae1eeb7a0c00001ad6153d3feb5766a1cc160"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-01 04:45:00.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a6173797bd789dc59b2b34b9bc51d2c427014b55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a6173797bd789dc59b2b34b9bc51d2c427014b55"}]},"branch":"refs/heads/master"},"3e9171c11dd29b9514aca41f51fcabfc1158fa40":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2023-02-02 01:36:09.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/14"}}},"commit":{"parents":[{"commit":"7c6f2c76ad9d600387723fb8545fc9ad4477fbe8","subject":"Implement sharing of zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c6f2c76ad9d600387723fb8545fc9ad4477fbe8"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-02 01:36:09.000000000","tz":0},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3e9171c11dd29b9514aca41f51fcabfc1158fa40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3e9171c11dd29b9514aca41f51fcabfc1158fa40"}]},"branch":"refs/heads/master","description":"Rebase"},"924abee02bf92af83ce5d4375a96ec487fabfe5e":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2023-02-03 00:12:27.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/15"}}},"commit":{"parents":[{"commit":"02c6c423d8db74273c56a0cdb443c94ab60f47fa","subject":"Implement sharing of zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/02c6c423d8db74273c56a0cdb443c94ab60f47fa"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-03 00:12:27.000000000","tz":0},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/924abee02bf92af83ce5d4375a96ec487fabfe5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/924abee02bf92af83ce5d4375a96ec487fabfe5e"}]},"branch":"refs/heads/master","description":"Rebase"},"0bb26071018670caf861d9e533d32ff3317c8d00":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2023-02-05 09:24:24.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/16"}}},"commit":{"parents":[{"commit":"cc3882f3ed892bc6d5dea0d4c8c832aac10d28c0","subject":"Implement sharing of zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cc3882f3ed892bc6d5dea0d4c8c832aac10d28c0"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-05 09:24:24.000000000","tz":0},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0bb26071018670caf861d9e533d32ff3317c8d00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0bb26071018670caf861d9e533d32ff3317c8d00"}]},"branch":"refs/heads/master","description":"Rebase"},"8078b64d26f9429d552e4833514aae6a804c0877":{"kind":"REWORK","_number":17,"created":"2023-02-05 21:43:40.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/17"}}},"commit":{"parents":[{"commit":"cc3882f3ed892bc6d5dea0d4c8c832aac10d28c0","subject":"Implement sharing of zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cc3882f3ed892bc6d5dea0d4c8c832aac10d28c0"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-05 21:43:29.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8078b64d26f9429d552e4833514aae6a804c0877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8078b64d26f9429d552e4833514aae6a804c0877"}]},"branch":"refs/heads/master"},"bc52fe776188817c7b3768588ec09f2c5c699bd3":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2023-02-07 02:56:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/18"}}},"commit":{"parents":[{"commit":"26d76bea0263c710d3945bafa1e220cb0298b043","subject":"Implement sharing of zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/26d76bea0263c710d3945bafa1e220cb0298b043"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-07 02:56:20.000000000","tz":0},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bc52fe776188817c7b3768588ec09f2c5c699bd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bc52fe776188817c7b3768588ec09f2c5c699bd3"}]},"branch":"refs/heads/master","description":"Rebase"},"6570aeab2a46f229c740157f01fdaa4d78264a51":{"kind":"REWORK","_number":19,"created":"2023-02-07 04:02:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/19"}}},"commit":{"parents":[{"commit":"26d76bea0263c710d3945bafa1e220cb0298b043","subject":"Implement sharing of zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/26d76bea0263c710d3945bafa1e220cb0298b043"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-07 04:02:22.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6570aeab2a46f229c740157f01fdaa4d78264a51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6570aeab2a46f229c740157f01fdaa4d78264a51"}]},"branch":"refs/heads/master"},"563853d912ba9c5943354ea59f74aab423b447c8":{"kind":"REWORK","_number":20,"created":"2023-02-07 04:45:36.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/20"}}},"commit":{"parents":[{"commit":"26d76bea0263c710d3945bafa1e220cb0298b043","subject":"Implement sharing of zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/26d76bea0263c710d3945bafa1e220cb0298b043"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-07 04:45:26.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/563853d912ba9c5943354ea59f74aab423b447c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/563853d912ba9c5943354ea59f74aab423b447c8"}]},"branch":"refs/heads/master"},"01c9671f6a07c2678d89810efe4ff26f86249716":{"kind":"REWORK","_number":21,"created":"2023-02-07 06:32:09.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/21"}}},"commit":{"parents":[{"commit":"26d76bea0263c710d3945bafa1e220cb0298b043","subject":"Implement sharing of zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/26d76bea0263c710d3945bafa1e220cb0298b043"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-07 06:32:01.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/01c9671f6a07c2678d89810efe4ff26f86249716"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/01c9671f6a07c2678d89810efe4ff26f86249716"}]},"branch":"refs/heads/master"},"5e03d087d4b0d8380ecbb8d2018f947b05c75985":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2023-02-16 06:07:52.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/22"}}},"commit":{"parents":[{"commit":"60edc59ff765b406e4b936deb4d200a2d9b411ce","subject":"Merge \"Implement sharing of zones\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/60edc59ff765b406e4b936deb4d200a2d9b411ce"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-16 06:07:52.000000000","tz":0},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e03d087d4b0d8380ecbb8d2018f947b05c75985"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e03d087d4b0d8380ecbb8d2018f947b05c75985"}]},"branch":"refs/heads/master","description":"Rebase"},"45b05ea234856f4995a47f823be17cfd5c039ce3":{"kind":"REWORK","_number":23,"created":"2023-02-17 06:57:15.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/23"}}},"commit":{"parents":[{"commit":"60edc59ff765b406e4b936deb4d200a2d9b411ce","subject":"Merge \"Implement sharing of zones\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/60edc59ff765b406e4b936deb4d200a2d9b411ce"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-17 06:57:07.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/45b05ea234856f4995a47f823be17cfd5c039ce3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/45b05ea234856f4995a47f823be17cfd5c039ce3"}]},"branch":"refs/heads/master"},"d5cc419e5060692492d03d60224d5bc954314f4e":{"kind":"REWORK","_number":24,"created":"2023-02-17 17:20:38.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/24"}}},"commit":{"parents":[{"commit":"60edc59ff765b406e4b936deb4d200a2d9b411ce","subject":"Merge \"Implement sharing of zones\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/60edc59ff765b406e4b936deb4d200a2d9b411ce"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-17 17:20:31.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d5cc419e5060692492d03d60224d5bc954314f4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d5cc419e5060692492d03d60224d5bc954314f4e"}]},"branch":"refs/heads/master"},"cd4a38ea96f1ef261ccdf26d858b6830946ca734":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2023-02-17 21:56:40.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/25"}}},"commit":{"parents":[{"commit":"409d6651281be91741a7796075201548888bd432","subject":"Merge \"Enable fail-fast on the gate queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/409d6651281be91741a7796075201548888bd432"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-17 21:56:40.000000000","tz":0},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cd4a38ea96f1ef261ccdf26d858b6830946ca734"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cd4a38ea96f1ef261ccdf26d858b6830946ca734"}]},"branch":"refs/heads/master","description":"Rebase"},"862fe761311c0d744644bb8703b8474596a5b703":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2023-02-17 23:30:35.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/26"}}},"commit":{"parents":[{"commit":"19f26d92b845144d68171f2f3d4eda4729198a9f","subject":"Merge \"Remove unused werkzeug requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19f26d92b845144d68171f2f3d4eda4729198a9f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-17 23:30:35.000000000","tz":0},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/862fe761311c0d744644bb8703b8474596a5b703"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/862fe761311c0d744644bb8703b8474596a5b703"}]},"branch":"refs/heads/master","description":"Rebase"},"6a03388cc50931ccaeda9c0a504a0e62d981203a":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2023-02-19 02:55:24.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/27"}}},"commit":{"parents":[{"commit":"fbfb1b0dd80bbcc2a6d1f09259972657d66fd998","subject":"[CI] Reduce workers and disable cinder to save on memory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fbfb1b0dd80bbcc2a6d1f09259972657d66fd998"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-19 02:55:24.000000000","tz":0},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6a03388cc50931ccaeda9c0a504a0e62d981203a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6a03388cc50931ccaeda9c0a504a0e62d981203a"}]},"branch":"refs/heads/master","description":"Rebase"},"e94da892fbf48a348ca9f026e70cbf5aecb8c030":{"kind":"REWORK","_number":28,"created":"2023-02-20 06:33:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/28"}}},"commit":{"parents":[{"commit":"fbfb1b0dd80bbcc2a6d1f09259972657d66fd998","subject":"[CI] Reduce workers and disable cinder to save on memory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fbfb1b0dd80bbcc2a6d1f09259972657d66fd998"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-20 06:33:37.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e94da892fbf48a348ca9f026e70cbf5aecb8c030"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e94da892fbf48a348ca9f026e70cbf5aecb8c030"}]},"branch":"refs/heads/master"},"9f42296df8888dc2056bc645407f98c524a1ff86":{"kind":"REWORK","_number":29,"created":"2023-02-20 06:49:42.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/29"}}},"commit":{"parents":[{"commit":"fbfb1b0dd80bbcc2a6d1f09259972657d66fd998","subject":"[CI] Reduce workers and disable cinder to save on memory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fbfb1b0dd80bbcc2a6d1f09259972657d66fd998"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-20 06:49:31.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9f42296df8888dc2056bc645407f98c524a1ff86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9f42296df8888dc2056bc645407f98c524a1ff86"}]},"branch":"refs/heads/master"},"d424d154fdcba12e227f78aa2dca07a19b1e8031":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2023-02-23 05:25:19.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/30"}}},"commit":{"parents":[{"commit":"8910a2014be52808cb3f4a42953b2176e852d0ad","subject":"Add hacking check for line continuation backslash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8910a2014be52808cb3f4a42953b2176e852d0ad"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-23 05:25:19.000000000","tz":0},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d424d154fdcba12e227f78aa2dca07a19b1e8031"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d424d154fdcba12e227f78aa2dca07a19b1e8031"}]},"branch":"refs/heads/master","description":"Rebase"},"4102fe1f1815236d2a888402737387c20fc6e52e":{"kind":"REWORK","_number":31,"created":"2023-02-23 18:00:31.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/31"}}},"commit":{"parents":[{"commit":"8910a2014be52808cb3f4a42953b2176e852d0ad","subject":"Add hacking check for line continuation backslash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8910a2014be52808cb3f4a42953b2176e852d0ad"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-23 17:59:24.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4102fe1f1815236d2a888402737387c20fc6e52e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4102fe1f1815236d2a888402737387c20fc6e52e"}]},"branch":"refs/heads/master"},"104c3416cda1e244e3d8aeb45e7f0e31c3b74802":{"kind":"REWORK","_number":32,"created":"2023-02-26 03:45:06.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/32"}}},"commit":{"parents":[{"commit":"8910a2014be52808cb3f4a42953b2176e852d0ad","subject":"Add hacking check for line continuation backslash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8910a2014be52808cb3f4a42953b2176e852d0ad"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-26 03:44:52.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/104c3416cda1e244e3d8aeb45e7f0e31c3b74802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/104c3416cda1e244e3d8aeb45e7f0e31c3b74802"}]},"branch":"refs/heads/master"},"b7c7069d5fc00b99e1af8514a4fccfd372d56c4b":{"kind":"REWORK","_number":33,"created":"2023-02-26 03:46:17.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/33"}}},"commit":{"parents":[{"commit":"8910a2014be52808cb3f4a42953b2176e852d0ad","subject":"Add hacking check for line continuation backslash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8910a2014be52808cb3f4a42953b2176e852d0ad"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-26 03:46:11.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b7c7069d5fc00b99e1af8514a4fccfd372d56c4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b7c7069d5fc00b99e1af8514a4fccfd372d56c4b"}]},"branch":"refs/heads/master"},"d4ec8f9311fccd7bc98ba61525d25eb8a44a7b9f":{"kind":"REWORK","_number":34,"created":"2023-02-26 04:04:42.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/34"}}},"commit":{"parents":[{"commit":"8910a2014be52808cb3f4a42953b2176e852d0ad","subject":"Add hacking check for line continuation backslash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8910a2014be52808cb3f4a42953b2176e852d0ad"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-26 04:04:36.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d4ec8f9311fccd7bc98ba61525d25eb8a44a7b9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d4ec8f9311fccd7bc98ba61525d25eb8a44a7b9f"}]},"branch":"refs/heads/master"},"4fbc2c8823ec432b617ce0c4e06b6ac57ae60d08":{"kind":"REWORK","_number":35,"created":"2023-03-12 05:45:19.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/35"}}},"commit":{"parents":[{"commit":"05a112177d491ed39e2a5c3001719ad05ffe16fc","subject":"Update for sqlalchemy v2.x compatibility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05a112177d491ed39e2a5c3001719ad05ffe16fc"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-03-12 05:45:05.000000000","tz":-480},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4fbc2c8823ec432b617ce0c4e06b6ac57ae60d08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4fbc2c8823ec432b617ce0c4e06b6ac57ae60d08"}]},"branch":"refs/heads/master"},"3401043fc15252dc212f6bfaa38bcd0f1a74eb7f":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2023-03-20 23:00:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/36"}}},"commit":{"parents":[{"commit":"5703810a40381d188b5d90fc68d7026c9b1c1150","subject":"Merge \"Update for sqlalchemy v2.x compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5703810a40381d188b5d90fc68d7026c9b1c1150"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-03-20 23:00:22.000000000","tz":0},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3401043fc15252dc212f6bfaa38bcd0f1a74eb7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3401043fc15252dc212f6bfaa38bcd0f1a74eb7f"}]},"branch":"refs/heads/master","description":"Rebase"},"fb9c2da977a36afc695b1e75a81aa85d8e56ef0a":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2023-04-04 03:48:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/871255/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/871255/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/871255/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/871255/37"}}},"commit":{"parents":[{"commit":"0f6a837a23f218158e487f99219c4a63d80696cb","subject":"Merge \"Update designate for RBAC \"direction change\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0f6a837a23f218158e487f99219c4a63d80696cb"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-01-22 02:13:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-04 03:48:34.000000000","tz":0},"subject":"Move to a batch model for incrementing serial","message":"Move to a batch model for incrementing serial\n\nThis patch moves the responsibility of incrementing the\nserial on a zone from central to the producer. This also\nmeans that NOTIFY is triggered by the producer after the\nserial has been incremented. The advantage of this approach\nis that we can now batch requests which means less work\nfor the DNS servers, and it removes the risk of\nrace-conditions when updating the serial. Finally, the\nproducer is sharded and is easy to scale which means that\nthis approach should scale well with many zones.\n\nThe disadvantage is that it may take up to 5 seconds longer\nfor the DNS record to be updated on the DNS server. This\ncan be lowered by increasing the frequency of the task\nthat is responsible for incrementing the serial.\n\nDepends-On: https://review.opendev.org/#/c/871266/\nChange-Id: I5e9733abaaa40c874e1d80d7b57e563df0f12cee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fb9c2da977a36afc695b1e75a81aa85d8e56ef0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fb9c2da977a36afc695b1e75a81aa85d8e56ef0a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
