)]}'
{"id":"openstack%2Fdesignate~872345","triplet_id":"openstack%2Fdesignate~master~I1de9b6cc16be2896136c7e99cf150c4b8e4603e9","project":"openstack/designate","branch":"master","topic":"bug/2004421","attention_set":{},"removed_from_attention_set":{"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2023-02-01 03:59:56.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-02-01 03:59:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1de9b6cc16be2896136c7e99cf150c4b8e4603e9","subject":"Add oslo.policy.enforcer entry point","status":"MERGED","created":"2023-01-31 18:58:24.000000000","updated":"2023-02-01 04:01:11.000000000","submitted":"2023-02-01 03:59:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872345-bug/2004421","meta_rev_id":"dc176e1eb79b1793168b9d90a2e3ad10b4af0bfa","_number":872345,"virtual_id_number":872345,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-01 03:59:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-01-31 19:05:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-01 02:37:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-01 02:37:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-31 19:05:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-01-31 20:30:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-01 02:37:22.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"f5d9859b2366954b7f68fef0a6fb9b14f1d7ae47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2023-01-31 18:58:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4367ddcbf99e4b1c90d26af45af0167290dd527","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-31 19:05:25.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"438254d3aaceeadd4113e2b20917ae69ad13ce0a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-01-31 19:05:59.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3818a7f87d30cdc75124f57472a487d1439e56d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 20:30:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e56fc92e37204e3c9002f4bfd63f78f7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/053d2e16b8aa4396810a6b327a374797 : SUCCESS in 2m 05s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/81c3836c261b440ea9d4c532fae74045 : SUCCESS in 51m 03s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/78f322f288714f1eba7bf2920a5b5051 : SUCCESS in 1h 13m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/89dcacad49aa48b88929b71aa2bde256 : SUCCESS in 1h 08m 59s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/fdb2666aceac451494ae593d118632c0 : SUCCESS in 55m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/68987c9ab5da4603869fa629b960be85 : SUCCESS in 58m 07s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/459f8095acd64490b0467c7f521b518d : SUCCESS in 1h 16m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bc91cc33b54443aeb0a5e7b06c113678 : SUCCESS in 1h 10m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a1c80a0698ee4782a889d44f35ff928b : SUCCESS in 58m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/27c865d690144b6898d1d36141bef398 : SUCCESS in 55m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/005643db1eba4b5fb35ace7b2e3d8d19 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ce550da25b5400cab4d4e0ade5be188 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8620cba7abd24c93852cfb42d6126672 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3eca2defbcfd4087921b87b075961601 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c5f206399a8472da9e1fc237f825edc : SUCCESS in 6m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d7a85b2ad4b44b4682b6d608b384982b : SUCCESS in 26m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/fd15f8b2eefc496c95f03cacf46d32d2 : SUCCESS in 7m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0e37163fc5c4ed992c8909fd086eb6b875f56034","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-01 02:37:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e107360e055eae45d7184028bd48eeb99338c7ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 02:38:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7feb1fe4806ff859655b57ed324077320695d199","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 03:59:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/675740e281f742acb8a034d7f6adecbf\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aa3f91395551439f87257ff5046a991e : SUCCESS in 2m 44s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2e118313559e444e89d38ccd1ec2217f : SUCCESS in 42m 42s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9dda90e7b49246dda8656e8b65c099ea : SUCCESS in 42m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8ed6feb686d448c28b8749f8ad2ab8c5 : SUCCESS in 46m 42s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b0ff6366cc26402da04940b0f7161805 : SUCCESS in 59m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/731d623b945e4912b68ea6e77bf0194a : SUCCESS in 41m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9e105b1f1f4a400d807c4db6a131ebf6 : SUCCESS in 45m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e5e25f9a7f3457ea3d069aae00ce978 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9adfcc28ec8448c09d2f1a0667df8c05 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a077e1185a846dbb996c1a6b6ee1a7a : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e99b4dcd27ca489a9dbd682895831b5e : SUCCESS in 8m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/03ef90bcb8164ed9a5906e59933e3ccb : SUCCESS in 26m 48s","accounts_in_message":[],"_revision_number":2},{"id":"33c63d7bd1655c33f10a99e06aea4d597c71bcde","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 03:59:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dc176e1eb79b1793168b9d90a2e3ad10b4af0bfa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 04:01:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29191fb41f964974a3f84c5ac6473d76\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baa4266f2ab542dd9da8ea3c306d63a8 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6389eb4c6858e7bd78bc65eb71c6ecb08d310acb","revisions":{"3cbb1e934a8f70d688a587d3d00eeea3cda022a9":{"kind":"REWORK","_number":1,"created":"2023-01-31 18:58:24.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/45/872345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/45/872345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/872345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/872345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/872345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/45/872345/1"}}},"commit":{"parents":[{"commit":"a9cf887da11f18dbda7c6c1332d430680a9edc14","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a9cf887da11f18dbda7c6c1332d430680a9edc14"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-01-31 18:56:45.000000000","tz":120},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-01-31 18:56:45.000000000","tz":120},"subject":"Add oslo.policy.enforcer entry point","message":"Add oslo.policy.enforcer entry point\n\nthis will allow usage of many oslo.policy scripts, e.g.\nto generate current effective policy, or to list redundant polices\ndefined in policy files.\n\nChange-Id: I1de9b6cc16be2896136c7e99cf150c4b8e4603e9\nCloses-Bug: #2004421\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3cbb1e934a8f70d688a587d3d00eeea3cda022a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3cbb1e934a8f70d688a587d3d00eeea3cda022a9"}]},"branch":"refs/heads/master"},"6389eb4c6858e7bd78bc65eb71c6ecb08d310acb":{"kind":"REWORK","_number":2,"created":"2023-01-31 19:05:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/45/872345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/45/872345/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/872345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/872345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/872345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/45/872345/2"}}},"commit":{"parents":[{"commit":"a9cf887da11f18dbda7c6c1332d430680a9edc14","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a9cf887da11f18dbda7c6c1332d430680a9edc14"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-01-31 18:56:45.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-01-31 19:05:25.000000000","tz":0},"subject":"Add oslo.policy.enforcer entry point","message":"Add oslo.policy.enforcer entry point\n\nthis will allow usage of many oslo.policy scripts, e.g.\nto generate current effective policy, or to list redundant polices\ndefined in policy files.\n\nChange-Id: I1de9b6cc16be2896136c7e99cf150c4b8e4603e9\nCloses-Bug: #2004421\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6389eb4c6858e7bd78bc65eb71c6ecb08d310acb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6389eb4c6858e7bd78bc65eb71c6ecb08d310acb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
