)]}'
{"id":"openstack%2Fdesignate~872716","triplet_id":"openstack%2Fdesignate~master~I3a90243dc323a2815ce7d973a5e52248ed1c2ad6","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-02-10 23:42:38.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I3a90243dc323a2815ce7d973a5e52248ed1c2ad6","subject":"Deprecate the agent framework and drivers","status":"MERGED","created":"2023-02-05 00:29:57.000000000","updated":"2023-02-11 00:56:13.000000000","submitted":"2023-02-11 00:54:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872716","meta_rev_id":"f1ac47cfff7c5323b89abc6510e1a03d0ffd88c0","_number":872716,"virtual_id_number":872716,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-11 00:54:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-02-06 16:05:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2023-02-10 23:42:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-05 03:45:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2023-02-10 23:42:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-05 01:42:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-05 03:45:13.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2023-02-06 16:05:35.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"4585a1764104fe773a721b5117d3e648b18a2305","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-05 00:29:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a57f673d5137c0cba4921b1418b73fe6426afdf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-05 01:42:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/979ffbf303544161941d9a93bb7a748a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/287dcbb2780e468db3031b25f86ca687 : SUCCESS in 46m 26s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/10dd994409aa4ced93f69e0deb67d0ba : SUCCESS in 1h 05m 41s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2bdddbc7cbf43b3ab69ee2f5f818e5d : SUCCESS in 52m 13s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/a5008fe7c9f847d095c30795da53a6aa : SUCCESS in 49m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4fb9320bd5f94d4eb6e0fc344e6bb0e5 : SUCCESS in 1h 05m 05s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b036fd7d8d0a43488f3e9c61d10dac02 : SUCCESS in 1h 04m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/75384f7b953f48c7b5f29a789b1211ae : SUCCESS in 1h 00m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a58ebdfdc264411dbe447a9f0b8e6ac3 : SUCCESS in 45m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f0d5180a32aa4fe1a785aaf97f473fcd : SUCCESS in 43m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84652aedf598475098fb1f3400f72efc : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48b4bdaca5eb4dfc972396e88db31d74 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21e6d9fa454d4dc69fa3eab6dd4c5ab0 : SUCCESS in 7m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9965687a4ab14b5583f98a150436f385 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52ff0531f0ce4a099b4ebbafdcb9017d : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/089291f3aa7c4f0d813b7448d72eabd5 : SUCCESS in 4m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/74428fc737264ee18b9e29be4105e7b6 : SUCCESS in 29m 13s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d1e1a4bd1f994c8fa010a13c9be8191d : SUCCESS in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72fd2bd6e69e707e22c8d93349532605f8229450","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-05 03:45:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"85a4b5736e5a499335f1e0aeaa4eefc7dbe8f09d","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-02-06 16:05:35.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95928ed13f433190aa8dc70ce73b66fbca69d7a9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-10 23:42:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"23685bae3518bd66117db53a5062d1dce3ddd0e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 23:43:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd6c23e3e492136cd5a9e9023112ca262e324d9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-11 00:54:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f23d7e0562d4fde97bbedb0d09c1daf\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/44eae0d34c40483ab74b3cb47245e943 : SUCCESS in 43m 47s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c24e3f58e63b419a94f504cf6a65abe6 : SUCCESS in 44m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6d601aa1b44c4942b1a89b08d6d67e6e : SUCCESS in 48m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/233a57048df147a39060c38883ee8195 : SUCCESS in 1h 06m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e72caacb792446b5bdca113e7236b496 : SUCCESS in 27m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a4c02e0296b64acb8ffb810cb5387f0e : SUCCESS in 32m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a690d777c7b24dc6b686f7482366fb73 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca0dc1ee97d64b9798158c25193cd594 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/facb5f5d8b0647a6845261d2af7ed05b : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e086bd0908c40f1bf0e14c7ee24db6e : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d42918db0cd54f2d9a4f3ed1af5ea080 : SUCCESS in 3m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/532c05a4a01f47a894552e933853ac8d : SUCCESS in 25m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a8a599bb48ef8414d8de7fab397de62f5aa369c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-11 00:54:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f1ac47cfff7c5323b89abc6510e1a03d0ffd88c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-11 00:56:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd4fd58250d04bc5af787ccb6f9abcf3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce9c161992de46f086d9f90e8f85850f : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a856b0a11b740848b71902fd7877b24 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63ca38975575f5b1eefe856ff1662a09b1e06f09","revisions":{"63ca38975575f5b1eefe856ff1662a09b1e06f09":{"kind":"REWORK","_number":1,"created":"2023-02-05 00:29:57.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/16/872716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/872716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/872716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/872716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/872716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/872716/1"}}},"commit":{"parents":[{"commit":"0e58c3d75729183f266fb9a7814d76eafe0ee3fa","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0e58c3d75729183f266fb9a7814d76eafe0ee3fa"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-02-05 00:28:23.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-02-05 00:29:55.000000000","tz":0},"subject":"Deprecate the agent framework and drivers","message":"Deprecate the agent framework and drivers\n\nThis patch marks the agent framework and drivers as deprecated in the Antelope release for removal in the \"C\" release.\n\nPlease see the announcement on the OpenStack discuss mailing list for more information about the deprecation:\nhttps://lists.openstack.org/pipermail/openstack-discuss/2023-January/031803.html\n\nChange-Id: I3a90243dc323a2815ce7d973a5e52248ed1c2ad6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/63ca38975575f5b1eefe856ff1662a09b1e06f09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/63ca38975575f5b1eefe856ff1662a09b1e06f09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
