)]}'
{"id":"openstack%2Fdesignate~873285","triplet_id":"openstack%2Fdesignate~stable%2Fxena~I31a1b1892a7112018b1261f956f51ed4eaa7cfb9","project":"openstack/designate","branch":"stable/xena","topic":"bug_fix-stable/xena","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-02-15 23:52:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I31a1b1892a7112018b1261f956f51ed4eaa7cfb9","subject":"Fix dns.resolver import for mdns","status":"MERGED","created":"2023-02-11 03:32:29.000000000","updated":"2023-02-15 23:55:48.000000000","submitted":"2023-02-15 23:52:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873285-bug_fix-stable/xena","cherry_pick_of_change":873070,"cherry_pick_of_patch_set":2,"meta_rev_id":"e61c900b161b37b715ccbb48b1ef36fccac89394","_number":873285,"virtual_id_number":873285,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-15 23:52:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-02-13 17:51:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2023-02-15 22:52:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2023-02-15 22:52:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-11 04:46:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 17:51:49.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2023-02-15 22:52:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"592a877c621990282d00d734ac2a64d0bf0b39fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-11 03:32:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"02994977859532123880ba1f6483a1436d5a5829","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-11 04:46:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8311450d35b4932aa52a45cfc32adec\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/533c6a5bb3714de2bf0c4cf5c2c4f886 : SUCCESS in 1h 02m 35s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/b520edde2a4645b6aa9d7dc47caa820e : RETRY_LIMIT in 11m 09s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e4a549a9a3db4eb79462df54973113f1 : SUCCESS in 50m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cf10a61425374446be44fb2d2bc13d2d : SUCCESS in 53m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/16600f631b244ae9af06d10034837f2f : SUCCESS in 1h 02m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/be807194b626400da8e6c33aa3397cf0 : SUCCESS in 44m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ed4ecd4d7904488fad91bb60bac813ad : SUCCESS in 50m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f3d90c4ea6041ccab73fdf2a14788fb : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84eaa8d841f146fd9b85851d40fbb925 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67188d4159b04fc9b9ffd17a147ad6aa : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bc332864ede4a249a8c3d34945b1ebf : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f053676b8e27432897facfa604bf0436 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff2ecbfcc8fc45e3b084eb7ecc6703f4 : SUCCESS in 6m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d7024b039a084db0bb1d7a5335318411 : SUCCESS in 26m 42s","accounts_in_message":[],"_revision_number":1},{"id":"33d38b3f08f0950e8188613f4990905c163bf70b","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-02-13 17:51:49.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f8a12a118e9ff3a299583e32ecb15fd0342898b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-15 22:52:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"225d367b83e320544207544ec2b63c79e2dcdd0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 22:53:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a094185468302018563a71c1e47713304539fe4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 23:52:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83c834ab298a4ef7b4847a8f00a173cf\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/995a0ed9a8e3477c9e994a5c8e8b9470 : SUCCESS in 34m 32s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/5014a26dff9f453e9a50d2b0a288d647 : SUCCESS in 50m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9c31660e734744c699b6c0173951cd90 : SUCCESS in 37m 32s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e293eda892124ad1a289d3b79fa39380 : SUCCESS in 57m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/77c9c9295e094cfbb19216a96ca02ae4 : SUCCESS in 52m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/598715fd1cb64f029b49375a83fa11c2 : SUCCESS in 48m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddf6a2cd79784a21b1de5dfa75f92fa8 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bcaf72b20494fc99e7aad27cf7c245b : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/046b6444c0ab4539ba0473711e275409 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beb8d1272f6745f1bd6edbff6bcb5cfe : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"94c7a65e1cd4e275c3a28dbb91ffd491eaefae39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 23:52:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e61c900b161b37b715ccbb48b1ef36fccac89394","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 23:55:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8baf2180f5464bfc87e1a10c3682283e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b70cb7efb434866b4bca35bafe2c341 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b94cf3ca15c6521386978dfd6504d614e880e1f","revisions":{"2b94cf3ca15c6521386978dfd6504d614e880e1f":{"kind":"REWORK","_number":1,"created":"2023-02-11 03:32:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/85/873285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/85/873285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/85/873285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/85/873285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/85/873285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/85/873285/1"}}},"commit":{"parents":[{"commit":"019b1ae9cb54f06f7b754024687e9f0fdfad9534","subject":"CI: Move queue configuration into project-template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/019b1ae9cb54f06f7b754024687e9f0fdfad9534"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-08 03:20:31.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-11 03:32:29.000000000","tz":0},"subject":"Fix dns.resolver import for mdns","message":"Fix dns.resolver import for mdns\n\nWe are currently failing with module \u0027dns\u0027 has no\nattribute \u0027resolver\u0027.\n\nCloses-Bug: #2006537\nChange-Id: I31a1b1892a7112018b1261f956f51ed4eaa7cfb9\n(cherry picked from commit 5cd63481c68ca5d2c8764a93dee5cbb8b45ecf63)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b94cf3ca15c6521386978dfd6504d614e880e1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b94cf3ca15c6521386978dfd6504d614e880e1f"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
