)]}'
{"id":"openstack%2Fdesignate~874991","triplet_id":"openstack%2Fdesignate~master~I0198f38afe3d5c32ea06d9e674ab0ff849f360e6","project":"openstack/designate","branch":"master","topic":"bug/1999774","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-03-31 21:27:35.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-03-24 20:38:41.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-03-31 21:27:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0198f38afe3d5c32ea06d9e674ab0ff849f360e6","subject":"Use SDK instead of neutronclient","status":"MERGED","created":"2023-02-24 08:54:45.000000000","updated":"2023-03-31 21:28:41.000000000","submitted":"2023-03-31 21:27:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"874991-bug/1999774","meta_rev_id":"d0f22c6909cbc1e131fd31cc23d2b6608db72748","_number":874991,"virtual_id_number":874991,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-31 21:27:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2023-03-27 15:03:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2023-03-25 06:06:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-28 15:10:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":2,"date":"2023-03-31 20:13:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-03-26 00:41:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":2,"date":"2023-03-25 06:06:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":1,"date":"2023-03-31 20:13:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-24 09:55:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-22 22:38:09.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2023-03-23 00:47:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-03-24 17:25:09.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2023-03-26 00:41:53.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2023-03-27 15:03:14.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2023-03-28 15:10:00.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"e7157fb614b48ebfa23571e220f85ef8e4d98add","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-24 08:54:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d06be73e165efeddeeac7b0d136c227ce4e9c699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 09:55:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1ce4e3ee5524858b7f953efe50da8b0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ce8654cec30946068b9aecb7265fb426 : FAILURE in 42m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4f7e733265d5450f8fcf21a113e0c310 : FAILURE in 52m 37s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/642729648f6547528b6780035502b8af : FAILURE in 42m 14s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4092ae46971141c5a14bb35805cc74fd : FAILURE in 43m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/aee0482261a94045a2bfffa0cbbdc419 : FAILURE in 37m 38s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a934d9d55f0d4facb4fd824c317d6b82 : FAILURE in 57m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2eb28ae03f8340a3bf0a2ebd6849ecb4 : FAILURE in 58m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ab17c4e10e34496fa6ef20cda8b7d50f : FAILURE in 39m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/307073889c214c9ea94ac4c1036895fc : FAILURE in 41m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d456400afd8545f5a0607552e8bb141a : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb60aece6df45bea800f8ef81b1950a : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1af2690146294a239731079050c4ee67 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a42ea6a8c58490aba1da161875f5adf : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/233f1abf7bee450d801b1a326d54fc30 : SUCCESS in 7m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/416ebfdbdf094699a368627c2239b195 : SUCCESS in 25m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4ea405e5d6cf4699bf9f96ee1077db57 : SUCCESS in 5m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"465ff57e9004781ce889c077c75af400104fed97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-24 10:25:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"80a3c604a32959d186174005d6bc3dd9a40d135c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 11:49:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ce18147920b4ac4b7de64aa27af688a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e80a056948c740b6a89218d7247f7d6c : FAILURE in 45m 43s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/bcc74743c9574a54afd8f191ee5ff004 : FAILURE in 1h 01m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/22d6a14c92f847f384fd23142592b437 : FAILURE in 53m 15s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/73c63614c70543a4970ccc39b9c4aa4c : FAILURE in 42m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f824b89ee60f460b96d3160b5ffd7501 : FAILURE in 42m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7ac48e3e6ce34105850ec7f9d2269de9 : FAILURE in 1h 06m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1337711fe5e4485eb39d5e4e22c22cd1 : FAILURE in 1h 02m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f404e505b02145f0b9cb493a1550e195 : FAILURE in 46m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6d3d2b0c725b419bb175561d89921b0c : FAILURE in 48m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dcde5e956424671bdcf473d0aea2a03 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cef84a85e2604f218a75123131239e2e : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1376677d5b2c487bb040b347f314afb1 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52fb91201dfa413bb8f16b4c83611662 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d96c943db7764956be56d0f19af15018 : SUCCESS in 7m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/04af503823ec4fac86994a89f14be752 : SUCCESS in 35m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/14744c1e552f419283fed33f7689ec38 : SUCCESS in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6001691e9d650d9498ba44495c7e5f5e504c2404","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-28 18:21:44.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"5bc5532e74146b6777c9e433d8a4b3dae46c4117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 19:39:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/19fc34badbb9421da6ab3b20062bb7cb\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b57fe45b202c4e9d8d11f982ccf672e6 : FAILURE in 46m 33s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4e0c2a6cef034e89b3d0ef31ad74d614 : FAILURE in 1h 07m 52s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8274f93ee8c049b0a5ca6c9f14204d5f : FAILURE in 56m 46s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/54b12ccb97a1446892e0b5fa13702bc1 : FAILURE in 47m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/aa043b7f18444c339917051564763acd : FAILURE in 38m 30s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ad51c1d9f7b04f90b88f2ae650ed94ce : FAILURE in 1h 00m 46s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/489b88e58e6745f69f99b37653b9834d : FAILURE in 1h 00m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0b1947bd445d4559b1c3dffdba105b67 : FAILURE in 39m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b510349eb15f469992400c45a54424f8 : FAILURE in 37m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9e329d791ef4f1b806c069244a635d4 : FAILURE in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d313d53d0fd54a6bb36ea27753fd1534 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b8ea42ee21c4141b7a925415d6ac5c5 : FAILURE in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89127b35b76a42499e23b21ba5242645 : FAILURE in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40b9552b77b14c89be063eb5e58e49b1 : SUCCESS in 8m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/46637e9b3fa8448aaaf376a779bb977c : SUCCESS in 26m 25s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9499b2add80b4ae183fffe48afd09ed3 : FAILURE in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"791e44b6d5185ec001f60fd760798163959e15e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-28 20:26:42.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"e7916f3f010fa0c00f91daac020d7b6dfeab01d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 21:40:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de796111ee784ce4b12045a85b455f33\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d44a1a8a9b9442a6a9cbffdf89eb5b36 : FAILURE in 47m 11s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8e17d0b34b2b43a98eb15da120db8ca1 : FAILURE in 1h 01m 31s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/530312117ab040dfb92f310046d14bec : FAILURE in 55m 41s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/00d67103afd24074a0dd69769daa1a65 : FAILURE in 49m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dabee609de734315931417fe2f6752e2 : FAILURE in 35m 49s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/845717900d81456abcb7a32fd1f91c54 : FAILURE in 1h 03m 34s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a1d31c051c26407ca4c1c699280b78b4 : FAILURE in 1h 07m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7e89b14813e0456581b158bfd6611352 : FAILURE in 50m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9164ff283311441c87558ea6960f6536 : FAILURE in 50m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11f231a877e64c3bb6422e0aa89cde95 : FAILURE in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfd3d63c2c114133ad9379cf5f6380c7 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b23a24207efc4f23ba52c0097301ed14 : FAILURE in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8fdb9c7f31574af48f55859a46b1bf2a : FAILURE in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ede3d21fcf764a4e9d80453181ffed1d : SUCCESS in 7m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5846512e9fe34ae88d19b6d40211a73d : SUCCESS in 25m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6a7ed37326b342789cec379dcb9a83c9 : FAILURE in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"623c0a626e1ae9606f546c81e80e5b605185aaaa","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-03 12:24:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e4b8557d6340f43e33fefc033cb96f9e3dffb82b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-07 16:40:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"daacd2d029d1eb9a1b610cebf32b1bca3d6025f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 17:43:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ff30845b53344108449a78ff0a01d8a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0714e4d2a64f4e6e9adc3daee3576d82 : FAILURE in 35m 26s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e7f70435f9de4c54bfae84c36f58288c : FAILURE in 55m 37s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf8ea190ec9a4e19b97b2c47e48347ac : FAILURE in 49m 13s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8417ddf784184d208f067610886e6deb : FAILURE in 51m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f6a7f0a0be404339964a5531b35f2cf4 : FAILURE in 35m 09s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bc16fdfa4fd34e19995759d2ed20795a : FAILURE in 54m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2b8584c2abfc498fbf933af834dcef8c : FAILURE in 50m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a66219f2905640ebb9301251ade21df5 : FAILURE in 33m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/99787747ab7044b6a8c97e0488fe0454 : FAILURE in 45m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f892973edf544bf904baaa4b470577b : FAILURE in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b57c79bc5d064eda8e2aac80ae89e9f5 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d22218e864b4494a855a19fd3f5c15fa : FAILURE in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d980f35b6da64c1b98e8dbf57d008fd0 : FAILURE in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebbcaecddafb4e64a01b72dc652a24e9 : SUCCESS in 9m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2cfa5d43a69d4636b449a89a5a302150 : SUCCESS in 29m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f7e18330ab174a8ca09ac025f2a5a572 : FAILURE in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9259210ab12d0e8551c8317c350d7051cb61bb5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-08 19:16:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"65c8cdfba389c64211905d9e73015052fa142c37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 20:27:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4e244bd1aec4e598e35420c7374d887\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/27e8864bdd084bf69da13663ae794189 : FAILURE in 44m 48s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0e59ab94f1d542f098fd8d3dfa9b1e53 : FAILURE in 53m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b96294fc1db47179ca325d44c0ee001 : FAILURE in 54m 06s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/816f005cd84b47938dccf0484e13bddf : FAILURE in 41m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ec5f40ffcf8e40cf9fd9223f6b49f627 : FAILURE in 54m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fa36081739a241c48b3b2b5fc9ad20cb : FAILURE in 58m 01s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e71eab74e2364718952a28a87c5d2bfa : FAILURE in 1h 05m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/01b48532eafc4382abff93bd2ccc0e90 : FAILURE in 42m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ab1a8352562a41548a0fa07e8acde5b2 : FAILURE in 46m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2883b013542b4830b2817d192999deba : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47220b3bfc784bb3b0cc44557da01e40 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9aebe91015c24fc1af696bdbf8abef81 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/320e4e29c24a4f538e1ce5ba6702918a : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a464bcfa4ca4aaca778fff2ad49bd5c : SUCCESS in 7m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dbffae467a1f4afebefd3d195e536c8b : SUCCESS in 34m 00s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/116eace997724a4990071b282f9dd1eb : SUCCESS in 4m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"52aac1c8b7b4cb12c72d5bf9ee53553c8cdecc83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-22 13:27:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e08d16ab0ffad394c1571f292b1a84ddf14cec92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 14:21:16.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5063e868377a4f04bb80d1f4696be395\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dcc7d81706c546faa9c03348f8bbf297 : FAILURE in 40m 27s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/52bc1f23db004841879b5d0d53648a9b : FAILURE in 50m 26s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/fdcdfe3eaddc4aeab1194a38b389a71d : FAILURE in 48m 55s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f5899482b6c14b009f7bf5a75702749f : FAILURE in 38m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/95dbfbee1ac94ab9928650534df9d65d : FAILURE in 42m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/dcdc7e658b0941629298bbdf45fa44b2 : FAILURE in 39m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7cf0a1950f3243d2acd712969eacf487 : FAILURE in 39m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5308e9184046468e9286d8c4bc8e4b18 : FAILURE in 39m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cd9aa38f10da4b57aba59e97f89f0823 : FAILURE in 37m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c8f5fefdd8844bd8cb5697b10bb8462 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf100bfae7b54b7e8bd9cb1893d17fd7 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8722b514fffc41338be962fedef48271 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/931bbce8ebcc4d87892153da6b3b08d2 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cef12d379fa9475fa37bd26cb3f01c84 : SUCCESS in 9m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e67592e003e548dea875d8b5b2212392 : SUCCESS in 36m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/63a47e74b65a44499ba512bc8037a58e : SUCCESS in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"95885b923831017a8cdcb8c43d1e9587bfa63899","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-22 22:38:09.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"483574465ab01a37b8ddc6265503e94ae09e2c5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-23 00:44:34.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4ca7fe3cbe07419bb39602226af3f539702a26ba","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-23 00:47:36.000000000","message":"Patch Set 8: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3675fa67d03d4c83511108d6aeb139880707891a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-23 00:49:44.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"28975b90c0b6af004a3130be13cac0c50be08e80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 01:35:09.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/634eb63974aa4462953bcbebc4e6bd1e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d5b30908b4984354abb9c745ae34f201 : SUCCESS in 44m 44s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e710f994e27741e89d47f795ff69b774 : SUCCESS in 38m 27s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/218485c5fbe648f19e47601ebd7ddbf5 : SUCCESS in 34m 45s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9db780c274d14398a6e44960ef845d8b : SUCCESS in 31m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d7f973c48804468abcf20278d23a6f07 : SUCCESS in 28m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d4fefa4e3ba04b51a32f1e629c8af442 : FAILURE in 30m 27s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/64e82bf513e343ea9433c9dd4335d299 : FAILURE in 30m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/09eafbde01204a08a3624fc0e8bc0d1b : SUCCESS in 29m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0785c718de1f476d879b82851dba9952 : SUCCESS in 31m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d10812115e348158574323335ef3010 : FAILURE in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aad8571ba854f418a872585c349e775 : FAILURE in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/747f7fa0e3934fb98ba4062cbd4a1fe4 : FAILURE in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15f46e24bafa4ddb81fcce01f38ceb28 : FAILURE in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82a31017f2d946a3aba45003ea8750fb : SUCCESS in 6m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f42cf8e3c6a248cbbb45c61b48c17f03 : SUCCESS in 27m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5b676c0461b14ae386f3b049fb726eaa : FAILURE in 3m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4db082e5d98b9f70112e4cc3219158b888d8f27f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-23 09:49:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fae4dac45aaad429cc036593c4fd923139b6fa77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-23 21:23:50.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"fd790c9acd9c986f74548e98fa5dbfacd06aa38c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-23 21:25:43.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"89d3849199e9add19a39f20f064c37b8065ddc06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 22:16:45.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a016a14e93d04ebdbf72d38859480a43\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6830020592c04f33af545854ab3b92bc : SUCCESS in 38m 28s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/15ee07bda78049d687c127fb224d6349 : SUCCESS in 50m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5471f0bd45f842e1b9821c3f30a38a76 : SUCCESS in 42m 45s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ceed25159bca46639777ae6910c59601 : SUCCESS in 36m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/382b38faa5fc429ea9f170abf13e5d0d : SUCCESS in 36m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5ae6a812311e49acbe08b7195e178df3 : FAILURE in 41m 01s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/183ee47cfd1e4da5ad2f484df52abdab : FAILURE in 37m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4099eeeb156f47319f2578e2356f753e : SUCCESS in 39m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c6a1aef19c34425093a8171837e444f5 : SUCCESS in 36m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fc9527754ad4550b4aa43b1fecdb9ad : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9d5b5a383f04e6493a0012078e13378 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1838238f53434ed1bac8a0cb78c11453 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/929254c65f5e4576a9a2245c395add2a : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/460388b8eabb4127b2068dd87a2f15ba : SUCCESS in 8m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/090d61d19895415f93f233eaa4b6ca73 : SUCCESS in 3m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1509764d1e84c709c05208d8745c8a6 : SUCCESS in 4m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4d0c816e501a4efaa251db26d016f15a : SUCCESS in 34m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5a3edaabaef54ccab5d45e86a172c9e0 : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c401e146257047b7e44023214bc521ae1f671a82","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-24 10:44:17.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ec42ed995960e912c907fc8e57bfc7c295a96306","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-24 10:51:54.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"168d5ab2b3c47e6fbcee687436ac14aad4a91e61","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-24 17:25:09.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"28ec282db5261ce4838074c6a736c507ebb02e27","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-24 20:38:41.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1b81b7939effc3735ea7fcf2fc10d1e227a4081c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-25 06:06:30.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"417c3a1de5d3bbf4b260b09260202b7b6045d95e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-25 07:20:12.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a91b874e00e442a993c839c963d3220d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2261289ec3b44137b40680de8903498c : SUCCESS in 52m 41s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/da86a78c9e85419c825496cf6aee39d8 : SUCCESS in 1h 01m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f57710e000946a4be0467916ba61c49 : SUCCESS in 51m 24s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6058665744934369a894c987192d866d : SUCCESS in 45m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dba66f5600e640dd9be46c8410aa4631 : SUCCESS in 47m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/aff797a286b94aea85e49633c42b1c46 : SUCCESS in 1h 08m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4a0efbf73696477e8778b167ac6774c7 : SUCCESS in 1h 07m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1304c6ededc5404ab3915db6a410b5cb : SUCCESS in 51m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7bf255f0e884458fbc90ce5dee868f18 : SUCCESS in 45m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bedc1e80f5164c199c099e235dad0ebd : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78b53581034543f78cbad3e077f82b33 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bc8abad33a34892a2a61af55728d7f7 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be6dd79a429a45068b1c7944a830cfd4 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e28d2af5c1e46479825fbe980bea5e9 : SUCCESS in 6m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3b242f8e59594c4586ebfc9369454e39 : SUCCESS in 2m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eaa3caa2b08541fc88cff00adb5bbbc9 : SUCCESS in 3m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d521249c14ac483091356573c64008b6 : SUCCESS in 26m 53s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2dd496140d704896bd0d6b20889788e0 : SUCCESS in 3m 44s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7f514f321a1bead3ac7f406f4ab8fdb70b6ee678","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-03-26 00:41:53.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"6f78ae22ccbd84433fc52b744f56129f0ea9de12","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-03-27 15:03:14.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0fa7c8021b8475bab9df0ad5b4eb87b6e7b73151","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-03-28 15:10:00.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e60feaa05bba84cf9f9b29a3a18856c46331ca26","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-31 20:13:45.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c45d4e9b15b8d3586740ee9ae661bd57d00a6445","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 20:14:18.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"43869872e929664a5a44a8d8a36632fd3c4f5f65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 21:27:35.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d2db63113ff436587d2431c40ce66cd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f0dcac167cd94dc38d4ee371cd297ab5 : SUCCESS in 52m 28s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/52e1a4384abb4ee8a93dd7cbd160c609 : SUCCESS in 52m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/727c7a2e62f84a489068af0f7da8e84d : SUCCESS in 44m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cab57707a00e4b5999763a4e207f8818 : SUCCESS in 1h 08m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/67beb9b82a534e92bd01e812182df86d : SUCCESS in 33m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/593d027ff663459eb7cdb223dbdb15d6 : SUCCESS in 49m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61f0d22be1cc403797da81fd25faf8da : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19c55e0a07df4a24a63068e3a6795540 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7919546bff14382a45139637a6370e8 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc7665cfefb8491b9d208bc373c3d8d6 : SUCCESS in 12m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c1b2e7727e964e9bbfa2744af98c76b9 : SUCCESS in 2m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50ca8fe8d0574ab1a560f533d913bf08 : SUCCESS in 3m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/320bacce190349a6aeff22735ebc38be : SUCCESS in 34m 44s","accounts_in_message":[],"_revision_number":10},{"id":"6e3f21a5e7aba04c4194b08e8a563d459f59ce05","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 21:27:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"d0f22c6909cbc1e131fd31cc23d2b6608db72748","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 21:28:41.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8a67ffda0f8408d929d992bb2d4c0c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/847b69c88fa0441ebc3c08fdc8e11b90 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe4fd1df9f4a468e812687743cc23267 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"fd09a0cfc392349ec3e7437c86aabef40c574cd4","revisions":{"0fab42fce344e45d185c1539e6f60b22c8c06650":{"kind":"REWORK","_number":1,"created":"2023-02-24 08:54:45.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/874991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/874991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/874991/1"}}},"commit":{"parents":[{"commit":"8910a2014be52808cb3f4a42953b2176e852d0ad","subject":"Add hacking check for line continuation backslash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8910a2014be52808cb3f4a42953b2176e852d0ad"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-22 13:20:28.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-24 08:54:19.000000000","tz":60},"subject":"WIP Use SDK instead of neutronclient","message":"WIP Use SDK instead of neutronclient\n\nChange-Id: I0198f38afe3d5c32ea06d9e674ab0ff849f360e6\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0fab42fce344e45d185c1539e6f60b22c8c06650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0fab42fce344e45d185c1539e6f60b22c8c06650"}]},"branch":"refs/heads/master"},"adb0d555a54a21fe315c392e0ec8b17528dc2a16":{"kind":"REWORK","_number":2,"created":"2023-02-24 10:25:37.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/874991/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/874991/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/874991/2"}}},"commit":{"parents":[{"commit":"8910a2014be52808cb3f4a42953b2176e852d0ad","subject":"Add hacking check for line continuation backslash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8910a2014be52808cb3f4a42953b2176e852d0ad"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-22 13:20:28.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-24 10:25:20.000000000","tz":60},"subject":"WIP Use SDK instead of neutronclient","message":"WIP Use SDK instead of neutronclient\n\nChange-Id: I0198f38afe3d5c32ea06d9e674ab0ff849f360e6\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/adb0d555a54a21fe315c392e0ec8b17528dc2a16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/adb0d555a54a21fe315c392e0ec8b17528dc2a16"}]},"branch":"refs/heads/master"},"964861d43c4dcf2e4acf90bc5182eae852726a9a":{"kind":"REWORK","_number":3,"created":"2023-02-28 18:21:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/91/874991/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/874991/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/874991/3"}}},"commit":{"parents":[{"commit":"8910a2014be52808cb3f4a42953b2176e852d0ad","subject":"Add hacking check for line continuation backslash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8910a2014be52808cb3f4a42953b2176e852d0ad"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-22 13:20:28.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-02-28 18:21:44.000000000","tz":0},"subject":"WIP Use SDK instead of neutronclient","message":"WIP Use SDK instead of neutronclient\n\nChange-Id: I0198f38afe3d5c32ea06d9e674ab0ff849f360e6\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/964861d43c4dcf2e4acf90bc5182eae852726a9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/964861d43c4dcf2e4acf90bc5182eae852726a9a"}]},"branch":"refs/heads/master"},"0783e1aec89d9b57e326eed4a0d33186f138a4a7":{"kind":"REWORK","_number":4,"created":"2023-02-28 20:26:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/91/874991/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/874991/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/874991/4"}}},"commit":{"parents":[{"commit":"8910a2014be52808cb3f4a42953b2176e852d0ad","subject":"Add hacking check for line continuation backslash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8910a2014be52808cb3f4a42953b2176e852d0ad"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-22 13:20:28.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-02-28 20:26:42.000000000","tz":0},"subject":"WIP Use SDK instead of neutronclient","message":"WIP Use SDK instead of neutronclient\n\nChange-Id: I0198f38afe3d5c32ea06d9e674ab0ff849f360e6\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0783e1aec89d9b57e326eed4a0d33186f138a4a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0783e1aec89d9b57e326eed4a0d33186f138a4a7"}]},"branch":"refs/heads/master"},"d8d7f4d4566f00c91caaa41d3012a6e87b53907e":{"kind":"REWORK","_number":5,"created":"2023-03-07 16:40:46.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/874991/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/874991/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/874991/5"}}},"commit":{"parents":[{"commit":"8910a2014be52808cb3f4a42953b2176e852d0ad","subject":"Add hacking check for line continuation backslash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8910a2014be52808cb3f4a42953b2176e852d0ad"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-22 13:20:28.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-03-07 16:40:27.000000000","tz":60},"subject":"WIP Use SDK instead of neutronclient","message":"WIP Use SDK instead of neutronclient\n\nChange-Id: I0198f38afe3d5c32ea06d9e674ab0ff849f360e6\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d8d7f4d4566f00c91caaa41d3012a6e87b53907e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d8d7f4d4566f00c91caaa41d3012a6e87b53907e"}]},"branch":"refs/heads/master"},"0b1fcaa7a35a65ca6f70f7c34e84f770c0d6562d":{"kind":"REWORK","_number":6,"created":"2023-03-08 19:16:11.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/874991/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/874991/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/874991/6"}}},"commit":{"parents":[{"commit":"8910a2014be52808cb3f4a42953b2176e852d0ad","subject":"Add hacking check for line continuation backslash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8910a2014be52808cb3f4a42953b2176e852d0ad"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-22 13:20:28.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-03-08 19:15:50.000000000","tz":60},"subject":"WIP Use SDK instead of neutronclient","message":"WIP Use SDK instead of neutronclient\n\nChange-Id: I0198f38afe3d5c32ea06d9e674ab0ff849f360e6\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0b1fcaa7a35a65ca6f70f7c34e84f770c0d6562d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0b1fcaa7a35a65ca6f70f7c34e84f770c0d6562d"}]},"branch":"refs/heads/master"},"e2fabfff965ffb769e96dfaa8ace3555508988f0":{"kind":"REWORK","_number":7,"created":"2023-03-22 13:27:50.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/874991/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/874991/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/874991/7"}}},"commit":{"parents":[{"commit":"5703810a40381d188b5d90fc68d7026c9b1c1150","subject":"Merge \"Update for sqlalchemy v2.x compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5703810a40381d188b5d90fc68d7026c9b1c1150"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-22 13:20:28.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-03-22 13:23:57.000000000","tz":60},"subject":"Use SDK instead of neutronclient","message":"Use SDK instead of neutronclient\n\nTODO:\n* Document somehwere if users need to be more careful with\nconfig.\n* Check with somebody the devstack plugin.sh and cfg part\n* I see some warnings in the logs like:\n\"The token used to make the request was project scoped\nbut the policy requires [\u0027system\u0027] scope\", I suppose I\nhave to update my environment for the new RBAC things, but\nbetter to check it in detail.\n\nChange-Id: I0198f38afe3d5c32ea06d9e674ab0ff849f360e6\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e2fabfff965ffb769e96dfaa8ace3555508988f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e2fabfff965ffb769e96dfaa8ace3555508988f0"}]},"branch":"refs/heads/master"},"b5918f770590b5ce5179237611f682f890ea3b1e":{"kind":"REWORK","_number":8,"created":"2023-03-23 00:44:34.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/91/874991/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/874991/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/874991/8"}}},"commit":{"parents":[{"commit":"5703810a40381d188b5d90fc68d7026c9b1c1150","subject":"Merge \"Update for sqlalchemy v2.x compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5703810a40381d188b5d90fc68d7026c9b1c1150"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-22 13:20:28.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-23 00:44:33.000000000","tz":0},"subject":"Use SDK instead of neutronclient","message":"Use SDK instead of neutronclient\n\nTODO:\n* Document somehwere if users need to be more careful with\nconfig.\n* Check with somebody the devstack plugin.sh and cfg part\n\nChange-Id: I0198f38afe3d5c32ea06d9e674ab0ff849f360e6\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b5918f770590b5ce5179237611f682f890ea3b1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b5918f770590b5ce5179237611f682f890ea3b1e"}]},"branch":"refs/heads/master"},"246bf27ce3387cff7cec01e73f9484cbc497b4e2":{"kind":"REWORK","_number":9,"created":"2023-03-23 21:23:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/91/874991/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/874991/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/874991/9"}}},"commit":{"parents":[{"commit":"5703810a40381d188b5d90fc68d7026c9b1c1150","subject":"Merge \"Update for sqlalchemy v2.x compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5703810a40381d188b5d90fc68d7026c9b1c1150"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-22 13:20:28.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-23 21:15:12.000000000","tz":0},"subject":"Use SDK instead of neutronclient","message":"Use SDK instead of neutronclient\n\nThe python-neutronclient has been deprecated for the CLI since Ocata and\nthe python bindings \"neutronclient\" has been deprecated for removal as\nof the 2023.1 (Antelope) release[1] in favor of using openstacksdk.\nThis patch migrates Designate from using the neutronclient to using the\nopenstacksdk for communicating with neutron.\n\n[1] https://docs.openstack.org/releasenotes/python-neutronclient/2023.1.html\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nChange-Id: I0198f38afe3d5c32ea06d9e674ab0ff849f360e6\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/246bf27ce3387cff7cec01e73f9484cbc497b4e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/246bf27ce3387cff7cec01e73f9484cbc497b4e2"}]},"branch":"refs/heads/master"},"fd09a0cfc392349ec3e7437c86aabef40c574cd4":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-03-25 06:06:30.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/874991/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/874991/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/874991/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/874991/10"}}},"commit":{"parents":[{"commit":"b8ec3b450b5242c28f19d3e8057c5573807a4954","subject":"Restart neutron on grenade client update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b8ec3b450b5242c28f19d3e8057c5573807a4954"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-22 13:20:28.000000000","tz":60},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-03-25 06:06:30.000000000","tz":0},"subject":"Use SDK instead of neutronclient","message":"Use SDK instead of neutronclient\n\nThe python-neutronclient has been deprecated for the CLI since Ocata and\nthe python bindings \"neutronclient\" has been deprecated for removal as\nof the 2023.1 (Antelope) release[1] in favor of using openstacksdk.\nThis patch migrates Designate from using the neutronclient to using the\nopenstacksdk for communicating with neutron.\n\n[1] https://docs.openstack.org/releasenotes/python-neutronclient/2023.1.html\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nChange-Id: I0198f38afe3d5c32ea06d9e674ab0ff849f360e6\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fd09a0cfc392349ec3e7437c86aabef40c574cd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fd09a0cfc392349ec3e7437c86aabef40c574cd4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
