)]}'
{"id":"openstack%2Fdesignate~877150","triplet_id":"openstack%2Fdesignate~master~I15864a76c8913f151154e2398ae86aea71883498","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-03-11 03:19:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I15864a76c8913f151154e2398ae86aea71883498","subject":"Removed unsupported test B309 from bandit","status":"MERGED","created":"2023-03-11 00:48:13.000000000","updated":"2023-03-11 03:20:33.000000000","submitted":"2023-03-11 03:19:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877150","meta_rev_id":"7d3e62878e5a8ce925edcfe4d1a99e1ea6b170cd","_number":877150,"virtual_id_number":877150,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-11 03:19:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-03-11 01:11:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-03-11 01:11:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-11 01:11:33.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-03-11 02:07:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d38faca943f68095f27c877863488c3ffbb31cf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-11 00:48:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"628b9881e70d88e9ae5f613443ffc2a0a6b189d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-11 00:51:01.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ad1f69c45fc6c9043b27f2455933ff503ca77482","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-11 01:11:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e25c7596e04fd287bb2d4f28dfb4e0e93ab251c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 02:07:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9535d6b0b6544df8a5144dba84e6cedd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f2423354e9ef4011b0d1979f2dd38469 : SUCCESS in 3m 31s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7900259504104972b008e874967e22de : SUCCESS in 33m 19s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ab0866bd8bbb4fe6bbb6662f591a8ea3 : SUCCESS in 44m 01s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/939c5d7cc73e4f6382d0c76043270e68 : SUCCESS in 38m 53s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f6ba410a28a745afa00f04df6c1f40d0 : SUCCESS in 34m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e437e7b2931d4554821815de89bf95c6 : SUCCESS in 37m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5bd48441a7ce4039bfa8ec2b047f9cb5 : SUCCESS in 47m 06s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/adaa6a874f294ffea1e7a7a8e7ee967c : SUCCESS in 1h 05m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/279f8630019a4368b2b6b31ced319c5d : SUCCESS in 1h 04m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0fc011f6d312484298827019c1f9ef2d : SUCCESS in 51m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54ac4ad684584599a6a3184a52930f81 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41175a9967a8422381cc3a0eb9cf094a : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdfcb33c2ecb4948a04bbf06ffb4995c : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a7a56d1b46e48838bc8ba3a1f5698e5 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8296ad6b7c3f4a9a86586e7c55a29e8d : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b070f564ccb4eccb30da098bb1dbb95 : SUCCESS in 3m 52s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5df2cdd436df439bb9245614fa97f92e : SUCCESS in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c90c00925dbd959aa5f2e31d48b51f34334f4911","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 02:11:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c11d0735e5a06634d7f41c446dbdbe09fbccdc19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 03:19:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a101a68551fb4d97bc6abb67e5cc3e0d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7cddaec25c584b57b49786eadf574505 : SUCCESS in 3m 21s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c71a96018bff4f8eaca4894f76c43d02 : SUCCESS in 45m 41s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4c64b83baf3c4f1e8fc3d3f513143577 : SUCCESS in 46m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b02126cba5dd4f6f8fa740389d047e0b : SUCCESS in 42m 32s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8bfd08e289014410b5d2c69d65951474 : SUCCESS in 1h 01m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c960f52399f0471bb1f87c0febb6d5b5 : SUCCESS in 43m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/10a2d683485a4323b0478b40911887d4 : SUCCESS in 49m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a9a0874184f475ca9c9ddb86a3aa459 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71985973655442afa49a737bf3ae29aa : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6de6f9260f34b82a5e456e4889b3d6c : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe47d0f45cf54a0d88efec1ea6b02ae4 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/696d7f21d3f54572b15bf81cc1e330eb : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"d8aa89816d5606d4ef77a7aa294d22f0c8f6bc2e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 03:19:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7d3e62878e5a8ce925edcfe4d1a99e1ea6b170cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 03:20:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c603a0189f01499db8c601132f2af0ff\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/23a213936ff64cf5bbae8c1baf7a7cba : SUCCESS in 51s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c8814f4f8fe427b87329c1c379002cd : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/568b90fac64b4aa49904997994da835f : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3","revisions":{"f315a2634cda260c1324a60a46804c9e7d91e990":{"kind":"REWORK","_number":1,"created":"2023-03-11 00:48:13.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/50/877150/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/50/877150/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/877150/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/877150/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/877150/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/50/877150/1"}}},"commit":{"parents":[{"commit":"dd2a945c78d01b8737e9573b7f1eab04f5a6a27a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dd2a945c78d01b8737e9573b7f1eab04f5a6a27a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-03-11 00:47:19.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-03-11 00:47:47.000000000","tz":-480},"subject":"Removed unsupported test B309 from bandit","message":"Removed unsupported test B309 from bandit\n\nChange-Id: I15864a76c8913f151154e2398ae86aea71883498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f315a2634cda260c1324a60a46804c9e7d91e990"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f315a2634cda260c1324a60a46804c9e7d91e990"}]},"branch":"refs/heads/master"},"758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-11 00:51:01.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/50/877150/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/50/877150/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/877150/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/877150/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/877150/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/50/877150/2"}}},"commit":{"parents":[{"commit":"dd2a945c78d01b8737e9573b7f1eab04f5a6a27a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dd2a945c78d01b8737e9573b7f1eab04f5a6a27a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-03-11 00:47:19.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-03-11 00:50:32.000000000","tz":-480},"subject":"Removed unsupported test B309 from bandit","message":"Removed unsupported test B309 from bandit\n\nB309 was removed with this patch and breaks the CI.\nhttps://github.com/PyCQA/bandit/commit/130a467434ff6b8f9bbd83100e409bc1fcb6f407\n\nChange-Id: I15864a76c8913f151154e2398ae86aea71883498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
