)]}'
{"id":"openstack%2Fdesignate~877164","triplet_id":"openstack%2Fdesignate~stable%2Fyoga~I0ace4d6c4ad6a6ee236e3af23805b01345d60a42","project":"openstack/designate","branch":"stable/yoga","topic":"stable/yoga/1927304","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-03-16 17:12:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0ace4d6c4ad6a6ee236e3af23805b01345d60a42","subject":"Validate MX records during recordset create or update","status":"MERGED","created":"2023-03-15 19:30:32.000000000","updated":"2023-03-16 17:13:45.000000000","submitted":"2023-03-16 17:12:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877164-stable/yoga/1927304","meta_rev_id":"3385af0a76d9f1d4b0805f19b969789a9dab0d0f","_number":877164,"virtual_id_number":877164,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-16 17:12:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2023-03-15 22:04:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":1,"date":"2023-03-15 22:04:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-16 16:03:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-16 16:03:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-15 19:30:32.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"CC"},{"updated":"2023-03-15 19:34:42.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2023-03-15 19:37:06.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-03-15 20:41:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-16 16:03:23.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"d9156308116ff10bb7c112915ae31ab1f852197d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-15 19:30:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1611b9ce743decd86622dfc639c6450015b3fda","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-03-15 19:34:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6c088b9167a8790d403363bc17b117c7dcf7cb72","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-15 19:37:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"24fa2f8bdc84bce35f2a400c650199626bd57b18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 20:41:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c7135ef52a3417a9542eb898bf303d8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7717d71d79ab4281abeb82a8259996a0 : SUCCESS in 41m 55s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/32c2ba1ece3942bf9be58448d94f47c3 : SUCCESS in 49m 30s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e63ce80d28ef4959a5efd40b882dfac1 : SUCCESS in 44m 36s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8e0f03d2e9e740bab764ed71aea969be : SUCCESS in 45m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a03b6bed4b0d4e869c09a9d51b2f2540 : SUCCESS in 44m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d16b02888eb647688364320a226b6536 : SUCCESS in 1h 04m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d7f5d53769f34f9ab996337f08e4e91b : SUCCESS in 46m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d23cb77b1fb24b6dbda3f7d0ef8d2a22 : SUCCESS in 46m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6e2a12e684c469ab38f488a4fb8186b : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15fbb3573bb449c5a4e1ab5b264b6920 : FAILURE in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9688cfcefd247609d7df10322afacf8 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d433efc91804588a9d782ae297109ab : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e98882809ff44c4394cb6c1b77739aa3 : SUCCESS in 7m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ae8cffeba1ea486eaffaa120dd179cc6 : SUCCESS in 32m 57s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/19268d9e75f342a08b29fe56f6e42446 : FAILURE in 4m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ffa9ffba9cece0ea845e769fa02f681db4ae48e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-15 22:04:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ad348efd81d873e3004b656f32a2c2db290a4d45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 23:11:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd2d5ed0e04743f2a7be1ea5ab0abc25\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c85ea76ca0d748ce9df14a380b04aec3 : SUCCESS in 36m 43s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/053f66787bde4616affa6aa70606f093 : SUCCESS in 1h 02m 31s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/24f442aa9b8f4f709bc799f1637ef343 : FAILURE in 37m 40s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0a6e97d46d45488ebc503482f00b1c75 : SUCCESS in 57m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f0ed19b12d954f27953e1b2857758615 : SUCCESS in 37m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/92e01b44778748d893b4bd8fcebd4566 : SUCCESS in 51m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f057ddfb65d44198b0ef639da0d09256 : SUCCESS in 33m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/97849a42dea146bca3bc94ed495b59b2 : SUCCESS in 44m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f234253c637540a082f878171c6c1d92 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a53dde4d49054793a7a27effbeaf7753 : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e42d9a4af1245ef8379791bb57722c2 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba5f46aed91c47c8a61a7af18be3917f : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcbf1002d6e04b6ea2555aff76e506f4 : SUCCESS in 6m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f97595a14d384f6db17e562679ed22b8 : SUCCESS in 31m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0b5211c809574c9a9583f6d62e9d9972 : FAILURE in 4m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cd589961709d341ebc7e245bc550679f894fc23d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-16 16:03:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f52918c3f64cd6d87bab8f6ae1a91c2c5a732510","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 16:04:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2362d38f152ad8488a99b33631de645885fe3154","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 17:12:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a81adb97785b433bb2f70c8bf50386bc\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/84c968fafa954f028065a512b8158771 : SUCCESS in 54m 07s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/5cb1278f3bc748498c857005f5f1c596 : SUCCESS in 51m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d2fbcb3c30f1498b9a36ecd558cdd04f : SUCCESS in 47m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/291dd5edc06a4b71ac980ac4b003e243 : SUCCESS in 1h 03m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b8422a30b04c4b459e54de62b6033e10 : SUCCESS in 51m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/66ec6e065e72449f8c84741b3d243b8d : SUCCESS in 41m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3cd038a15e74d61bbdef5bb6772d3f4 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8288c5443d764e57a9a8f799e6de0dc3 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19e06569cd794a648da1caced0770313 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab2333294aa642148904bc2a796a3ede : SUCCESS in 10m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3089fb8522d34480be718cff8f9fbfab : SUCCESS in 27m 03s","accounts_in_message":[],"_revision_number":2},{"id":"3f136dd46c0106f56b92b9c086dba29ab2ebfd89","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 17:12:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3385af0a76d9f1d4b0805f19b969789a9dab0d0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 17:13:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d29300877f15432284868855f4a211de\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a7088bf33d49b2ac51b81b35d84853 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"86a8cc5f525f6e2ab5e287c027b3bd18b9fbfbd3","revisions":{"6f04005f95fa2052c00166ddc2e0d519da645128":{"kind":"REWORK","_number":1,"created":"2023-03-15 19:30:32.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/64/877164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/64/877164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/877164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/877164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/877164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/64/877164/1"}}},"commit":{"parents":[{"commit":"1f30672bde3148f868fbaece8a58c20413e47b12","subject":"Fix dns.resolver import for mdns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f30672bde3148f868fbaece8a58c20413e47b12"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-05-06 04:56:11.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-03-15 19:16:20.000000000","tz":-420},"subject":"Validate MX records during recordset create or update","message":"Validate MX records during recordset create or update\n\nIf invalid MX record is provided during recordset create command or\nvalid record is updated to invalid record using recordset set command,\ncauses recordset creation/updation failure. This leads to zone enter\nin ERROR state and all further recordset creation calls fail. Validate\nthe records within recordset during create or update call.\n\nCloses-Bug: #1927304\nChange-Id: I0ace4d6c4ad6a6ee236e3af23805b01345d60a42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f04005f95fa2052c00166ddc2e0d519da645128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f04005f95fa2052c00166ddc2e0d519da645128"}]},"branch":"refs/heads/stable/yoga"},"86a8cc5f525f6e2ab5e287c027b3bd18b9fbfbd3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-15 22:04:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/64/877164/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/64/877164/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/877164/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/877164/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/877164/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/64/877164/2"}}},"commit":{"parents":[{"commit":"d27b006aac5735b334cc27ca8d00ff3f06ad9c0a","subject":"Removed unsupported test B309 from bandit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d27b006aac5735b334cc27ca8d00ff3f06ad9c0a"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-05-06 04:56:11.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-03-15 22:04:03.000000000","tz":0},"subject":"Validate MX records during recordset create or update","message":"Validate MX records during recordset create or update\n\nIf invalid MX record is provided during recordset create command or\nvalid record is updated to invalid record using recordset set command,\ncauses recordset creation/updation failure. This leads to zone enter\nin ERROR state and all further recordset creation calls fail. Validate\nthe records within recordset during create or update call.\n\nCloses-Bug: #1927304\nChange-Id: I0ace4d6c4ad6a6ee236e3af23805b01345d60a42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/86a8cc5f525f6e2ab5e287c027b3bd18b9fbfbd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/86a8cc5f525f6e2ab5e287c027b3bd18b9fbfbd3"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
