)]}'
{"id":"openstack%2Fdesignate~877456","triplet_id":"openstack%2Fdesignate~master~Ic27796fdad347ed44634ee8fa6367cfa93d8c8fa","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"31945":{"account":{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},"last_update":"2025-09-26 12:03:40.000000000","reason":"Change was abandoned"},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2025-09-26 12:03:40.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic27796fdad347ed44634ee8fa6367cfa93d8c8fa","subject":"Add a new floating IP notification handler","status":"ABANDONED","created":"2023-03-15 10:27:22.000000000","updated":"2025-09-26 12:03:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c5c1d0c1eac2d77d146bd0ce373bb39cd92ba7c1","_number":877456,"virtual_id_number":877456,"owner":{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-03-15 12:24:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-15 12:24:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e44e7b6a56d7c2631d22ceab040bbdfeaf99f47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},"date":"2023-03-15 10:27:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b22531bf1d169d449c8722d85ee6bce121141dde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},"date":"2023-03-15 10:35:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ed3099f4ee9ef056c6f1eeb787141d52727485f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2023-03-15 10:53:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9a936871aa5466de9e3d3a4d39f453f79ae34024","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2023-03-15 11:10:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6ff2ba707b2b6128f3351af98282b1656dee9d3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 12:24:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1daf900734324938a54ef1ff58cad2a0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0c70a85cd0b94ba681778266073c78bf : SUCCESS in 2m 29s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/814201190fd74bdfb41632c02aa84809 : SUCCESS in 41m 35s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8e9db422d2c84c2e96e8c8370a20ad2b : SUCCESS in 53m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca24815285d942e0a99430647c68b257 : SUCCESS in 54m 48s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/224595853b0b41248343d93a27d0f5f3 : SUCCESS in 41m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7348184c92c24b1c8c6fb46bf26643f9 : SUCCESS in 42m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/29dea9a8c3fb49d98859807a24280081 : SUCCESS in 53m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/30a8a0adbce04db780927f6901fbb1ed : SUCCESS in 58m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/31c6b9552ea043529009fb76a16c4b8a : SUCCESS in 35m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ac700ab76850494cbf825f5d8edea0db : SUCCESS in 46m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b050a4f984a94acfb11f104bd383c8d9 : FAILURE in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e3e512d47264ecfa232a42944fa8b65 : FAILURE in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a5d59982e83472b890fbb09ae53d191 : FAILURE in 8m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e3856ef2ce141d782e8b9b96f8810c1 : FAILURE in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc1078c703554b4c9c53f8f035e76660 : SUCCESS in 7m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/21ef8144cf2345e8a9740c6d4f46b8fc : SUCCESS in 27m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1ea79d01a2004a1f8fd98b9ced9f5232 : FAILURE in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c5c1d0c1eac2d77d146bd0ce373bb39cd92ba7c1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-26 12:03:40.000000000","message":"Abandoned\n\nI am abandoning this patch as it is fixed on this patches chain and didn\u0027t receive attention for 2.5 years. https://review.opendev.org/q/topic:%22ptr_records_use_designate_db%22","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"415d37df3d9f3c521db89bc391fdf31b84f1f969","revisions":{"f78f8e6a0407b69b30ab0463d198dab6e7fdb293":{"kind":"REWORK","_number":1,"created":"2023-03-15 10:27:22.000000000","uploader":{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},"ref":"refs/changes/56/877456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/56/877456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/877456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/877456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/877456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/56/877456/1"}}},"commit":{"parents":[{"commit":"758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3","subject":"Removed unsupported test B309 from bandit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3"}]}],"author":{"name":"Axel Jacquet","email":"axel.jacquet@infomaniak.com","date":"2023-03-15 10:09:51.000000000","tz":60},"committer":{"name":"Axel Jacquet","email":"axel.jacquet@infomaniak.com","date":"2023-03-15 10:27:19.000000000","tz":60},"subject":"Add a new floating IP notification handler","message":"Add a new floating IP notification handler\n\nThe older handler needed a zone_id to be able to delete the\nFloatingIP associated recordsets. This new handler works\nwithout the zone_id, and make it possible for operators to\nhave their recordsets automatically deleted when a floating\nIP is deleted even if they manage more than a single public\nIP range.\n\ncloses-Bug: #1746627\nChange-Id: Ic27796fdad347ed44634ee8fa6367cfa93d8c8fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f78f8e6a0407b69b30ab0463d198dab6e7fdb293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f78f8e6a0407b69b30ab0463d198dab6e7fdb293"}]},"branch":"refs/heads/master"},"9ce260453ae82514266041512cec8efba0bdb538":{"kind":"REWORK","_number":2,"created":"2023-03-15 10:35:25.000000000","uploader":{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},"ref":"refs/changes/56/877456/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/56/877456/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/877456/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/877456/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/877456/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/56/877456/2"}}},"commit":{"parents":[{"commit":"758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3","subject":"Removed unsupported test B309 from bandit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3"}]}],"author":{"name":"Axel Jacquet","email":"axel.jacquet@infomaniak.com","date":"2023-03-15 10:09:51.000000000","tz":60},"committer":{"name":"Axel Jacquet","email":"axel.jacquet@infomaniak.com","date":"2023-03-15 10:35:12.000000000","tz":60},"subject":"Add a new floating IP notification handler","message":"Add a new floating IP notification handler\n\nThe older handler needed a zone_id to be able to delete the\nFloatingIP associated recordsets. This new handler works\nwithout the zone_id, and make it possible for operators to\nhave their recordsets automatically deleted when a floating\nIP is deleted even if they manage more than a single public\nIP range.\n\ncloses-Bug: #1746627\nChange-Id: Ic27796fdad347ed44634ee8fa6367cfa93d8c8fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ce260453ae82514266041512cec8efba0bdb538"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ce260453ae82514266041512cec8efba0bdb538"}]},"branch":"refs/heads/master"},"3f61cfef891e1f37cec4cbf543473f723e91c8a3":{"kind":"REWORK","_number":3,"created":"2023-03-15 10:53:19.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/56/877456/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/56/877456/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/877456/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/877456/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/877456/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/56/877456/3"}}},"commit":{"parents":[{"commit":"758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3","subject":"Removed unsupported test B309 from bandit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3"}]}],"author":{"name":"Axel Jacquet","email":"axel.jacquet@infomaniak.com","date":"2023-03-15 10:09:51.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2023-03-15 10:52:55.000000000","tz":60},"subject":"Add a new floating IP notification handler","message":"Add a new floating IP notification handler\n\nThe older handler needed a zone_id to be able to delete the\nFloatingIP associated recordsets. This new handler works\nwithout the zone_id, and make it possible for operators to\nhave their recordsets automatically deleted when a floating\nIP is deleted even if they manage more than a single public\nIP range.\n\ncloses-Bug: #1746627\nChange-Id: Ic27796fdad347ed44634ee8fa6367cfa93d8c8fa\nCo-Author: Thomas Goirand \u003czigo@debian.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f61cfef891e1f37cec4cbf543473f723e91c8a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f61cfef891e1f37cec4cbf543473f723e91c8a3"}]},"branch":"refs/heads/master"},"415d37df3d9f3c521db89bc391fdf31b84f1f969":{"kind":"REWORK","_number":4,"created":"2023-03-15 11:10:04.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/56/877456/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/56/877456/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/877456/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/877456/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/877456/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/56/877456/4"}}},"commit":{"parents":[{"commit":"758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3","subject":"Removed unsupported test B309 from bandit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3"}]}],"author":{"name":"Axel Jacquet","email":"axel.jacquet@infomaniak.com","date":"2023-03-15 10:09:51.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2023-03-15 11:09:56.000000000","tz":60},"subject":"Add a new floating IP notification handler","message":"Add a new floating IP notification handler\n\nThe older handler needed a zone_id to be able to delete the\nFloatingIP associated recordsets. This new handler works\nwithout the zone_id, and make it possible for operators to\nhave their recordsets automatically deleted when a floating\nIP is deleted even if they manage more than a single public\nIP range.\n\ncloses-Bug: #1746627\nChange-Id: Ic27796fdad347ed44634ee8fa6367cfa93d8c8fa\nCo-Author: Thomas Goirand \u003czigo@debian.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/415d37df3d9f3c521db89bc391fdf31b84f1f969"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/415d37df3d9f3c521db89bc391fdf31b84f1f969"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
