)]}'
{"id":"openstack%2Fdesignate~877502","triplet_id":"openstack%2Fdesignate~stable%2Fyoga~I15864a76c8913f151154e2398ae86aea71883498","project":"openstack/designate","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-03-15 23:45:18.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I15864a76c8913f151154e2398ae86aea71883498","subject":"Removed unsupported test B309 from bandit","status":"MERGED","created":"2023-03-15 20:49:46.000000000","updated":"2023-03-16 01:57:04.000000000","submitted":"2023-03-16 01:56:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877502","cherry_pick_of_change":877150,"cherry_pick_of_patch_set":2,"meta_rev_id":"abe05662ccd39c6fbe50f79c6fb54715d5076da9","_number":877502,"virtual_id_number":877502,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-16 01:56:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-03-15 21:32:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-03-15 21:32:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-15 21:32:32.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-03-15 22:01:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59861e4ef30646ff29523de402246b3779a43801","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-15 20:49:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c857bb0b0f4bb8eae13ac0acb952ec626436c3d7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-15 21:32:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"25ae6ab6889a1a0c4f3eb2e51ff02364f3650843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 22:01:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfbe5dfea9d94fbd8b6c75e27c986666\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0fd9969a19c045b48d1e6509d9c63643 : SUCCESS in 49m 42s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8dfc2942d378464f93368408c142f596 : SUCCESS in 43m 01s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b1cdeaf042e42ab992c54589dbe41fa : SUCCESS in 38m 31s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/553b40f28a434635af77a4d7c34ab9f1 : SUCCESS in 52m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6696a75164e34abf8b942c85e5b0be4a : SUCCESS in 34m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a344820f9ba04c56aa42fed32561f8bd : SUCCESS in 1h 06m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2ef4b50942c3494681856c6667eaf0ab : SUCCESS in 41m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/42e8817a44584106abbe6f73acb570da : SUCCESS in 33m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b50b182e8b641e79d1ef2a6785a4256 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/263d3d3103fd4d499d0238c843e8a3d3 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd6d7ab84b2d489e89cbfe67298727f0 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f14fcbb078474c608b35ed5b2d8f207b : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a1de72c4fb849349faec48f190549d9 : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21d5aaa477ad4e2bb570ff36b16614b9 : SUCCESS in 4m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9fddc30d9718486898e10f4d0f5edc04 : FAILURE in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"721f6b64911848251f404e47aa4896cce8f18406","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 22:01:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34b74470b21bebb2958874db10d8e99acc9fc019","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 23:38:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae6c9d44900340dba1628dc68d3c9425\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/820dabd7cee5447d98ab081d17038bd9 : SUCCESS in 53m 32s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/927b32cb0c0b4446aacc8e90aa49bf32 : SUCCESS in 51m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/63af2b11feef4025a7f8dac6c3369126 : POST_FAILURE in 53m 45s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fed9b49e2ad640bf934e57f03449894a : SUCCESS in 1h 08m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6423372ba5f04be7978cc921d6a3ad25 : SUCCESS in 40m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/db717cd8f7c7419b8a8a05e6bb53a693 : SUCCESS in 56m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0d1c605d3c245ad95d391784fe7cad0 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a5f522491a44086b5afde06a0bd7493 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/387e7dcbd1bf4d6aa379e5a2a0c8cba5 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d896f40c436a4637952899be4ecb6d90 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e03ad222950452c9b128b45bc0fbefb : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"8046a7ec80aa75ced5f9e278562b9a23b0ba2c96","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-15 23:45:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c0b9d5fb1e61b6220b0c4fed496dec8710881a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 00:46:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a904dbb7cf9469087e8f0ca1d28167a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2b731e5355c44eb7a5fe835137078d5e : SUCCESS in 47m 21s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9c9127825abf4c02ae3fb13d8fc32c93 : SUCCESS in 40m 20s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c87cf58e4cf49b4833117e8c34d2d3d : SUCCESS in 54m 03s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/61c9f00f084a4f1b8546845e1c8145c8 : SUCCESS in 36m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1cddd93be6af4bbfb6c8067e2e69f7ba : SUCCESS in 48m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a55f51a6ebcb4ce4ac83ed79d1da1692 : SUCCESS in 46m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b7586c28656b445e8887d501d2569cca : SUCCESS in 48m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c144c26b6efa40c2ba1bcb8cfb2703cb : SUCCESS in 39m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a85b87654eaa46b980a526091b7f2011 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16a6dc398fe44851a42bbf41086e0bbf : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4b471aadd674386a7101f02a22aa978 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45a8f28f4ccf40dba8f761027cfad4b6 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2789d59428484d5e984f49d64cde6797 : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63055858789c48a798482e74e208d274 : SUCCESS in 3m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/eb8e910952ea4762ac808f0de8672321 : FAILURE in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"46133e7bcd6a95ae40ee8e70ad965720cc531bd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 00:47:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d9f0fa66e6ea47416401ae370f87407daf4385de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 01:56:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b58e0ad116484052a00c672ad9f71277\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fc3ea2b7d772425695027b18de7f98bc : SUCCESS in 53m 06s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d658e045180242e18948e4e2246509fb : SUCCESS in 51m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/edc5703694c44ff2b2682f7e72533617 : SUCCESS in 46m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/23e357c41c6649cd9e8eedfbf86e304b : SUCCESS in 1h 03m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d72f890ef4d0446b810bd1348851767a : SUCCESS in 53m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f80bdc701dda4caa9e330211da1d52c9 : SUCCESS in 47m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/002e1a4063774323ae2edf1725ad2625 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b41472fcc3e402dad2f28b5ef6dd20c : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac1889a51a454063afea038ddf99f289 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35673db7ad12489ba772fe5b52e6ab03 : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5522770280b4f08ae9e6613828c93c7 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"aa96c568c72dd0a6e5429bd95ee865f80e6cccce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 01:56:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"abe05662ccd39c6fbe50f79c6fb54715d5076da9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 01:57:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a9507bbe4e641c09fccc2495c132b95\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f91fda4b20a0464194e1b7127f9a1a54 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f050c87ab58a4a7f949ca39e20278b6b : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d27b006aac5735b334cc27ca8d00ff3f06ad9c0a","revisions":{"d27b006aac5735b334cc27ca8d00ff3f06ad9c0a":{"kind":"REWORK","_number":1,"created":"2023-03-15 20:49:46.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/02/877502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/02/877502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/02/877502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/02/877502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/02/877502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/02/877502/1"}}},"commit":{"parents":[{"commit":"1f30672bde3148f868fbaece8a58c20413e47b12","subject":"Fix dns.resolver import for mdns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f30672bde3148f868fbaece8a58c20413e47b12"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-03-11 00:47:19.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-03-15 20:49:46.000000000","tz":0},"subject":"Removed unsupported test B309 from bandit","message":"Removed unsupported test B309 from bandit\n\nB309 was removed with this patch and breaks the CI.\nhttps://github.com/PyCQA/bandit/commit/130a467434ff6b8f9bbd83100e409bc1fcb6f407\n\nChange-Id: I15864a76c8913f151154e2398ae86aea71883498\n(cherry picked from commit 758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d27b006aac5735b334cc27ca8d00ff3f06ad9c0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d27b006aac5735b334cc27ca8d00ff3f06ad9c0a"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
