)]}'
{"id":"openstack%2Fdesignate~877768","triplet_id":"openstack%2Fdesignate~stable%2Fzed~I0ee95fc56130e51bf5c799d252e79a469492b7db","project":"openstack/designate","branch":"stable/zed","topic":"bug/2012246","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-22 03:54:39.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-03-22 05:46:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0ee95fc56130e51bf5c799d252e79a469492b7db","subject":"[coordination] backend_url should be secret","status":"MERGED","created":"2023-03-20 18:52:36.000000000","updated":"2023-03-22 05:47:53.000000000","submitted":"2023-03-22 05:46:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877768-bug/2012246","cherry_pick_of_change":877767,"cherry_pick_of_patch_set":3,"meta_rev_id":"29ce4fbd23beea049f7ce30b05cdbf910cafd6f4","_number":877768,"virtual_id_number":877768,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-22 05:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-21 21:21:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-21 21:21:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-20 18:52:36.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-03-20 19:53:42.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2023-03-20 20:24:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-21 21:21:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"258d34fa10fd7a2a26d8cd358e41c89d0049a779","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-20 18:52:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e2118265976e966a699428c524e13ebfd8a3c96a","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-03-20 19:53:42.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f64ded1a4b56101a5a3d59391a982aea8c64c6a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 20:24:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ea0d58ff5514ed89e03c6b54a247a80\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/573be131ddca4b4cab36d077f4854bbb : SUCCESS in 34m 31s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8df627fb763543efb40eb4c8d746fd05 : SUCCESS in 1h 02m 06s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d75b1738e4a4dfba83ed1eb0e48d6ce : SUCCESS in 1h 00m 05s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/968057e3064441c3bd0425f116ba1a85 : SUCCESS in 35m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/390ca58c89a04cb6b57ceaa2ac74153b : SUCCESS in 53m 41s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ee380d781c74441aa764d32a6d903472 : SUCCESS in 1h 13m 17s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8cacb5725e3247d39968649ce79a96ad : SUCCESS in 1h 12m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/058c631e81734f1a8379bf9345f4a977 : SUCCESS in 44m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e0578b9ad1144af6b18174e226b78fca : SUCCESS in 38m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a86abf96355e4ee28cc887606eb0625e : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9080b8d9c5064f66878ef69689072a01 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82b5c3aed66d437385863c28c1b05e7e : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1c2a6a6351c4810abd7ed225803f721 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b40478f79744bc3a7107b55ec25f1db : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cfeb31a808a466f848f0bc72fcfe8bb : SUCCESS in 6m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/742d6396c68c446a8d6486087576bbfd : SUCCESS in 35m 44s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ee7e03e5b44c4560a7b05ca5ba1d3e9c : FAILURE in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a9ed11762a8f1d854220dd3efcd515ba395f487","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-21 10:55:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a75cda19868280b6fd04f7fda123ef6b8f30caa4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-21 10:55:47.000000000","message":"Topic bug/2012246-stable/zed removed","accounts_in_message":[],"_revision_number":2},{"id":"5b3d518825c464fc34b133fed4a6140c10b7268b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-21 10:55:51.000000000","message":"Topic set to bug/2012246","accounts_in_message":[],"_revision_number":2},{"id":"34f4cd1872711e73891c41f9b72434ceaf3acabc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 12:13:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b4cc6ce3ece4a8d8e1c9ba7ebe68e24\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b5ad305ac8364d0bb91cfc10259689dc : SUCCESS in 45m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4a32f53cbffa4f1b9e698442fceb6db5 : SUCCESS in 1h 00m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/47f65410fe2448f1bc581276285214a5 : SUCCESS in 59m 03s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/56e131a239724c3197c89d5768a90cfa : SUCCESS in 44m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/24d5f7a945a4448e8e2575889af574ce : SUCCESS in 43m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8ae9fff2b218477bad07537bd3eae929 : SUCCESS in 1h 00m 50s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2d5987d639044f42aec193be975a7878 : SUCCESS in 1h 11m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f34c839caf1d4087a02563e05736dde4 : SUCCESS in 44m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0339d0463cb34f1fbfd514f5a21fc927 : SUCCESS in 49m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edeeb29cb06645e0b39e1d144e8bc59f : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c514f8c8260f4d1393159ebe11c4c251 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/327a50780c664a8ba6bfbfc88515db61 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/536945e5b31c454981047d6467300cbd : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3bdd3fba0dd94aada00c90205a2eed9a : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18bcfd606ba847158736ab6707560d68 : SUCCESS in 7m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5a82441977e34aa89821d83f95da5377 : SUCCESS in 28m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cc85e40383634434902b1ac0a2d582f7 : FAILURE in 3m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5b495bd0ca8769a5434153379c25e35170ccbeb2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-21 21:21:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ef425cea0b134c0f797235b09145014063d40ccf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 21:22:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b12ca3520ab4554e799cb238b2d7640727297bac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 22:35:03.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08ecbc070b7949498d2ba291d26c3669\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ee44db789e20443daa586237dd1f9183 : SUCCESS in 41m 13s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c50bea7021c64a328abcee4f75fcdf1d : SUCCESS in 34m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d27806f2ed3d477a97b0148d195b26f6 : SUCCESS in 47m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8e36bf03643c4b4087cedcf852ca8f70 : SUCCESS in 56m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/48ee319a352645c997571fa71b79c6b8 : FAILURE in 37m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d79ab87bdd0b45f78a76d7d3ddff9036 : SUCCESS in 32m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82b5bdecd15c4337a6f6d8c7f0c91e56 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7adcbfcb55ed4d0fae3945cc7bd597f7 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6f61e417b994a78aa937a5f466d2d6d : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68b25530eb5249de95325eee7621104a : SUCCESS in 8m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/10093294bf9a40e9b8d7ea2257acebd8 : SUCCESS in 31m 39s","accounts_in_message":[],"_revision_number":2},{"id":"bda475c4b0a796b1909096e00b6dea4705ccb715","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-21 23:50:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"645bf46df70d35d8569c12d5b14c339673104a82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 00:44:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea574129f91141ff9805fed0d2633cd1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c655f0dbe42f4a1e9672e9f72540390c : SUCCESS in 34m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/df3cb5267f9740838426546b0083f078 : SUCCESS in 39m 33s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d210d9ff727414aab9192f13cc7aebf : SUCCESS in 33m 41s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f68a87ed2175487892fbda6169d9d001 : SUCCESS in 31m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/19e0b92cf60048728454f69058425396 : FAILURE in 31m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c1851c74a1504519b364cddb7423fd12 : SUCCESS in 47m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/309595c231944f96910076ea1b66ce69 : SUCCESS in 50m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9492d28a2f0e40ba983029cc5ee55dfb : SUCCESS in 28m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dfab2b1fa5654fe1a19b97615b9104a8 : SUCCESS in 31m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/798137fa4df242b796a7c6a5e6fb1964 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbfe6593fc58473bbefb8f34c008857b : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a4c18585e654993a064876d95c75761 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fac56af654c40c7add9916210e24805 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ca883c2d41445ea9b385eb5caaf5080 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/898f584ce0dc4f21b840cd48c002c91a : SUCCESS in 7m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3b9dd10a4307436a8c96830bf5e5f944 : SUCCESS in 26m 58s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d8afb60df3a645ae9f76dfe88c238f05 : FAILURE in 3m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"11ba209183f69d9e463de41bada49ef1abe361c2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-22 00:46:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f3f8cde4ae5865ff6f504577e32451beaa9f272","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-22 01:27:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"14f6a569d8d139b5ee8506871c9e86478c46900d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 01:55:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/206a3f8c99884d4abb0bc70ebe3f0169\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0ef9aac51bed4bbcaac6f90e1b8194af : SUCCESS in 39m 26s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/40b33c5088424789b5abb8559ba93f7e : SUCCESS in 49m 24s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3ef00a8bce24008bb3c09903d407da3 : SUCCESS in 45m 57s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e2522dbb04c4470d900484969b97c4cc : SUCCESS in 45m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/246185d25973411aad44c0e3c8c3b74a : SUCCESS in 43m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8f9633c45fcc463ea5a35efa16995e03 : SUCCESS in 58m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d206392490ff4536be2ac6ed362afd1f : SUCCESS in 52m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6b774c8de4c44032810c2b3ad70c27bf : SUCCESS in 32m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9b6822c70a984b35a77e38a3f0e88f77 : SUCCESS in 31m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c4a9381811a4299aed5617f6ebf9d35 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/617a3cab3a104c678c8096a2a8b068bf : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1546d025800840b6a3796511bd5eadbf : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c356df13e6324060ab2d1df07ebd52f9 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bda29722c6a247ec91187295c9c50683 : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65c1ced74a9f4aa999f42b866f45d1a3 : SUCCESS in 7m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a219010cf94d4a6e82a09ce65223a237 : SUCCESS in 34m 53s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/dc8370a00dd14173bb9373403d31623b : FAILURE in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d2f8bd90828d93bdc4bddc6ecfafc2f443d616bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 01:55:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3036d19943de02fc6d17cf27d4c5ed52a6deb933","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 03:15:42.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/54a18b2750b44011b49439e463499c28\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f37e8f1cf4e5417f9b5902151e72a5b4 : TIMED_OUT in 1h 12m 52s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6919b9fa039f48688e885f7f03c84cc2 : SUCCESS in 1h 08m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/694ab7be33ed43d289775b07b779507e : SUCCESS in 41m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e70443ac6eb74a25b1394f5d65c4e5de : SUCCESS in 1h 04m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/31e7e80db70e411f8f46d25fbf5941b2 : FAILURE in 39m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/aa71c59a04fe4aef8ee2131a328c6010 : SUCCESS in 43m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/018b24e6a4ea432995b179e29454f48c : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/365d85a1226d4fb3b58fd916ca50dd9c : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86b3cacbeda24e1e8c295e76d296f50b : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eca0e6c04a50449fae78fdd50c0ba6d9 : SUCCESS in 9m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/88b2a3dfba504161987a6af11e643927 : SUCCESS in 36m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1e4cc50924239e15bbb13d465dcf48e739d82601","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-22 03:54:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8679836bb8b0e1c97e27b82735c7f6d4b5b9b819","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 04:48:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e6e39b6c87a4804a9897985df43f110\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/929c1ace03d345fe9ffa8db0ad6371c5 : SUCCESS in 33m 33s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/98275b877a6a4ecf91d8c8888185f635 : SUCCESS in 44m 29s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8d9171045fe4eea8df0cd284a47132f : SUCCESS in 34m 35s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7e8f64ab7144454692728c2e80a12d05 : SUCCESS in 34m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/050d8b581f424131b86fcd4a86ace17f : SUCCESS in 35m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f4e97d58edfb4c34a9584ce904d46b99 : SUCCESS in 48m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/22f44127dcbc458180580a0bbf9239b6 : SUCCESS in 47m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d85115b769c743e89afebb22ef0ae3f3 : SUCCESS in 48m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1636ea3346e44d07937c67ebf56b3638 : SUCCESS in 35m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf13badcf68245b295d3ed0e41b39cc1 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55576ac6a4f24f39adccd8e7fb07ebd1 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b04612e156341efb565fb254adb283d : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbef8a4124c2470ca4862301bb54e0b2 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4281bd1ea3241c0905da840f02bb9ed : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8d2e8b9a33845f78fa691d09954497b : SUCCESS in 7m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9d314ec6214949e38550f0d96d6483f6 : SUCCESS in 27m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/855f7907b931474eb8589e61b722d52b : FAILURE in 4m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1c000be9a51ea5b5c33450450633d01a8c2d77f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 04:49:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c5971dccca82498f5181244d04b39818e3407a83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 05:46:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8699cea9bbfd44e5be58000138df22fa\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e89e56a7519c4bff868557afc080b2dc : SUCCESS in 39m 20s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c3612cb9bf7e4dbb89b142860c91af5d : SUCCESS in 44m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/43f2122562044ccfa14a578369b1de9b : SUCCESS in 50m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/76c86a0b5e4940d9a9d1da82e1b39ac3 : SUCCESS in 52m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/11ed43d08c93468090472e74437f4d57 : SUCCESS in 46m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/809074a244a7471aba01e2d5d9f4b9d7 : SUCCESS in 44m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4db8eab84a294dbb802d71e9175d2db0 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a08a4d39fbfd4bfca19fd3a97b649e1a : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/240c3fa101944adfb5836fbd25e100db : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbcbbd9001c9438fbd341507bb7c738a : SUCCESS in 7m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/edefc71b7df44ecfbbb50a4d559999c0 : SUCCESS in 31m 26s","accounts_in_message":[],"_revision_number":2},{"id":"78119efcc50c0daaac7410ac4da98a3a9207b7f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 05:46:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"29ce4fbd23beea049f7ce30b05cdbf910cafd6f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 05:47:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6893e47efcb440fdaf436e8f792d935f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c83e4937aa1d4a79b14fe5d218d51d47 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"be775801e33313305b5d90dca10bae166e6938c8","revisions":{"1322b5a2a6e82b579c8f1d6804c4a2e3b068e18c":{"kind":"REWORK","_number":1,"created":"2023-03-20 18:52:36.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/68/877768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/877768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/877768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/877768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/877768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/877768/1"}}},"commit":{"parents":[{"commit":"24ac8d7d987e37685201c509491b68363373dd4d","subject":"Removed unsupported test B309 from bandit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/24ac8d7d987e37685201c509491b68363373dd4d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-20 09:26:17.000000000","tz":540},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-03-20 18:52:36.000000000","tz":0},"subject":"[coordination] backend_url should be secret","message":"[coordination] backend_url should be secret\n\nThe backend_url option can sometimes contain secrets.\n\nFor example when redis coordination backend is used and authentication\nis enabled in redis, the plain redis password is put as an URL element.\n\n[coordination]\nbackend_url\u003dredis://:password@127.0.0.1:6379\n\nCloses-Bug: #2012246\nChange-Id: I0ee95fc56130e51bf5c799d252e79a469492b7db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1322b5a2a6e82b579c8f1d6804c4a2e3b068e18c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1322b5a2a6e82b579c8f1d6804c4a2e3b068e18c"}]},"branch":"refs/heads/stable/zed"},"be775801e33313305b5d90dca10bae166e6938c8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-21 10:55:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/877768/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/877768/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/877768/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/877768/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/877768/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/877768/2"}}},"commit":{"parents":[{"commit":"24ac8d7d987e37685201c509491b68363373dd4d","subject":"Removed unsupported test B309 from bandit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/24ac8d7d987e37685201c509491b68363373dd4d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-20 09:26:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-21 10:55:39.000000000","tz":0},"subject":"[coordination] backend_url should be secret","message":"[coordination] backend_url should be secret\n\nThe backend_url option can sometimes contain secrets.\n\nFor example when redis coordination backend is used and authentication\nis enabled in redis, the plain redis password is put as an URL element.\n\n[coordination]\nbackend_url\u003dredis://:password@127.0.0.1:6379\n\nCloses-Bug: #2012246\nChange-Id: I0ee95fc56130e51bf5c799d252e79a469492b7db\n(cherry picked from commit 541395c42414c40d105ce206e2f200456bbc375f)\n(cherry picked from commit bda31ec62c3eaa7fcdd25d69f97742744debf915)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/be775801e33313305b5d90dca10bae166e6938c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/be775801e33313305b5d90dca10bae166e6938c8"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
