)]}'
{"id":"openstack%2Fdesignate~878561","triplet_id":"openstack%2Fdesignate~master~I5ce96268cb39ae4a4a8d82a75ac192b2310455a1","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-03-24 22:39:20.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I5ce96268cb39ae4a4a8d82a75ac192b2310455a1","subject":"Restart neutron on grenade client update","status":"MERGED","created":"2023-03-24 21:21:26.000000000","updated":"2023-03-24 23:36:18.000000000","submitted":"2023-03-24 23:35:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878561","meta_rev_id":"998cdd578e51f464b38990c0e299dc20106623c4","_number":878561,"virtual_id_number":878561,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-24 23:35:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-03-24 22:39:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-24 22:35:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-03-24 22:39:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-24 22:35:07.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2023-03-24 22:40:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e63e4c02257273e6de6864da0acb34ae504d851","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-24 21:21:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e36daf2c1013387a64674f3d12ad5efe696ace1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-24 22:35:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f5050cd6b81d737d5654e39f53bdc20b7c35df47","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-24 22:39:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39e55038f6e02376b13cbcbde0b0e851384828f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 22:40:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24e254aad5c8440282fe942befca77ed\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8fa1194d833c4054bb514448e244ee25 : SUCCESS in 43m 57s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0b73d198a079482f8bb336dd1f043b8f : SUCCESS in 54m 08s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8d378e1e48647dfadc125196dce1dce : SUCCESS in 53m 00s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6f32ef2af13d4f7db25a4b0a5c6ef727 : SUCCESS in 48m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fdd670420e974f2c812d32d2dd3316a7 : SUCCESS in 50m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/38632b9e97fe49a78b35958e7c849dfa : SUCCESS in 1h 09m 22s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e0d866d7447241b2980cb2056336c844 : SUCCESS in 1h 07m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/baaf1e69f4af4d93b691c6038b53a213 : SUCCESS in 49m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6cdc8d0f5c0d43eeae9c86c495adbd6f : SUCCESS in 48m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e988f06f4c4a40bbb639e59c602835a2 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8fcdb793c1a43d8983d003e36146025 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c25c4ef40d3347cc8a2f232780242af6 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da2bfe2278d6432aae35fd62bf4bf529 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a55996b9a11946e3a4934d0bafc7659b : SUCCESS in 7m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/410f696c63b54b85a286730d41780de1 : SUCCESS in 29m 17s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e6930be3b3a24de29942bdd611363452 : SUCCESS in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"20abd6f22fc9b0d16699bd30356cf7202870201c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 22:40:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85768eb00c36dadc53736c940809dfa0033f97c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 23:35:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6adb1a5360d4467fb607d98c02028aed\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cf0e922e9332464991d184d3d8b9b483 : SUCCESS in 36m 36s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/76a7f50a398a4d4fa64d1c7ea4fecbfe : SUCCESS in 41m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7bbf79e3d306437fa9262d848a8c7bec : SUCCESS in 30m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/92b451b152914d069f0ca864f1dd4ad9 : SUCCESS in 52m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/58c1e55c1b004b209c0253b092559d74 : SUCCESS in 37m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/935d3e60903343469debbccbd827e4c8 : SUCCESS in 32m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07468527881947a8b853032e8009a7e4 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5b81c59c6e743c0b7e38d32be60b87c : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cc98906c8e74e70b130259e702d056e : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6957249704543faaac2fb1d0d761317 : SUCCESS in 6m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ff06419ed0694ac68a50fb2803f90dbb : SUCCESS in 35m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4c110874590824867b9330ae24f15c47e8de9ef5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 23:35:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"998cdd578e51f464b38990c0e299dc20106623c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 23:36:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94bfc0b0b0bf4ba8886242af118dd5d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7102117ee4ca4b91b97d16670baf4844 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8ec3b450b5242c28f19d3e8057c5573807a4954","revisions":{"b8ec3b450b5242c28f19d3e8057c5573807a4954":{"kind":"REWORK","_number":1,"created":"2023-03-24 21:21:26.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/61/878561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/878561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/878561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/878561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/878561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/878561/1"}}},"commit":{"parents":[{"commit":"5703810a40381d188b5d90fc68d7026c9b1c1150","subject":"Merge \"Update for sqlalchemy v2.x compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5703810a40381d188b5d90fc68d7026c9b1c1150"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-24 21:18:11.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-24 21:21:24.000000000","tz":0},"subject":"Restart neutron on grenade client update","message":"Restart neutron on grenade client update\n\nDuring a grenade upgrade the Designate grenade plugin updates the\npython-designateclient. This can cause the installation location to change\nto under the /opt/stack/new path.\nThis patch makes sure we restart neutron q-svc after this update to make sure\nit can still find the client.\n\nChange-Id: I5ce96268cb39ae4a4a8d82a75ac192b2310455a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b8ec3b450b5242c28f19d3e8057c5573807a4954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b8ec3b450b5242c28f19d3e8057c5573807a4954"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
