)]}'
{"id":"openstack%2Fdesignate~878729","triplet_id":"openstack%2Fdesignate~master~I1937e215dbd072b0a095df659c75f17a3f48c937","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-04-03 22:06:54.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I1937e215dbd072b0a095df659c75f17a3f48c937","subject":"Update designate for RBAC \"direction change\"","status":"MERGED","created":"2023-03-28 00:19:18.000000000","updated":"2023-04-03 23:07:01.000000000","submitted":"2023-04-03 23:05:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878729","meta_rev_id":"944f99a3daf97fcbd95ad207fdd4c7775865dad9","_number":878729,"virtual_id_number":878729,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-03 23:05:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-04-03 14:21:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2023-04-03 22:06:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-31 22:19:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2023-04-03 22:06:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-28 01:29:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-31 22:19:49.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2023-04-03 14:21:49.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"f8034e07692e4c5c347f505dc1be0c86a92abdf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-28 00:19:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5507e51c4e573b13abbad74413244c1f6ce7188a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-28 00:19:38.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a31149019d9cca674596873581c178f41ba49b25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 01:29:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d2c22a114294a8084647580d998cf3a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0e508d902a394d26929b7381419065e3 : SUCCESS in 30m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1e5e1d8eccfc4fc58abec56c4ca6b202 : SUCCESS in 1h 02m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/67e805400c164f48a4e81b295366f91a : SUCCESS in 52m 19s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d7590b875b6847b5950d807bed0b834d : FAILURE in 21m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6441d90a1bdf4047b9e13c7ffd88b186 : SUCCESS in 45m 07s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/25c02ea640b44a2da2fa07f1dc950a8b : SUCCESS in 1h 02m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/53a83c3206464ca6b038d35937429fe6 : SUCCESS in 1h 09m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f65db0b2257c4e00b599048b19586fcd : SUCCESS in 31m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/73bec7662b104e7caaf9eefb93de1c39 : SUCCESS in 48m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14433b89d410411d9506bf5891271d1d : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38e83853aaa04301b3e4ff01e3c8dd0d : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4de7e8065e44a13bea0eed8ff5c000b : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cb9feba89e54462a462bfcf54e8ee25 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a46c0dfda92c44c0aed729e30756a732 : SUCCESS in 7m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9ee0be04d25a458a98b8acbad7d78f6d : SUCCESS in 27m 13s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d5b7e40c682547c582034a54c84f5bb7 : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed5588dfd4776756133701bbba75f797af2ee07b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-29 21:05:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4ba0047423c20f601ce3f24b32baf59382f81b99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-29 21:08:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"309a746ba919dfb5c88efcc344efaff98a92967d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 22:15:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb6c7aae7bdc4edea6d4e1a77e67f663\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/212e87720ccc4698ae98242dd36409a0 : SUCCESS in 54m 21s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2d67e2351e2b478a9396c9cf1b4abbf8 : SUCCESS in 58m 23s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e59dbe8715e46a9a3df8767b111c31b : SUCCESS in 36m 39s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c207822a657c45c0ba02b809f8fbf708 : FAILURE in 19m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d67e446d59e14552b17f9b30dc6b8d27 : SUCCESS in 32m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/496fb632d88b472b8d06ddc6de528164 : SUCCESS in 55m 13s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/61acce2665a446af9a511fe3728e0497 : SUCCESS in 46m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1c04b49a263c4b49a75d6e1dd4ea8720 : SUCCESS in 50m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a57083fbab984cee8f973b169ca60768 : SUCCESS in 57m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bd1cfb4dc95484888ff545fab083571 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/581ad013059f40de95a32bbdb72c094b : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fa0bcabf515412ab8ff7f7ac64c97b1 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0962fbb70a1e4ec89a10c608c66d1803 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f382436e5738461c8618abd8ead87502 : SUCCESS in 6m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1d96c4382d114151920ed36ebf00eb9d : SUCCESS in 30m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/32655607b3bc4810850dd2c9ccd9d679 : SUCCESS in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b93cdf0f6095b613161b64d41a99250c2b67fa14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-31 00:39:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"88e99920dbb60c63e91bd8424f60dfb72ac1bd66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 01:49:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9871d7c8d1734d4ab7431549d502b278\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4124c7a15b3a4e5cbf7dc3de2eb4342e : SUCCESS in 32m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ff5fcd67fda947edbff0b95c5537de07 : SUCCESS in 1h 03m 51s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/55d0ca0ad5ed40349ccb43f3fdd93dac : SUCCESS in 35m 11s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/11043edb0755413fa2131dcf42df3b7f : SUCCESS in 31m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/81302fbb9f9a4aef943d83a03699a607 : FAILURE in 47m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/dc7cc0b38e274055859585650e0be3d0 : SUCCESS in 43m 54s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bcbb9b9a5ae64e849a08d5917ed434bc : SUCCESS in 46m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/42de3e1067e34c0db2fc77e6ce605a0c : SUCCESS in 46m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/234d8bc4771a496a83f0ee9da4177d12 : SUCCESS in 32m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45bfb7581ad14bb6ae9973ad91777bc3 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b3b1167957546a88a7bd48e7bda1655 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d9a40a100c247388bf27535f1c49844 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b063bbe97d61497592833a7017a59925 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db27e6f76cd44e719b3bae3e90209980 : SUCCESS in 6m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ab4c1610c8854ac2966f99aa284d1ddc : SUCCESS in 27m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/56e60f832ede40ee941f02fcc59de7e7 : SUCCESS in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"de41ae5238ac436a83f1c458df625b41c53ac2cc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-31 14:29:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a36e765b75fab3e3b4a57a5b5831b3604a35b8c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 15:39:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcb8ada3b85a42db8e04c40566b5aaa5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a2fd7afe5eab43059ac8d4a4043237e3 : SUCCESS in 37m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/162e4cb83fe440d880719a06671cf844 : SUCCESS in 46m 34s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e76d1505ae442518b28c104f3b42a98 : SUCCESS in 40m 24s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2ade73dd0e34463498657a32467e57b4 : SUCCESS in 34m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bdc2bf457e31448788d8ce5eafc9848a : SUCCESS in 31m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0c125c6c197f42979a86cfdd4d1a0140 : SUCCESS in 1h 03m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4d414fb288374a67ae85af88bb438830 : SUCCESS in 59m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/722511ef4d374794b8456290fdccb5bb : SUCCESS in 32m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6e3573c8303349c58cbc2e340246ff05 : SUCCESS in 41m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd0137b8bccc42439ce14cb9bb0141cd : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6059b21bcd5d45bb9466e7ba9d9fb508 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/887517393c254f88848d7aa08b9962d7 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa4a3a6662c245c3ba78062f76685645 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b1a555b3110475c80fae8b72c3fd1cd : SUCCESS in 8m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9e679118165f4ce18875bf7f4fb9c5dc : SUCCESS in 28m 46s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5ba06de189b1436e9f349090b6cd21c6 : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c1820a14e03f8254bdaaf1b609df6c033b72b440","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-31 22:19:49.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c4a0e406c6ef1ecdeb73796ed6677bd050f2b0eb","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-04-03 14:21:49.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8fbfa8f09f7fd89e8b9c8bb16b464edb7878f138","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-03 22:06:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"90ee446f4b8e967ac251deaa7bf8374b64b7f27b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 22:08:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9c473120a984d05362d14a754617337a7799fba3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 23:05:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4eb018e5081846e889b1433f7289052b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ad25acb9d8a54ac1ba6148f3a017fdac : SUCCESS in 38m 22s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4171871e930d43569c6359fb467af068 : SUCCESS in 36m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9a03c7406cea445c857e6fe3c09c6047 : SUCCESS in 34m 25s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/527d60e1c4b54fe285ba7a8d46de0a1b : SUCCESS in 49m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0838e20af437413ebd88364f456335eb : SUCCESS in 48m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c4383bfa7fb44d24b3580aec66f436c1 : SUCCESS in 44m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbae213a8c1f47158660b6c5503b91bc : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8891de88a6154e68b7c858f096aba00f : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66c66ba8931b429383e1f603a297eaae : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51312eb0afdc4ff68558203855f54dc8 : SUCCESS in 7m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0ac92b2251c04b73b259fa44a1eb1370 : SUCCESS in 27m 20s","accounts_in_message":[],"_revision_number":4},{"id":"1911cac9aac96c0e1a2fdd57993b4d5e27ad8145","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 23:05:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"944f99a3daf97fcbd95ad207fdd4c7775865dad9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 23:07:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b57f4b7d3e374951a18575b91881c29a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a9347531cf64837a1a211b92fc030a1 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c2e51939b43afcdc20e24f86595655bf588a81a0","revisions":{"4dbeaf2d899326fe3d4189e8a398efeafdc02290":{"kind":"REWORK","_number":1,"created":"2023-03-28 00:19:18.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/29/878729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/878729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/878729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/878729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/878729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/878729/1"}}},"commit":{"parents":[{"commit":"b8ec3b450b5242c28f19d3e8057c5573807a4954","subject":"Restart neutron on grenade client update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b8ec3b450b5242c28f19d3e8057c5573807a4954"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-28 00:16:32.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-28 00:19:17.000000000","tz":0},"subject":"Update designate for RBAC \"direction change\"","message":"Update designate for RBAC \"direction change\"\n\nThe RBAC goal has changed[1] and system scope is no longer going to be\nused. This patch updates Designate to align to this change in direction\nby removing the system scope from the policies.\n\n[1] https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#direction-change\n\nChange-Id: I1937e215dbd072b0a095df659c75f17a3f48c937\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4dbeaf2d899326fe3d4189e8a398efeafdc02290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4dbeaf2d899326fe3d4189e8a398efeafdc02290"}]},"branch":"refs/heads/master"},"75814fe6f7935221537f3f9de612763be651c6ce":{"kind":"REWORK","_number":2,"created":"2023-03-29 21:05:52.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/29/878729/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/878729/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/878729/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/878729/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/878729/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/878729/2"}}},"commit":{"parents":[{"commit":"b8ec3b450b5242c28f19d3e8057c5573807a4954","subject":"Restart neutron on grenade client update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b8ec3b450b5242c28f19d3e8057c5573807a4954"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-28 00:16:32.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-29 21:05:00.000000000","tz":0},"subject":"Update designate for RBAC \"direction change\"","message":"Update designate for RBAC \"direction change\"\n\nThe RBAC goal has changed[1] and system scope is no longer going to be\nused. This patch updates Designate to align to this change in direction\nby removing the system scope from the policies.\nIt also updates the functional tests to be ready for the switch to using\nthe new keystone roles by default.\n\n[1] https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#direction-change\n\nChange-Id: I1937e215dbd072b0a095df659c75f17a3f48c937\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75814fe6f7935221537f3f9de612763be651c6ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75814fe6f7935221537f3f9de612763be651c6ce"}]},"branch":"refs/heads/master"},"70459b049519ef9985e827c6bec171d979adc2db":{"kind":"REWORK","_number":3,"created":"2023-03-29 21:08:04.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/29/878729/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/878729/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/878729/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/878729/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/878729/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/878729/3"}}},"commit":{"parents":[{"commit":"b8ec3b450b5242c28f19d3e8057c5573807a4954","subject":"Restart neutron on grenade client update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b8ec3b450b5242c28f19d3e8057c5573807a4954"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-28 00:16:32.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-29 21:07:58.000000000","tz":0},"subject":"Update designate for RBAC \"direction change\"","message":"Update designate for RBAC \"direction change\"\n\nThe RBAC goal has changed[1] and system scope is no longer going to be\nused. This patch updates Designate to align to this change in direction\nby removing the system scope from the policies.\nIt also updates the functional tests to be ready for the switch to using\nthe new keystone roles by default.\n\n[1] https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#direction-change\n\nChange-Id: I1937e215dbd072b0a095df659c75f17a3f48c937\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/70459b049519ef9985e827c6bec171d979adc2db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/70459b049519ef9985e827c6bec171d979adc2db"}]},"branch":"refs/heads/master"},"c2e51939b43afcdc20e24f86595655bf588a81a0":{"kind":"REWORK","_number":4,"created":"2023-03-31 00:39:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/29/878729/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/878729/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/878729/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/878729/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/878729/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/878729/4"}}},"commit":{"parents":[{"commit":"b8ec3b450b5242c28f19d3e8057c5573807a4954","subject":"Restart neutron on grenade client update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b8ec3b450b5242c28f19d3e8057c5573807a4954"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-28 00:16:32.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-31 00:38:50.000000000","tz":0},"subject":"Update designate for RBAC \"direction change\"","message":"Update designate for RBAC \"direction change\"\n\nThe RBAC goal has changed[1] and system scope is no longer going to be\nused. This patch updates Designate to align to this change in direction\nby removing the system scope from the policies.\nIt also updates the functional tests to be ready for the switch to using\nthe new keystone roles by default.\n\n[1] https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#direction-change\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/879111\nChange-Id: I1937e215dbd072b0a095df659c75f17a3f48c937\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c2e51939b43afcdc20e24f86595655bf588a81a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c2e51939b43afcdc20e24f86595655bf588a81a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
