)]}'
{"id":"openstack%2Fdesignate~879208","triplet_id":"openstack%2Fdesignate~master~I1b56c492436821f650d1ba669614d92595d2f476","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-04-04 04:57:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1b56c492436821f650d1ba669614d92595d2f476","subject":"Fix sharing a zone with the zone owner","status":"MERGED","created":"2023-03-31 21:00:58.000000000","updated":"2023-04-04 04:58:25.000000000","submitted":"2023-04-04 04:57:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879208","meta_rev_id":"c3c46ba2a15cb59e9584a87181095d5d4b6886ac","_number":879208,"virtual_id_number":879208,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-04 04:57:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-04-03 14:51:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-31 21:55:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-04 03:50:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-31 21:55:36.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2023-03-31 22:16:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-03 14:51:03.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"53be0384f906f9191a45c6da908a87b8dd0cff71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-31 21:00:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b96945cc78a7eb1b2b37bf8d4b61d672f16c1c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-31 21:02:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4babf4ff96a3d0477c0f43cbcb7d21fb4a4ac30f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-31 21:55:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bcdcf73e225bcb2787fdd062e714401478d88846","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 22:16:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f740c2ca913d4f11aee6b11ccdf4dc63\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/31070ac56911402e9f212738bfc20a4b : SUCCESS in 30m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9ef7dffa6470441893935afc84b41908 : SUCCESS in 40m 33s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/18f506b97b21409d9f17e6bfa9a04116 : SUCCESS in 38m 55s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7605f36aebaf4616ba9c1fffca27d364 : SUCCESS in 31m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ea4efeb38a7a4d1baf8be8a313409f60 : SUCCESS in 29m 09s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8c675eb82bb4430b90e2dbc141e41e54 : SUCCESS in 45m 00s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2bdf96cf82e84f6891330bcdc70b7715 : SUCCESS in 1h 07m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4f92d6864d8a49fbb32ee2ec305abefe : SUCCESS in 44m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7d35d3a2e00a4ade9eddd355a3e74e5a : SUCCESS in 47m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54a592cfea4a492b8f5d3be0b21f0a3c : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b1aacada1494446b2cfff75b011d22c : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0416314deb9e4d6b983672ea84f5b354 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/daa5ec6ab4ba43a78e93875903a730a1 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b1d5148c6064081a487255d85506ea3 : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48bb68e704dd4ffdbaaa2dbdee12e543 : SUCCESS in 3m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/12b59342a9ed44d4891a982d1c0a81e7 : SUCCESS in 29m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e76ed1d4716e464e9b08a716d3e5f61f : SUCCESS in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"013bc24b9feb9562d5ffe823d3c93507b34b35d6","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-04-03 14:51:03.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d48565b9e8d23422a49c04b3b8f2bfe607a6da8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-04 03:50:47.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b79ce1f1bc771b301f28e4c2135b3bef348a322c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 03:51:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"92c62e4a81ed2fae742ae8033292326075f9d1e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 04:57:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8cdb390029f4ee3a1b80d81073b8475\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ffd9dd2ffa9e4243bb87f1afc1c924e1 : SUCCESS in 47m 40s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/78d352112c274220859538d4aebe70b5 : SUCCESS in 47m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/57bd94dc2301401682c664149d84c8ec : SUCCESS in 49m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a61db2ad10b5460c813db41d5718c576 : SUCCESS in 1h 01m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4e8c8095b6e048089836f1dce681110f : SUCCESS in 45m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/322e13fae37c4abfb3786459b24c99c4 : SUCCESS in 47m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83469c440ba749d29e31e9e9441ef36c : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1f66cb28f0c429cb10e1cdf9f17072a : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07d3630b6c9e4a2f8c9bb2a2b85bf40b : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1750029b5d94cc69fb852b9ed927cd3 : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e31c5eb812ec4547a9a6f79d11c341b1 : SUCCESS in 3m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a21e845c50424b57a4959fd01113fcdb : SUCCESS in 29m 46s","accounts_in_message":[],"_revision_number":2},{"id":"4a6e691ea716e513abd3134c99e2b12b96c4ec17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 04:57:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c3c46ba2a15cb59e9584a87181095d5d4b6886ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 04:58:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0349ee5b113748b8b138f0f1e9f30062\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2930697a8a548f4aa2b463ed11e6662 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15c7ba0ddd8742308bf574d55cf0abc3 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"edcd2e09989852da94dd1333fcab0eab244d09d7","revisions":{"6d5602e72630649735b1670c0521f26c17c4795c":{"kind":"REWORK","_number":1,"created":"2023-03-31 21:00:58.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/879208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/879208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/879208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/879208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/879208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/879208/1"}}},"commit":{"parents":[{"commit":"15fcba15e80a47b9c43518e35d127e675d5c3f42","subject":"Merge \"Removed unnecessary import in sql code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15fcba15e80a47b9c43518e35d127e675d5c3f42"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-31 20:57:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-31 21:00:56.000000000","tz":0},"subject":"Fix sharing a zone with the zone owner","message":"Fix sharing a zone with the zone owner\n\nThere was a bug that allowed users to create a zone share with the zone\nowner. This would then cause issues deleting the zone share as the zone\nowner owns the NS and SOA recordsets in the zone.\nThis patch raises a BadRequest if the user attempts to create a zone\nshare for the zone owner.\n\nCloses-Bug: #2011585\nChange-Id: I1b56c492436821f650d1ba669614d92595d2f476\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6d5602e72630649735b1670c0521f26c17c4795c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6d5602e72630649735b1670c0521f26c17c4795c"}]},"branch":"refs/heads/master"},"edcd2e09989852da94dd1333fcab0eab244d09d7":{"kind":"REWORK","_number":2,"created":"2023-03-31 21:02:24.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/879208/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/879208/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/879208/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/879208/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/879208/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/879208/2"}}},"commit":{"parents":[{"commit":"15fcba15e80a47b9c43518e35d127e675d5c3f42","subject":"Merge \"Removed unnecessary import in sql code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15fcba15e80a47b9c43518e35d127e675d5c3f42"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-31 20:57:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-31 21:02:24.000000000","tz":0},"subject":"Fix sharing a zone with the zone owner","message":"Fix sharing a zone with the zone owner\n\nThere was a bug that allowed users to create a zone share with the zone\nowner. This would then cause issues deleting the zone share as the zone\nowner owns the NS and SOA recordsets in the zone.\nThis patch raises a BadRequest if the user attempts to create a zone\nshare for the zone owner.\n\nCloses-Bug: #2011585\nChange-Id: I1b56c492436821f650d1ba669614d92595d2f476\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/edcd2e09989852da94dd1333fcab0eab244d09d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/edcd2e09989852da94dd1333fcab0eab244d09d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
