)]}'
{"id":"openstack%2Fdesignate~879305","triplet_id":"openstack%2Fdesignate~master~Ia95b9fcc3d3fe0bc5b4d5b24895d6b16727d6741","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"Ia95b9fcc3d3fe0bc5b4d5b24895d6b16727d6741","subject":"Use find_recordset when looking up recordsets internally","status":"ABANDONED","created":"2023-04-05 05:13:54.000000000","updated":"2023-04-07 23:45:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c5785e26a23868f7a587d9622bf73d2dd591d12c","_number":879305,"virtual_id_number":879305,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-07 07:19:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-05 06:11:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd1b905a18b9496a17e58e3a3ea3fa4ab52a9fa1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-05 05:13:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3010538fbf10fae89d207e544ecaa236081596a8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-05 05:20:49.000000000","message":"Topic find removed","accounts_in_message":[],"_revision_number":1},{"id":"c0439698040c4555216a107a203ec9cd12c1ec1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 06:11:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02bf3ac209c542e08102fc2e6321164e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cda8b54084074ca8bfbd6354c2145cf8 : SUCCESS in 32m 20s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9486013ffe1e4130a801a0237e2a767e : SUCCESS in 36m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0393416e33014c4eaf937cb53d333840 : SUCCESS in 33m 48s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d112b311c8604a5da76e94dc665cf780 : SUCCESS in 37m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/46ea67374a1d4b09b8c22813edb42271 : SUCCESS in 31m 19s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5d9d9e364741481aa90c3bbb60e7395d : SUCCESS in 45m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2c2b75a7e3ed477f9174c94b27ecaaf1 : SUCCESS in 44m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/564877be3cb3493d8df15bf52a09c0e5 : SUCCESS in 51m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/548afb33984c47b089c43b12066626bf : SUCCESS in 32m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73434660f9884aadb9a2410d854c8a99 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d2cdcc558be41d3936c00d952e24247 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/825a3395daa04c6aa774afae09839428 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8cea067e3d14ac9b22e3b2eb9118cd6 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3fecb2b82384deeac3c0484d45200e8 : SUCCESS in 8m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ac69fd03394548cab2bd7d3d5d560f6c : SUCCESS in 45m 47s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8a7275ca797746fcaf8826dccb2dc515 : SUCCESS in 4m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"681a1d6fc7de82a49abf618612678a6eb156bd17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-07 05:55:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"changekind:NO_CHANGE\")\n","accounts_in_message":[],"_revision_number":2},{"id":"1ceb4842eb3dc420caf0540838f04120291d8240","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 07:19:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/841b490a06e042e8b13d3a014385291d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e7cfc691092945cd85e8ac42a77def53 : SUCCESS in 47m 19s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3b3e5a2a543e49d5906a2d4c2e9d84b1 : SUCCESS in 49m 42s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/61da28ba772e4bd1bd9b4b58862c209d : SUCCESS in 36m 33s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/87bfefdd75314692ac442f63c4ace2d5 : SUCCESS in 33m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/790dc8c230234cafac86b01e3eea51eb : SUCCESS in 33m 17s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bff51061ed9749bfa6673c0fe58b0f3b : SUCCESS in 49m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8d823e2df89248059d7ac1267e7c238a : SUCCESS in 49m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5999da8148b0455e9d16924ab13ca2d7 : SUCCESS in 32m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/146ff7d4313e4f6d8762c1e1e4f7ce3b : SUCCESS in 33m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59fac7e023804e09bec566b03ed45f44 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aadd62828f0f4b7cbdfc5bee60e8c9ae : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fc8d3b757f6462d97df2489ef166c82 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6ad95a3223c4cbeb4ae7d45262d60b0 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0504140f8dc6406a8c816edb8a78daec : SUCCESS in 7m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a2f6903d1f6c40c4ba832b56b27f7ce2 : SUCCESS in 27m 48s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a68f79c2c5e44ded869397d887d02ec7 : SUCCESS in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c5785e26a23868f7a587d9622bf73d2dd591d12c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-07 23:45:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"44e29c88eed7ea2f348088334dd4c11ac0e6b034","revisions":{"77404d79eb5602462c0f6839a69266f031427d8f":{"kind":"REWORK","_number":1,"created":"2023-04-05 05:13:54.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/05/879305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/05/879305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/879305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/879305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/879305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/05/879305/1"}}},"commit":{"parents":[{"commit":"6591b53e509f17fd236903c86b751f3a1742f07e","subject":"Merge \"Fix sharing a zone with the zone owner\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6591b53e509f17fd236903c86b751f3a1742f07e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-05 05:04:56.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-05 05:13:47.000000000","tz":-420},"subject":"Use find_recordset when looking up recordsets internally","message":"Use find_recordset when looking up recordsets internally\n\nThis patch replaces the usage of get_recordset in some\ninternal calls with the faster find_recordset. get_recordset\nhas some additional checks that are unnecessary for internal\ncalls.\n\nChange-Id: Ia95b9fcc3d3fe0bc5b4d5b24895d6b16727d6741\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77404d79eb5602462c0f6839a69266f031427d8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77404d79eb5602462c0f6839a69266f031427d8f"}]},"branch":"refs/heads/master"},"44e29c88eed7ea2f348088334dd4c11ac0e6b034":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-07 05:55:14.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/05/879305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/05/879305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/879305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/879305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/879305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/05/879305/2"}}},"commit":{"parents":[{"commit":"f4ce71c8f80778b555eb1c0a29b67b8c32ed5f97","subject":"Merge \"Move to a batch model for incrementing serial\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f4ce71c8f80778b555eb1c0a29b67b8c32ed5f97"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-05 05:04:56.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-07 05:55:14.000000000","tz":0},"subject":"Use find_recordset when looking up recordsets internally","message":"Use find_recordset when looking up recordsets internally\n\nThis patch replaces the usage of get_recordset in some\ninternal calls with the faster find_recordset. get_recordset\nhas some additional checks that are unnecessary for internal\ncalls.\n\nChange-Id: Ia95b9fcc3d3fe0bc5b4d5b24895d6b16727d6741\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/44e29c88eed7ea2f348088334dd4c11ac0e6b034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/44e29c88eed7ea2f348088334dd4c11ac0e6b034"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
