)]}'
{"id":"openstack%2Fdesignate~879834","triplet_id":"openstack%2Fdesignate~master~Iadeaefe361638ec092eab1757f3b89e3020631af","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"5572":{"account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"last_update":"2023-05-22 21:27:14.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-05-22 20:28:37.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Iadeaefe361638ec092eab1757f3b89e3020631af","subject":"Refactor managed record logic","status":"MERGED","created":"2023-04-10 03:52:12.000000000","updated":"2023-05-22 21:28:12.000000000","submitted":"2023-05-22 21:27:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879834","meta_rev_id":"38faeb8ad83dd8ac0007670346545d976b8f8fb9","_number":879834,"virtual_id_number":879834,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-22 21:27:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-22 19:41:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-05-22 20:28:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-10 05:10:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-24 17:59:52.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2023-05-22 19:41:41.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"4291b08ecefe4cfbf251e97b92277a9927e618cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-10 03:52:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0c175167f2e1781b5efa08af4c2f86343fff484","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-10 03:55:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aaaa389ab090791b6c6cb526182773f21d9caae3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-10 04:06:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5787676b623330e18f625e2cb1e4166c814b5d88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-10 04:07:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"057726df296c6496541cbe81e0969bf3994364a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-10 04:14:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f9c8c21a17a5a21a9ce17d0998ed2e67ec52a1e3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-10 04:17:55.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"09baf8fa8b6fdcde02ad8fac13abb74bc1b6f6b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 05:10:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1ebcff0633a4bd5ae6d0848138751fa\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e332937329984f59877fe9de26db3fae : SUCCESS in 40m 01s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d558a0f0e6d24b3a9c48f401701fb6e9 : SUCCESS in 36m 50s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/07ad19e0afa7450ba21a92f8014062ff : SUCCESS in 41m 10s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/32be447cc606400fa10ab34f4b55dc53 : SUCCESS in 46m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/58f117d6e2ff432fb9b5c5fcbde154c0 : SUCCESS in 39m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a3ab746e977a4843ad3d018bc6ce0089 : SUCCESS in 50m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ecd3551084d444af9234a16181ccbd72 : SUCCESS in 52m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ac6cac0ab08a4df8b6b2ceee2befd337 : SUCCESS in 44m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/969cea31e92a46ad977c27eec27e8808 : SUCCESS in 36m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/118c0596c2f64efabb1e51c3527120f1 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/780620a32109492da506efcbb79eb5d3 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1baa3c5010248938cf07d7b0db048a2 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94e8fe19c49a45f6905a91027856227d : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2424a6e825eb4a30a7158de767bebf81 : SUCCESS in 6m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2647302edc874ab0beaa2702bc46b3aa : SUCCESS in 28m 00s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6c4698b3b1c94e0483054f7bb475f935 : SUCCESS in 3m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"16603036c273f40223bb8623818bc07ef1f07f0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-10 05:54:41.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"ea008728714edbd73fee51fcbb09fef788c4f029","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-10 05:57:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"14b2866d420eb0c236edd267ae9a3a91c39df699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 07:08:42.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dfc893f9998a4d0fba152d863a5feac2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/50d2c34ff01245698bb2a2c853c3ae7b : SUCCESS in 47m 59s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/db355526f4234b55aed5d7496a3a8073 : SUCCESS in 40m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f739f3685c6046d3a70fbf6f2c6862e2 : SUCCESS in 35m 16s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/320b2bdc58a84096b3f2f07b2456cae6 : SUCCESS in 40m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/18be9b6a04554c81b7a21c46f2287b98 : SUCCESS in 31m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/11977e4167ae41ae83c1874ec577a690 : SUCCESS in 1h 05m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b641a0e0280c4fa2b99dcf0989b6f28a : SUCCESS in 44m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5789ed3f78a24d05b0fefdb7cbef7bd9 : SUCCESS in 49m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4f5ddee7ad1447789ab80192666c053e : SUCCESS in 48m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d359cdd8e5dc43b385e2e5636eeb5410 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa5ce8158efe41fd9ee41dece17a40ac : FAILURE in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/952f86edc464411b93de529fdf19d521 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7f12a9cf436451eb065c2fd69d5f30f : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e50c2a9280e41ed9a7d819a162084d2 : SUCCESS in 8m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eb5a2248ef744385981b992be0102288 : SUCCESS in 35m 29s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/65366bbd42954508b666e0f6fc79d3cc : SUCCESS in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fb52b26cde1ab70268a703589401d52055cae58a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-11 01:43:19.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"e162eedf585108605be011234ef0ef7d00252ebf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-14 23:26:01.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":8},{"id":"d643fe8f95b5990de28eb477ec657a7d3a5bdc1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-14 23:53:03.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d278d511bb7fa4d61968f13be28ea96a56fe3cac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-15 00:49:47.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a98fc5958493471398c566332335c837\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9f1971be35874749b7d5328d7675a250 : SUCCESS in 36m 23s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d12d80c0e37d49f3b8d39f94435e58cb : SUCCESS in 38m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/117f6f786ab249ceb588271e83f280ae : SUCCESS in 39m 08s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1be3d719f15e4c5fb73f88db622a4439 : SUCCESS in 35m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d688109cd2b945a39424f0a61ef456fc : SUCCESS in 47m 12s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/410056dd099a41259820fecd6e10543c : SUCCESS in 45m 06s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ab5763d5c3574c8db39ffeedc489649a : SUCCESS in 47m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cfdb67c7e90b4650b16bee630bc553a0 : SUCCESS in 31m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b3800d18e5fb454291c9c0d99e854942 : SUCCESS in 51m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e463cacd93144c08b83b9b824e5478d : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b04d15de594412c912e1f499036204c : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e3070782bcc44a1a0a7f7572f6af406 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24bc9b99234049d299676f0a117eae8d : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edacbc9bad594632b04174e494bf9fab : SUCCESS in 7m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/99cc6c86e17246c182992ed03ad1e6de : SUCCESS in 28m 30s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9f0870772da94a65b27ca98b0be0776a : SUCCESS in 3m 42s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9b07612b7641222b32992f45008a8829d672bc0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-15 02:00:01.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"85d679b9db532c60d29931996c27478f5444483f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-15 03:04:39.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d53e004031044f28e55e3134fc58827\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9be7e2391ece4911bfe7b4e07b959bd9 : SUCCESS in 43m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8ad780a9339f4d0daf0d627688a3d00e : SUCCESS in 45m 50s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8d1e0fe772d4e26a03c24ce7375d4d9 : SUCCESS in 43m 31s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/21dd6ce544b844d388ac9d3d92da4894 : SUCCESS in 45m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/be9a3514600a49d98725d8a160b0a7e8 : SUCCESS in 38m 35s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/15ab1e100b574f77beed088d09d2f0dc : SUCCESS in 1h 02m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/da13ddc2b5ba435784d5bd1eebc3332e : SUCCESS in 1h 00m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6287f927d0be4568972e248daff4a208 : SUCCESS in 44m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a126d5e443044c71a0ac4555d0d46ec0 : SUCCESS in 50m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ab9cbd8b3fe487bbc304befab14d4c0 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9902e43e3f4487a830fe53f24d70606 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abd1c29fe0a940dc9b7718ede974acff : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14828f153c8c4e45b3df2f76f1368ca0 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce63d5bde1424229b6e216044bc9b049 : SUCCESS in 7m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0bb1cdf6857140eba380a293e4b1622f : SUCCESS in 31m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/80f6215e26654baea86367dca5e0f986 : SUCCESS in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"cf8969be12dcc6bf458ffb2bc448f074cd795548","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-20 20:25:41.000000000","message":"Topic new_sink_rpc removed","accounts_in_message":[],"_revision_number":10},{"id":"dc557a442053e8a37bde002350a46ad2b6ac3fa5","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-04-24 17:59:52.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"5a3cf3fdb606a8beb7eb8e6d07555bd5ede0a544","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-01 02:49:07.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"79e380e66c1b234dea6126033ceabb29bcf633c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-01 03:31:46.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":12},{"id":"f4991abad837ed89293422628467e9eb8cf2109b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-01 03:37:29.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"e6004242d684c2c86b16204571837be21637f884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-01 04:54:51.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a3f9594c4d04548a96547fb36f87cac\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5359193f55684f33b95c1f1065e0d728 : SUCCESS in 47m 19s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7b1f944c39574c2b955a308375aefd5b : SUCCESS in 45m 57s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/dcbe1c291034436fa1505b6e59c0f508 : SUCCESS in 43m 56s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b496b56c24b9417d83a187dfc54e5f27 : SUCCESS in 42m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ddc9c11ec5fe4c2a893b38d90d15c111 : SUCCESS in 47m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a722868d614e4fc693253275482a2877 : SUCCESS in 1h 10m 57s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0b59efc7a36e44abbfab74d788167848 : SUCCESS in 52m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9011973969284416a4e68bad89762826 : SUCCESS in 44m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2edec6ba23784d9d984b32505adbf686 : SUCCESS in 45m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6bb55da66d94fdf881a19a5a7d73291 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78b622f11aed43efadb4f93df17c7b46 : FAILURE in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15f5ae80a9a24cd1971e36ead857199c : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d22da32b6ef47cf8124f196ba4e5953 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0813f817b73e411ca345e9bea9ae757e : SUCCESS in 7m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7659b1e0c5874840b9c6a869fa6d28af : SUCCESS in 43m 52s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6ba9ffda47a84d738c4e6bf23fe047ed : SUCCESS in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"a55cdcad831c07e149c95694484b57338c159782","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-01 06:31:43.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"07ce7a0ca99b4993ee00ffcb1aa4bbf2b7c89eb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-01 07:34:25.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/710bba3f4dba47eca49cd52400634c42\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/029ad32def28494097a18ea2ca230a40 : SUCCESS in 40m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/67005044a70a4b48adc55283db593934 : SUCCESS in 49m 29s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c94daba7b3764405a0bc547443b3795b : SUCCESS in 39m 01s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ed139ca7a56a4bc29094ebd288ac6f68 : SUCCESS in 38m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2a39af9a77264811b8f994cfa127fe1c : SUCCESS in 31m 40s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9279464124314ce5ba679a8416fd0488 : SUCCESS in 53m 25s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6b269a45fc944202896a2b47b650f1f4 : SUCCESS in 50m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6eb9fc8e2e3f4aa18d5bfd11ff80c486 : SUCCESS in 49m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b8c8babcf15943baa3547d545b711ff6 : SUCCESS in 29m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/394a383b26594f6ba29ba26e8512d476 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35b7fde9314b41e9af103c3e050d71f1 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff79052237674abc9598f84b69e31037 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97fabab760174f5cb2c9e695b0d48e3b : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3438c7b2fc4c457889865acd481cf07a : SUCCESS in 6m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/876a5c7aca54464d927d3a5636f3d5f4 : SUCCESS in 36m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ef765e8286ff486cbd0ac12b418c00a6 : SUCCESS in 3m 43s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"02e9bd1df43e6cf63d26422aac4f3fa0eb596b37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-09 16:59:02.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"f73bc44716bc48b17601a0467b4116fa178fe770","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 17:58:54.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2d7b7d17b6d4040aad6c10bd2a2fae3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4fcedc68bb814013a73f92143f38c8a1 : SUCCESS in 41m 41s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c779ad3341594538966cbdb9801c7224 : SUCCESS in 50m 31s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/736e14d38cb24469a7156363d33b324a : SUCCESS in 49m 12s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9f6f6c8c00e14778bd6b4546bfa614b7 : SUCCESS in 39m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7581a8cab6c848769e01cfc97be76016 : SUCCESS in 34m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c07b1cfacbfd498981723e18e935f667 : SUCCESS in 55m 48s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/86b99af6d6564de6959b9e60df66596a : SUCCESS in 56m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4d132edda4c74ffd88e5480a35453c14 : SUCCESS in 40m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0d75f32185a94bfab03fe8c160c12d75 : SUCCESS in 38m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5bf6e00b4024e1cac09f856d75df3e5 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78b21f274a72433c80956c64094df37e : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3843bd10aa14c148804d6d34a7a79c8 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d54bb4d4428e4173950a138e0aed7c7e : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a0bdcd46e044d97b4fed1993305fcd3 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9221bd39a594fbe838c85063bb0b41d : SUCCESS in 8m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ecddbf16895c4009a37f89f66d297dff : SUCCESS in 37m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ede115ad2341452a968f4779d5ff206b : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"3f4e8f249dec607ae4ed1af763ccb39397439e4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-12 01:41:41.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"3e13c34f89cc4794f9db89f05c84b4afeafe49e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 02:38:44.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57c9db578c1b43a3932530f5d66124cb\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/899644fb5d44478587ebc57077aa1b28 : SUCCESS in 38m 40s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ab26eaf10bee4b15b8199dcbf8b4d488 : SUCCESS in 47m 03s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad14e5fbaa474248a81a9adaae15c986 : SUCCESS in 48m 57s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/469d199e7cbf43d8a5833e72741af533 : SUCCESS in 41m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9f735c5dd56c46d68691bc69af877d8d : SUCCESS in 35m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3f38d865b79a401f80f9b2601c07aa1f : SUCCESS in 51m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dae12e1650664179934677f4b8281a2e : SUCCESS in 55m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/34e5f17f797742abbd3097befdaaadd3 : SUCCESS in 40m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1576dff60f044e84a329c6379fedd146 : SUCCESS in 40m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d7fc7e5e52045c590c672f354e50ea0 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbc4af3a49554ccbafd654b2497c3e91 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61a41d8117b649c3abfb4dd54ab6fc12 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68875adc37a84adb8534967c01eb8e0e : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e472a90db5474d6a82730b08f5b6eb34 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a79238e47a554b1b9f71c62107cb2937 : SUCCESS in 7m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5905ddabbb34440faea4283f9597f0f6 : SUCCESS in 33m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/21462a75ebca40c7a913be81c680507d : SUCCESS in 4m 29s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"588e767b8b7b560d7b67433ef4ee7c699308046c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-22 19:41:41.000000000","message":"Patch Set 16: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"28df3ecdbd1baf9f46888ee368c6c6e57722ae0e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-22 20:28:37.000000000","message":"Patch Set 16: Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"cc5a0504ece9769093edf1d90af85f5c3093ecc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 20:29:09.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"2280fe8420e03c933a5c73cbb24ea9ba08f43273","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 21:27:13.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3e7e7c3d8c64d358d019557d1347bb2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/80adb9dc57f148638502fc507cb4d464 : SUCCESS in 39m 18s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8c9e027605974682a3c73564cc0634bc : SUCCESS in 40m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c5239652b3204e0582a0fe405a1b79a4 : SUCCESS in 41m 10s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e1e6656cae4f4c8292fe85dc3fb5e9e6 : SUCCESS in 56m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5e1d8c0344ed47b9a17b315d629bb729 : SUCCESS in 38m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/753fbb88279d4f68a7ffaa104ac6be08 : SUCCESS in 42m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97cd600b376640658ee0d013c0756419 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20f036ff7d294d178105b87755dc073e : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49c986ad632e40c2928b2319649c64a8 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e76f2250de05474a81896e1952a47a73 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/349fd426c4f44b1aabbd54a51db063f8 : SUCCESS in 7m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/07e196e0fadf4fbcbcac7953899f0086 : SUCCESS in 40m 41s","accounts_in_message":[],"_revision_number":16},{"id":"9533f3e22b06682f5fd1290a8007bbb4f8ad69dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 21:27:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"38faeb8ad83dd8ac0007670346545d976b8f8fb9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 21:28:12.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eab8312c97fe43a2971d1abf96f0f64f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57cd7671a7ae4024bc2a2dd0e752f6a8 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"1b4281b3ab08d507196ab55b3b3bdac8e7bbfa53","revisions":{"b64ed247ef2a36e21a1652fde8288ddcfcc449dc":{"kind":"REWORK","_number":1,"created":"2023-04-10 03:52:12.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/34/879834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/879834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/879834/1"}}},"commit":{"parents":[{"commit":"e19380bdbfaf458984252669c580e8512e2461f1","subject":"Use ids when removing sink managed records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e19380bdbfaf458984252669c580e8512e2461f1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:51:27.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:51:27.000000000","tz":-420},"subject":"Move sink database logic to central and reuse ptr code","message":"Move sink database logic to central and reuse ptr code\n\nChange-Id: Iadeaefe361638ec092eab1757f3b89e3020631af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b64ed247ef2a36e21a1652fde8288ddcfcc449dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b64ed247ef2a36e21a1652fde8288ddcfcc449dc"}]},"branch":"refs/heads/master"},"0acf0230be5105cecdfdc92a06f6e92b7af5376f":{"kind":"REWORK","_number":2,"created":"2023-04-10 03:55:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/34/879834/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/879834/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/879834/2"}}},"commit":{"parents":[{"commit":"e19380bdbfaf458984252669c580e8512e2461f1","subject":"Use ids when removing sink managed records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e19380bdbfaf458984252669c580e8512e2461f1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:51:27.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:54:41.000000000","tz":-420},"subject":"Refactor managed record logic","message":"Refactor managed record logic\n\nWe have two almost identical code paths for sink and ptr\nrecords. This patch unifies the two paths.\n\nChange-Id: Iadeaefe361638ec092eab1757f3b89e3020631af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0acf0230be5105cecdfdc92a06f6e92b7af5376f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0acf0230be5105cecdfdc92a06f6e92b7af5376f"}]},"branch":"refs/heads/master"},"26f768ba94f220522de6c921ac42aacecb58d36c":{"kind":"REWORK","_number":3,"created":"2023-04-10 04:06:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/34/879834/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/879834/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/879834/3"}}},"commit":{"parents":[{"commit":"e19380bdbfaf458984252669c580e8512e2461f1","subject":"Use ids when removing sink managed records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e19380bdbfaf458984252669c580e8512e2461f1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:51:27.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 04:06:36.000000000","tz":-420},"subject":"Refactor managed record logic","message":"Refactor managed record logic\n\nWe have two almost identical code paths for sink and ptr\nrecords. This patch unifies the two paths.\n\nChange-Id: Iadeaefe361638ec092eab1757f3b89e3020631af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/26f768ba94f220522de6c921ac42aacecb58d36c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/26f768ba94f220522de6c921ac42aacecb58d36c"}]},"branch":"refs/heads/master"},"b4727162544e1eacfab5d347f4fa7b7daf57d997":{"kind":"REWORK","_number":4,"created":"2023-04-10 04:07:39.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/34/879834/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/879834/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/879834/4"}}},"commit":{"parents":[{"commit":"e19380bdbfaf458984252669c580e8512e2461f1","subject":"Use ids when removing sink managed records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e19380bdbfaf458984252669c580e8512e2461f1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:51:27.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 04:07:32.000000000","tz":-420},"subject":"Refactor managed record logic","message":"Refactor managed record logic\n\nWe have two almost identical code paths for sink and ptr\nrecords. This patch unifies the two paths.\n\nChange-Id: Iadeaefe361638ec092eab1757f3b89e3020631af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b4727162544e1eacfab5d347f4fa7b7daf57d997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b4727162544e1eacfab5d347f4fa7b7daf57d997"}]},"branch":"refs/heads/master"},"771abe80d97957134d381599482ae3b525fc1bc2":{"kind":"REWORK","_number":5,"created":"2023-04-10 04:14:26.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/34/879834/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/879834/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/879834/5"}}},"commit":{"parents":[{"commit":"e19380bdbfaf458984252669c580e8512e2461f1","subject":"Use ids when removing sink managed records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e19380bdbfaf458984252669c580e8512e2461f1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:51:27.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 04:14:18.000000000","tz":-420},"subject":"Refactor managed record logic","message":"Refactor managed record logic\n\nWe have two almost identical code paths for sink and ptr\nrecords. This patch unifies the two paths.\n\nChange-Id: Iadeaefe361638ec092eab1757f3b89e3020631af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/771abe80d97957134d381599482ae3b525fc1bc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/771abe80d97957134d381599482ae3b525fc1bc2"}]},"branch":"refs/heads/master"},"8be3d5021d1be2dc7174370ea0f20be288759184":{"kind":"REWORK","_number":6,"created":"2023-04-10 05:54:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/34/879834/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/879834/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/879834/6"}}},"commit":{"parents":[{"commit":"e19380bdbfaf458984252669c580e8512e2461f1","subject":"Use ids when removing sink managed records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e19380bdbfaf458984252669c580e8512e2461f1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:51:27.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 05:54:32.000000000","tz":-420},"subject":"Refactor managed record logic","message":"Refactor managed record logic\n\nWe have two almost identical code paths for sink and ptr\nrecords. This patch unifies the two paths.\n\nChange-Id: Iadeaefe361638ec092eab1757f3b89e3020631af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8be3d5021d1be2dc7174370ea0f20be288759184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8be3d5021d1be2dc7174370ea0f20be288759184"}]},"branch":"refs/heads/master"},"564a44fdc82ecdfa779077d2d6d48417c5af80bb":{"kind":"REWORK","_number":7,"created":"2023-04-10 05:57:08.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/34/879834/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/879834/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/879834/7"}}},"commit":{"parents":[{"commit":"e19380bdbfaf458984252669c580e8512e2461f1","subject":"Use ids when removing sink managed records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e19380bdbfaf458984252669c580e8512e2461f1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:51:27.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 05:57:00.000000000","tz":-420},"subject":"Refactor managed record logic","message":"Refactor managed record logic\n\nWe have two almost identical code paths for sink and ptr\nrecords. This patch unifies the two paths.\n\nChange-Id: Iadeaefe361638ec092eab1757f3b89e3020631af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/564a44fdc82ecdfa779077d2d6d48417c5af80bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/564a44fdc82ecdfa779077d2d6d48417c5af80bb"}]},"branch":"refs/heads/master"},"10b03f86ea394fc45e171c905e0ec6f4f95364b0":{"kind":"REWORK","_number":8,"created":"2023-04-14 23:26:01.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/34/879834/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/879834/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/879834/8"}}},"commit":{"parents":[{"commit":"ae3bb0ddf359cd0a0d7b5b4705e331d104c6fce0","subject":"Merge \"Use ids when removing sink managed records\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ae3bb0ddf359cd0a0d7b5b4705e331d104c6fce0"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:51:27.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-14 23:25:15.000000000","tz":-420},"subject":"Refactor managed record logic","message":"Refactor managed record logic\n\nWe have two almost identical code paths for sink and ptr\nrecords. This patch unifies the two paths and should drastically\nlower the overhead of each of these requests.\n\nChange-Id: Iadeaefe361638ec092eab1757f3b89e3020631af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/10b03f86ea394fc45e171c905e0ec6f4f95364b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/10b03f86ea394fc45e171c905e0ec6f4f95364b0"}]},"branch":"refs/heads/master"},"5f7a27e91a07dd72ca2dbb3f2c57a5fc4a8c405c":{"kind":"REWORK","_number":9,"created":"2023-04-14 23:53:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/34/879834/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/879834/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/879834/9"}}},"commit":{"parents":[{"commit":"ae3bb0ddf359cd0a0d7b5b4705e331d104c6fce0","subject":"Merge \"Use ids when removing sink managed records\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ae3bb0ddf359cd0a0d7b5b4705e331d104c6fce0"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:51:27.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-14 23:52:56.000000000","tz":-420},"subject":"Refactor managed record logic","message":"Refactor managed record logic\n\nWe have two almost identical code paths for sink and ptr\nrecords. This patch unifies the two paths and should drastically\nlower the overhead of each of these requests.\n\nChange-Id: Iadeaefe361638ec092eab1757f3b89e3020631af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f7a27e91a07dd72ca2dbb3f2c57a5fc4a8c405c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f7a27e91a07dd72ca2dbb3f2c57a5fc4a8c405c"}]},"branch":"refs/heads/master"},"f36906d57b14ff87f8128bedfa0d4b71b6ebe61d":{"kind":"REWORK","_number":10,"created":"2023-04-15 02:00:01.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/34/879834/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/879834/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/879834/10"}}},"commit":{"parents":[{"commit":"ae3bb0ddf359cd0a0d7b5b4705e331d104c6fce0","subject":"Merge \"Use ids when removing sink managed records\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ae3bb0ddf359cd0a0d7b5b4705e331d104c6fce0"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:51:27.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-15 01:59:55.000000000","tz":-420},"subject":"Refactor managed record logic","message":"Refactor managed record logic\n\nWe have two almost identical code paths for sink and ptr\nrecords. This patch unifies the two paths and should drastically\nlower the overhead of each of these requests.\n\nChange-Id: Iadeaefe361638ec092eab1757f3b89e3020631af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f36906d57b14ff87f8128bedfa0d4b71b6ebe61d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f36906d57b14ff87f8128bedfa0d4b71b6ebe61d"}]},"branch":"refs/heads/master"},"a073f7a14891238167eeeb599fdeecaa8e9d6705":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-05-01 02:49:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/34/879834/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/879834/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/879834/11"}}},"commit":{"parents":[{"commit":"cfc8865fbe28516df5ac4a45aeaf346caead4a9c","subject":"Fix invalid fallback value for BaseAddressHandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfc8865fbe28516df5ac4a45aeaf346caead4a9c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:51:27.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-01 02:49:07.000000000","tz":0},"subject":"Refactor managed record logic","message":"Refactor managed record logic\n\nWe have two almost identical code paths for sink and ptr\nrecords. This patch unifies the two paths and should drastically\nlower the overhead of each of these requests.\n\nChange-Id: Iadeaefe361638ec092eab1757f3b89e3020631af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a073f7a14891238167eeeb599fdeecaa8e9d6705"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a073f7a14891238167eeeb599fdeecaa8e9d6705"}]},"branch":"refs/heads/master","description":"Rebase"},"61fc056f3f5795917a8befe676181a398304f450":{"kind":"REWORK","_number":12,"created":"2023-05-01 03:31:46.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/34/879834/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/879834/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/879834/12"}}},"commit":{"parents":[{"commit":"cfc8865fbe28516df5ac4a45aeaf346caead4a9c","subject":"Fix invalid fallback value for BaseAddressHandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfc8865fbe28516df5ac4a45aeaf346caead4a9c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:51:27.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-01 03:31:38.000000000","tz":-420},"subject":"Refactor managed record logic","message":"Refactor managed record logic\n\nWe have two almost identical code paths for sink and ptr\nrecords. This patch unifies the two paths and should drastically\nlower the overhead of each of these requests.\n\nChange-Id: Iadeaefe361638ec092eab1757f3b89e3020631af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/61fc056f3f5795917a8befe676181a398304f450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/61fc056f3f5795917a8befe676181a398304f450"}]},"branch":"refs/heads/master"},"3d8144e1fc7522444c3ee85c0ae04fd2afd276b1":{"kind":"REWORK","_number":13,"created":"2023-05-01 03:37:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/34/879834/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/879834/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/879834/13"}}},"commit":{"parents":[{"commit":"118294ac1443507ad8ac9a18d99102f02850c2b4","subject":"Fix invalid fallback value for BaseAddressHandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/118294ac1443507ad8ac9a18d99102f02850c2b4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:51:27.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-01 03:37:20.000000000","tz":-420},"subject":"Refactor managed record logic","message":"Refactor managed record logic\n\nWe have two almost identical code paths for sink and ptr\nrecords. This patch unifies the two paths and should drastically\nlower the overhead of each of these requests.\n\nChange-Id: Iadeaefe361638ec092eab1757f3b89e3020631af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3d8144e1fc7522444c3ee85c0ae04fd2afd276b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3d8144e1fc7522444c3ee85c0ae04fd2afd276b1"}]},"branch":"refs/heads/master"},"ee911857d96d80f85dd9b7edca4233dbbf4f790d":{"kind":"REWORK","_number":14,"created":"2023-05-01 06:31:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/34/879834/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/879834/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/879834/14"}}},"commit":{"parents":[{"commit":"118294ac1443507ad8ac9a18d99102f02850c2b4","subject":"Fix invalid fallback value for BaseAddressHandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/118294ac1443507ad8ac9a18d99102f02850c2b4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:51:27.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-01 06:31:32.000000000","tz":-420},"subject":"Refactor managed record logic","message":"Refactor managed record logic\n\nWe have two almost identical code paths for sink and ptr\nrecords. This patch unifies the two paths and should drastically\nlower the overhead of each of these requests.\n\nChange-Id: Iadeaefe361638ec092eab1757f3b89e3020631af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee911857d96d80f85dd9b7edca4233dbbf4f790d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee911857d96d80f85dd9b7edca4233dbbf4f790d"}]},"branch":"refs/heads/master"},"debb68ddda6cd1ee14fa8fa556166ee4723aaae1":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2023-05-09 16:59:02.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/34/879834/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/879834/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/879834/15"}}},"commit":{"parents":[{"commit":"f70053f2aaa8b8d46690c94530bb2fb5f908347c","subject":"Merge \"Use monotonic time to protect from system time jumps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f70053f2aaa8b8d46690c94530bb2fb5f908347c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:51:27.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-09 16:59:02.000000000","tz":0},"subject":"Refactor managed record logic","message":"Refactor managed record logic\n\nWe have two almost identical code paths for sink and ptr\nrecords. This patch unifies the two paths and should drastically\nlower the overhead of each of these requests.\n\nChange-Id: Iadeaefe361638ec092eab1757f3b89e3020631af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/debb68ddda6cd1ee14fa8fa556166ee4723aaae1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/debb68ddda6cd1ee14fa8fa556166ee4723aaae1"}]},"branch":"refs/heads/master","description":"Rebase"},"1b4281b3ab08d507196ab55b3b3bdac8e7bbfa53":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2023-05-12 01:41:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/34/879834/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/879834/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/879834/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/879834/16"}}},"commit":{"parents":[{"commit":"590064e8d986660e294d469e39e95241857778b5","subject":"Merge \"Remove unused test code for coordination\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/590064e8d986660e294d469e39e95241857778b5"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-10 03:51:27.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-12 01:41:41.000000000","tz":0},"subject":"Refactor managed record logic","message":"Refactor managed record logic\n\nWe have two almost identical code paths for sink and ptr\nrecords. This patch unifies the two paths and should drastically\nlower the overhead of each of these requests.\n\nChange-Id: Iadeaefe361638ec092eab1757f3b89e3020631af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b4281b3ab08d507196ab55b3b3bdac8e7bbfa53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b4281b3ab08d507196ab55b3b3bdac8e7bbfa53"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
