)]}'
{"id":"openstack%2Fdesignate~880327","triplet_id":"openstack%2Fdesignate~stable%2F2023.1~I02b5d7499440154160c89ed63a2f70652fe72145","project":"openstack/designate","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-04-20 21:07:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I02b5d7499440154160c89ed63a2f70652fe72145","subject":"Use ids when removing ptr records","status":"MERGED","created":"2023-04-13 20:18:43.000000000","updated":"2023-04-20 21:08:05.000000000","submitted":"2023-04-20 21:07:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880327","cherry_pick_of_change":879833,"cherry_pick_of_patch_set":2,"meta_rev_id":"986394e8aa80a6526c548b6895ef9ab51935d7f5","_number":880327,"virtual_id_number":880327,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 21:07:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-04-20 15:08:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2023-04-20 19:24:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2023-04-20 19:24:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-13 21:45:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 23:21:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2023-04-20 15:08:31.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2023-04-20 19:24:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"76e87e462d304adfb806fd7d72cade07d0958534","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-13 20:18:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f1fb78fe81165052a8b8731b38314a150cd7d9d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 21:45:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb2bab7fca3a4c17ac80f02aa26a6fd3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/19839be648844bc18c45311d98faed4a : SUCCESS in 37m 10s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c3390b20bb214202ae8551035f83ec71 : SUCCESS in 54m 14s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb3f4f0b43a146e1a23f71f61f892305 : SUCCESS in 41m 23s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e270fb09cfb64469a5e80116d32d49cb : FAILURE in 46m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/95f916d151e24befa6502356b3cc1944 : POST_FAILURE in 47m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c4f6dabd532444d09af1865733576e71 : SUCCESS in 1h 17m 22s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/065a3195cb934845a308f8cfeacc03a2 : SUCCESS in 1h 18m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6153f73d91c94880bff068ad88eb733f : SUCCESS in 47m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/77fe273bf5ff4367825b491ed0bed08c : SUCCESS in 49m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/471ecab254054a228823a2f73b50a2cc : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a09103dc06ca4e97a4184e67519bca45 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7375371609e4a1a80c2d9d2b889e3fa : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98e61dcf9aaf4b018a2a4021b1daf0f2 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85ff4d538eef403f95371ed772f947a9 : SUCCESS in 7m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1f6d5c7d25344e0cbf547c3f19c6f4b2 : TIMED_OUT in 1h 02m 31s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9153a234e4714d83abf3a0d67f0a88ea : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0921f0be2dcb66a856f8365b7217289d30c5edcb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-13 23:21:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a58169dc05ab259dfd9b1c0fdf137ebbed3ec47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 00:47:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bea2cb02c66a4feabfd8033998a45a38\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a5a601be63304f4a8d1c8d7432c5eca4 : SUCCESS in 48m 10s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/dc92b97c2ce74c6381e2af062dd52089 : SUCCESS in 1h 01m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f56f427ff66046a0a40c924d783adfcb : SUCCESS in 37m 34s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/5928a7aad96e482baccbd9182a2b6d28 : FAILURE in 45m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5db33de71d044cea9ff67b3f6240b1e2 : SUCCESS in 46m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b6aa2c48af194703a80a9fd62f4f023d : SUCCESS in 1h 01m 10s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3b19da87126a481ca3f21b381dc54f8a : SUCCESS in 1h 01m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2177099b9faf4a7492402b7ba765ce5d : SUCCESS in 45m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f3b462979ba54bddb4dde71fff0d3871 : SUCCESS in 50m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d815171916fb4c2b81aae368ed8d5ffe : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03de07772b6f46a1a7c7509e2071de00 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34093efdf6db475db6dd868f1b594e90 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23379e65d95c467d8a032e54733e2382 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad785546edf7442db2f18362176ecb7f : SUCCESS in 7m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c06ecb2219424bc0b93c3bba89531687 : SUCCESS in 27m 37s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c8645d92b6bf4ab6aded6efc52f35abc : SUCCESS in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"55843abfcfc07f755dbf142b1a239768e43de759","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-14 17:29:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbc5cdf6c1a583ef99e2fac788a87096ea7a4c73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 18:45:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/59fc6dd3b9294579b2d6792ba0948fad\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/101cfe055f014151ba1d2da0b90e4f98 : SUCCESS in 51m 10s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5e3d0d3f11b74577863fd20b811cb35a : SUCCESS in 43m 41s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b82da09d00d5494f955e0734ffe870de : SUCCESS in 33m 19s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4605322512474ab69168f029bb7f2fdf : FAILURE in 48m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fabdb4396ee3440bba1b305f241c7f9d : SUCCESS in 28m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4528751e2e8644fab7fbb7e1451d7f6f : SUCCESS in 1h 10m 08s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e8656761b84c4b7f94506b4978a5a3ae : SUCCESS in 44m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/edf9fabc3bab41f4ae04453b67775b86 : SUCCESS in 42m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bc882b72e35e431ca5288b665d1dc481 : SUCCESS in 48m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e1b980a908e4d98b746c9262f6cd7ab : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b157d4d61a4d4031924f3bff357f1890 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b69e701e669f4e27ba9202edca299fca : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e0adaa5c82d43139a228061f5da39b7 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cb67284133642e581e4354ed1a885d2 : SUCCESS in 7m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2528f14325184b33a32d92219a9b2935 : SUCCESS in 32m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/37b056b6363f4a50a43bc36a8a69d34a : SUCCESS in 4m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d720b2a3d0f5d12a11faf5da051481c6e19c4efa","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-18 23:54:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8fd719c541ee2814fa19adb12942e68e5959262","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 01:11:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5fee43f99d1f4da19058e2270482f670\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ec2a8daf01954997815dc38c9971daba : SUCCESS in 47m 24s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5c084c799fb94ef4b5d561d961b94057 : SUCCESS in 38m 41s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/486049be540e423ab6aa904fa03f60ae : SUCCESS in 35m 53s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/cf53b9d7080047c881043500e37b1fc3 : FAILURE in 49m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4481d54a024744efabf4ecfd800a5716 : SUCCESS in 42m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/69e02df9e2874d43a4391979c97c7464 : SUCCESS in 1h 07m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1462b8c6282346fc84962ecddac54700 : SUCCESS in 1h 10m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7f7151b6952b4f4eac484517621524a0 : SUCCESS in 39m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f4873ddad9ff4ab6bf07f1b1220b72c3 : SUCCESS in 47m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7edfd054a134339a2fd6929f0246207 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01728edaa5d44d69830710b26a456f26 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04d6f8e6d2c74eec8c4abe49587f2872 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c6b5b0efe374b9ba5cf86443a5227e1 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25ffbd34dd194015824c73bcaa22397a : SUCCESS in 7m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e3c80949d720481a9b341bb7a19f083e : SUCCESS in 33m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2dd47d4e1b114e79a47580d53aac295b : SUCCESS in 3m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b927fbd8c1132c67a620bc0a7c5a6bb3fc44fae","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-20 03:30:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ffb87603964bd47eade4b9357e37cf20a33a241","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 04:29:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b3c1832f13c4e52856c28e3eaa2f6cc\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6fb05103c88d4f18bb39b8fd0207fda1 : SUCCESS in 36m 58s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1a93411dcc0546a3a1454caea69a1548 : SUCCESS in 49m 03s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a44debbb52ee4947abb6b67dd4c3ff9b : SUCCESS in 37m 05s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/535fdac3e2914b54904127ce320366e2 : SUCCESS in 40m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1e56ebd1c62f4edc91a69efc4f3d4a84 : SUCCESS in 39m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e179ebebcb0c48a6ba603ffa51351965 : SUCCESS in 56m 57s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c8fcab951429486ca645fb2621fb5621 : SUCCESS in 53m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ca796abb96c64ffe8f3e695e6fa5bfa0 : SUCCESS in 37m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/961a71a9f85f49abb14f6d878e095ceb : SUCCESS in 37m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7f9f3fc2e7d41acad05dd7aec050cde : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a00befc897d34ab6b32068ba6576f6e0 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03aca94db0f5467eb130d97d9f49fe2f : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b682411ed37c4d6ca925cf12cdcf9047 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d7366af8ec141d3b0448c3548c4f829 : SUCCESS in 8m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5c0fd9dcbed44697a404d42223597514 : SUCCESS in 34m 27s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/59ba4fafa4564953946eb2c5c6e7d33e : SUCCESS in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff1ae8f089952623c8b9c2856bb94c9657765e50","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-04-20 15:08:31.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a21360fad2af7d30ef2da17456da50761fa4740b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-20 19:24:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a1c86a90c1e100ab329dde64b7189ebe1a6c9e73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 19:25:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f92ab76f10aa36abf6f4a86b9ce47c475ff80ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:07:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab40700194d046a68253d2418276e2ad\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/97fe5fc06b124e57ba4a3e00eb8d5c13 : SUCCESS in 55m 49s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9de0aabd7bec4868b29803d26b274045 : SUCCESS in 40m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6e27d20cb1a84ee4bba0d3e2c201db68 : SUCCESS in 50m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3be02c352354469895fe088f1e35e83c : SUCCESS in 57m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0fd6bbf8709642738aed8d63ae945e7f : SUCCESS in 48m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8048663ffb3f47458724751e3cddae46 : SUCCESS in 45m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59b6f923f5ca4f5a9669cc894096fee7 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45135ce2eabd4cac822abc65c7a81714 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10205dd2ab58438da2d1d6e8e58e3d34 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c84dbecd85f4964adeb86d9a2ff52e6 : SUCCESS in 7m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/71b3873bf95b4a6ead338bb9308bcf70 : SUCCESS in 32m 34s","accounts_in_message":[],"_revision_number":1},{"id":"be50fe880541c9f3de80e0436b73f24f495c8b3c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:07:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"986394e8aa80a6526c548b6895ef9ab51935d7f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:08:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78b54a207aef45c7adfff3e13a44b1dd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae95adf48113492784df537893c75d62 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1c7344d71160ee0d2c55ad51ba9a2024c8e98b1d","revisions":{"1c7344d71160ee0d2c55ad51ba9a2024c8e98b1d":{"kind":"REWORK","_number":1,"created":"2023-04-13 20:18:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/27/880327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/880327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/880327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/880327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/880327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/880327/1"}}},"commit":{"parents":[{"commit":"4c08082bea6793334d3a215cdc0c3f051cdf6785","subject":"Fix sharing a zone with the zone owner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4c08082bea6793334d3a215cdc0c3f051cdf6785"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-09 18:38:46.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-13 20:18:43.000000000","tz":0},"subject":"Use ids when removing ptr records","message":"Use ids when removing ptr records\n\nIf the record status changes during the removal process,\nthe current implementation will fail.\n\nChange-Id: I02b5d7499440154160c89ed63a2f70652fe72145\n(cherry picked from commit b3161ece234b5e2ee7658060aaa936af625bef38)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1c7344d71160ee0d2c55ad51ba9a2024c8e98b1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1c7344d71160ee0d2c55ad51ba9a2024c8e98b1d"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
