)]}'
{"id":"openstack%2Fdesignate~880328","triplet_id":"openstack%2Fdesignate~stable%2Fzed~I02b5d7499440154160c89ed63a2f70652fe72145","project":"openstack/designate","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-04-20 21:17:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I02b5d7499440154160c89ed63a2f70652fe72145","subject":"Use ids when removing ptr records","status":"MERGED","created":"2023-04-13 20:18:56.000000000","updated":"2023-04-20 21:18:48.000000000","submitted":"2023-04-20 21:17:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880328","cherry_pick_of_change":879833,"cherry_pick_of_patch_set":2,"meta_rev_id":"020bd789df44b0672f1c2958c341e288076dc5be","_number":880328,"virtual_id_number":880328,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 21:17:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-04-13 23:21:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-04-20 19:25:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-13 21:34:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 23:21:27.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"61cd42ec009e5f0fd861a554988d7bcdeba34352","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-13 20:18:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9f54fcaa7ed99361167e5e18ac8362ebff6864fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 21:34:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a782f8b5833f4b49a371c82c558284e5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d3bd226c9670460eb9a1d936be823da1 : SUCCESS in 49m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a0eae8a302f6496b9e3fdfc91d26d95a : SUCCESS in 45m 24s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0be52977827149e1b76ff5b65f9704d2 : SUCCESS in 41m 30s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/578f2b2330cd4c3eb3971d7210d46195 : SUCCESS in 42m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f669e0ed672c4662a891432186a08b85 : SUCCESS in 46m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c03c8ba1a9174c8582c2aa5b47f1b667 : SUCCESS in 1h 06m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/86a1a4a4c2cf4dd0a0b5e4480f8e3202 : SUCCESS in 59m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/88af7e699f6d42ae8fb0a5384d8b4be5 : SUCCESS in 43m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7efee304168246fbac3c64e26810a896 : SUCCESS in 50m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a15776948c6f48728b0c4a89d7340447 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/826c5a245ccb486787391a039f7e2338 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07fc9e19aac440668e297a570990cd07 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6aa829bdf8e146388d1f16c8f2ba1090 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f80d70bd891406ca5d5c65553f8ff62 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52215ff48ebc4acd8a40f6ed05b1b702 : SUCCESS in 6m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9533785047864156bde9bcf49d886a73 : SUCCESS in 39m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5386652521ef4dceb8d2d8a9aa50040d : FAILURE in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf3e218c3938a0618fc96c54a629d7d75d81c1e1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-13 23:21:27.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"26fdcc8e604fe15553e727a13836978e1eb51c79","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-20 19:25:46.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f803577c5c212b4e84084fcd9582bed2dca0da53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 19:26:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a296adf63ebcb589574af7f5f74295f8bd97da6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:17:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83f92c6fd41c4efe97161156d84865c3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b051312a317f4e70b77dd6c4d4df50e6 : SUCCESS in 43m 46s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0ed5eb9de1504da292aefcacbb82b55a : SUCCESS in 51m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5fdbf810352244f08d5c389caebee52d : SUCCESS in 49m 21s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/48c3198ab6e0478fbc83185add3d4f6b : SUCCESS in 1h 07m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f1a0a47964c8415aa9f0e861f644b44b : SUCCESS in 32m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c819fe7a9c794a2f80c809d9af412359 : SUCCESS in 40m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ec3fad675be44d38600195d3489aa7d : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19d6319d4f03457eb607888b4377bb00 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b51e07ab457d41d4a4dcb6da6b45c4ac : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6bb679242ec4627b90e33e08311f560 : SUCCESS in 7m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/24ae83d893c24ae7a607ecb2bf9edab6 : SUCCESS in 31m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0ac450a5941f0af73a482c37a367c30e912bcbc0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:17:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"020bd789df44b0672f1c2958c341e288076dc5be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:18:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c83d1e2d1a447d09786ba82155251e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a71d3eb62d1d4ad3ae67ee28cdf58762 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb67571d90ca0ed079ed391a6ed3970998491ebc","revisions":{"eb67571d90ca0ed079ed391a6ed3970998491ebc":{"kind":"REWORK","_number":1,"created":"2023-04-13 20:18:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/28/880328/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/28/880328/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/28/880328/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/28/880328/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/28/880328/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/28/880328/1"}}},"commit":{"parents":[{"commit":"be775801e33313305b5d90dca10bae166e6938c8","subject":"[coordination] backend_url should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/be775801e33313305b5d90dca10bae166e6938c8"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-09 18:38:46.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-13 20:18:56.000000000","tz":0},"subject":"Use ids when removing ptr records","message":"Use ids when removing ptr records\n\nIf the record status changes during the removal process,\nthe current implementation will fail.\n\nChange-Id: I02b5d7499440154160c89ed63a2f70652fe72145\n(cherry picked from commit b3161ece234b5e2ee7658060aaa936af625bef38)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb67571d90ca0ed079ed391a6ed3970998491ebc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb67571d90ca0ed079ed391a6ed3970998491ebc"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
