)]}'
{"id":"openstack%2Fdesignate~880329","triplet_id":"openstack%2Fdesignate~stable%2Fyoga~I02b5d7499440154160c89ed63a2f70652fe72145","project":"openstack/designate","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-04-20 22:39:41.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I02b5d7499440154160c89ed63a2f70652fe72145","subject":"Use ids when removing ptr records","status":"MERGED","created":"2023-04-13 20:19:03.000000000","updated":"2023-04-21 00:53:23.000000000","submitted":"2023-04-21 00:52:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880329","cherry_pick_of_change":879833,"cherry_pick_of_patch_set":2,"meta_rev_id":"49b5a10481a7a5cd8791b3636736cf82e953adcf","_number":880329,"virtual_id_number":880329,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-21 00:52:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-04-13 23:21:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-04-20 19:27:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-13 21:21:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 23:21:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"992cff0107e4a5486748a12994796818703ec87e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-13 20:19:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a471271293bd0e436fad8b761653080b5399b439","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 21:21:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97941db866084db1a45b8a75bc9aecf4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a157527e7a3a4a89803e5381f124687b : SUCCESS in 48m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a040c78e88a64e82a73e9a93d32e5a2c : SUCCESS in 44m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a8d46e0ca844a3f86c7e36a612a6a5f : SUCCESS in 42m 35s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8eb03bc19b884eeea0f760d25134fbc9 : SUCCESS in 53m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3efc6bd824394e4198e5e3285ffd9066 : SUCCESS in 51m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/de398dee8f14463a82f4d208f534284a : SUCCESS in 47m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/434222b6c47f4c33b9d4242d6002541a : SUCCESS in 50m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/daa4908416f94beca25a3bee94381aec : SUCCESS in 51m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01c59cda3fcc41e2abefd858076594cc : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e29515693b04d79a894257027977356 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a1977501221401dab4ca9d1eb1bcdaf : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f4e825351a04215ac5d30bf03f970ff : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f477eb318e104ccdbed5557ce5e277e1 : SUCCESS in 8m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4973adab10774bb6bdede1dd123ba2a4 : SUCCESS in 35m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/411fb63e16b24db792c3e8955efae6aa : FAILURE in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"81484c55f4b5c8a95e6877b71fe3ae24bd45a0e3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-13 23:21:49.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ce01b295b092e4ae6802f798e39c8aa16ed1d1b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-20 19:27:30.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba07a6fe00a5a7605d56d4485fa9759f75b03ff7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 19:28:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b11fce8a0d450393bed8155f1ae30dc55171a88c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 22:03:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/050d6b033a564a51bfa827f3bddc3c56\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/53c5c89b0c384fb088ce3614debe63b2 : SUCCESS in 52m 55s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2f62d1cf31a042bb9ccdff4a08ab3242 : SUCCESS in 49m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/aca86684e48d4592b4d74fea720889a1 : SUCCESS in 48m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3a5ed1323ec240dda2676dc8d89d918c : SUCCESS in 1h 03m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/38b906112b114f5ab355c84edecbf1a5 : SUCCESS in 47m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d8a65cdc54de4723a8e6032023b7e615 : SUCCESS in 54m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82ee8944393e4a5f9f33ff75d35cba57 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb08f5a805b44493859ea43c3004b2bd : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d29d8b3bdcf14758b537dc5e8f0f4c92 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b53abe5c1fb4f9b862b2951968bf72d : SUCCESS in 7m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3c85d2f537d24c0390b96f147ddae529 : FAILURE in 14m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b385dbf519cf187117fd94f83ac6f244d2b98f8d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-20 22:39:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"453389baaa3853200b8afc239dbe09306a976f93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 23:44:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be65d10569a84fd98a82f4123377fc7f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9dcca5f70f014b8caa44495a731a739e : SUCCESS in 34m 55s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0fbd110705be48d3969621240f0aaaf0 : SUCCESS in 38m 24s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e0010ccc86c4e0c943ee205b2eeebe8 : SUCCESS in 37m 50s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2588737afdd848938cc15c85cadd9f80 : SUCCESS in 34m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/380536bf9ac340d3a5423f9bf7233ceb : SUCCESS in 48m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/81ebea28a327474cae807e893dba5c09 : SUCCESS in 1h 00m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ca09686e85da4058affae4869e4d91ac : SUCCESS in 35m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bde20b361ce44b0ba37ebf41f2b55b2d : SUCCESS in 49m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39e5952c56ee4a5b9bdf1dbfadfdd683 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5561af28f594489a1d718fb808d33df : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efe6ceffd97c4d229f6e584d5cabe54d : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9da02eee5b1a425abf000cff8ca1256b : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7e081a0cf244ee5b0cc658e8db7634b : SUCCESS in 7m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5054ed8d7bfc403c933e1cf20307703c : SUCCESS in 33m 04s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6b78493152e94f97908dde029a90fe03 : FAILURE in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"527e9e62bcdd40e5d2b58864647a72df98080f50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 23:45:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e0dc190581b771f0a732e46a0299e79276702573","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 00:52:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b37469e61aef4b7ca045115b6c7538ae\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f59e2f73734d43efa36c3c97a4e226aa : SUCCESS in 51m 19s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/5cff2ebc02e44b859511ca773949baa7 : SUCCESS in 51m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e9e07d0c851d461c86f8378ce895c910 : SUCCESS in 50m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/26af476130194fb086ebd38d0b819ab0 : SUCCESS in 1h 02m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bd33468bb17742f8996da883d49fddaa : SUCCESS in 41m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c3fab49344cd47e6837b357607df4b3b : SUCCESS in 47m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7afbce3c44994e218cbf83068e310416 : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ca1f803a6494d73a11703e929482810 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc9c8594f2ac4004b9b38b551b0dd737 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb18dfec5ca34f2dbb245112c73c9017 : SUCCESS in 7m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3f7f898662794d6691bd172e5d8b8eb8 : SUCCESS in 39m 27s","accounts_in_message":[],"_revision_number":1},{"id":"61bc775ed70dd3e0b2007f8861655b86eaa69167","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 00:52:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"49b5a10481a7a5cd8791b3636736cf82e953adcf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 00:53:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dcb8ae9389f4e13a5ffd44d7c9dab42\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e8c0da7267a4b6aac3c514e87482747 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1df48dfee1246cf7f719c790c6964c0863756717","revisions":{"1df48dfee1246cf7f719c790c6964c0863756717":{"kind":"REWORK","_number":1,"created":"2023-04-13 20:19:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/29/880329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/880329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/880329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/880329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/880329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/880329/1"}}},"commit":{"parents":[{"commit":"f1981d057f08f9229eba2e8191dc06b22be1cffd","subject":"[coordination] backend_url should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f1981d057f08f9229eba2e8191dc06b22be1cffd"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-09 18:38:46.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-13 20:19:03.000000000","tz":0},"subject":"Use ids when removing ptr records","message":"Use ids when removing ptr records\n\nIf the record status changes during the removal process,\nthe current implementation will fail.\n\nChange-Id: I02b5d7499440154160c89ed63a2f70652fe72145\n(cherry picked from commit b3161ece234b5e2ee7658060aaa936af625bef38)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1df48dfee1246cf7f719c790c6964c0863756717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1df48dfee1246cf7f719c790c6964c0863756717"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
