)]}'
{"id":"openstack%2Fdesignate~880370","triplet_id":"openstack%2Fdesignate~stable%2Fxena~I02b5d7499440154160c89ed63a2f70652fe72145","project":"openstack/designate","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-04-20 22:03:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I02b5d7499440154160c89ed63a2f70652fe72145","subject":"Use ids when removing ptr records","status":"MERGED","created":"2023-04-13 20:19:22.000000000","updated":"2023-04-20 22:05:08.000000000","submitted":"2023-04-20 22:03:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880370","cherry_pick_of_change":879833,"cherry_pick_of_patch_set":2,"meta_rev_id":"b5336b06427e39d6296967e653c2d0f96d1fc3b1","_number":880370,"virtual_id_number":880370,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 22:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-04-13 23:22:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-04-20 19:27:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-13 21:19:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 23:22:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"9ea4448460711a89e66aa60526ed3d44e9b69fc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-13 20:19:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b07a8bde8b33f178bc29cedf9cb38f97e4b5ba97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 21:19:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fb1f444c24849a496122e76666e24e5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1d9616d75ee64ffc8ce438912e6a41c0 : SUCCESS in 31m 03s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b46d78cf756540e6b6d4e00709e6afae : SUCCESS in 46m 49s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/716c98b042854728aedf7f8e00cfdfb0 : SUCCESS in 34m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f83dab49b83740338ab83f8b8a9deddb : SUCCESS in 33m 55s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/360b23a1d75a4018a5ffc6c2abac87ce : SUCCESS in 54m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5707d7d6253a4759bc71bb579619ec1c : SUCCESS in 46m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/380a2a28d57d4bd8a33e426a2e2238e4 : SUCCESS in 36m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6eeadc9bd9b490abb4facbdfb1e64b5 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae3044de6d2e49ffa79819d4125416f3 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13d452ed0b01400ab6ca1c38649ec9cb : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/544e1bc23bf34477bf5b11e53a7d4693 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/677455bdcbac46a6aa015accc599bc86 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b2bb84ff6e84dc0b4803c919ec2d530 : SUCCESS in 6m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c35d86b689d745259a06ea136fbfbf81 : SUCCESS in 41m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f54b2058bb87431b0c8c93a413a5caebf925d7d9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-13 23:22:01.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"47c6862efe7cc5c73d2535241f4691e58acdcf9a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-20 19:27:20.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be4c9b63cf42eccafaabf63cf699450351ba0cd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 19:28:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6239f83c57d6c8c34d3e3115ca02ca24397f4a7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 22:03:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4faeae3675b4698af61e49525ed9f1f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fad82487226c4e248a8b5d6655342ace : SUCCESS in 42m 43s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/089f5ab8633841b5ba5eb6c0a3980afb : SUCCESS in 39m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3a0909cee4154dc78290b7f153218750 : SUCCESS in 40m 03s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/416eb8fea4c24c9ca6f7e0feea67bae5 : SUCCESS in 59m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/667ed45995604c9b884421c542068793 : SUCCESS in 41m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/75fd72e19efc40438e5209580286647b : SUCCESS in 41m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2740a046e6a8419491c346eb8bd3481a : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/936e93e4638243d0add27db148274c42 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51c93a10a53842348c4ceb75b84b615c : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93687dd196194a809c0c434ee63f1c80 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5b3d528f3c34f1cc5a3fcebf313d3f57174a3113","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 22:03:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b5336b06427e39d6296967e653c2d0f96d1fc3b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 22:05:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50ae0dc814b242139dfcc9d4ab2becc0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c844184743f4cf39f99eddfa363f93a : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9c3b63d46630d0f049239739d0d2e8973bd2784","revisions":{"c9c3b63d46630d0f049239739d0d2e8973bd2784":{"kind":"REWORK","_number":1,"created":"2023-04-13 20:19:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/70/880370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/880370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/880370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/880370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/880370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/880370/1"}}},"commit":{"parents":[{"commit":"81f3475971823ee82fb9c73f20bc64051d0cba36","subject":"[coordination] backend_url should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81f3475971823ee82fb9c73f20bc64051d0cba36"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-09 18:38:46.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-13 20:19:22.000000000","tz":0},"subject":"Use ids when removing ptr records","message":"Use ids when removing ptr records\n\nIf the record status changes during the removal process,\nthe current implementation will fail.\n\nChange-Id: I02b5d7499440154160c89ed63a2f70652fe72145\n(cherry picked from commit b3161ece234b5e2ee7658060aaa936af625bef38)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c9c3b63d46630d0f049239739d0d2e8973bd2784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c9c3b63d46630d0f049239739d0d2e8973bd2784"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
