)]}'
{"id":"openstack%2Fdesignate~880371","triplet_id":"openstack%2Fdesignate~stable%2Fwallaby~I02b5d7499440154160c89ed63a2f70652fe72145","project":"openstack/designate","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-04-20 21:17:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I02b5d7499440154160c89ed63a2f70652fe72145","subject":"Use ids when removing ptr records","status":"MERGED","created":"2023-04-13 20:19:30.000000000","updated":"2023-04-20 21:19:00.000000000","submitted":"2023-04-20 21:17:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880371","cherry_pick_of_change":879833,"cherry_pick_of_patch_set":2,"meta_rev_id":"5729757dde791b557d7b2f2d4ca0f49f65b5e440","_number":880371,"virtual_id_number":880371,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 21:17:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-04-19 13:32:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2023-04-18 18:37:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2023-04-20 19:26:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-13 21:12:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-18 18:36:48.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2023-04-18 18:37:03.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-04-19 13:32:34.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"26ae2911a0a7fb4e0fb96b4e17093d98c1f2a546","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-13 20:19:30.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f5d8b269beed751303258f9a4f62d96995bf2171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 21:12:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6b442b70d9a44f6b3e13701a539a969\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/240c30b2ad7542378a1781afc5d3701b : SUCCESS in 46m 20s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/402cfb24a7c6454aac9e000b8f421029 : RETRY_LIMIT in 8m 46s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d6303b53bdcb4357b9c8848a18ce1b07 : SUCCESS in 47m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8a79e76f91b44b82a48b79b070c24493 : SUCCESS in 36m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/27b37d05abe54185bee3ef29ac941666 : SUCCESS in 43m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ab3bb1885b204cbe91fe385cc76a6e69 : SUCCESS in 43m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8805bbdeb93c4e6eb8cbb8240568a781 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40ce2c1722cf45ec85ded2c05ff1a15e : FAILURE in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0e80d3dcc66427a9671ba3a5532b4c4 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dbd004226c446d299f9ecf43effb5ca : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55e875a4727144829d4a56bc691d21a3 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8f9db8ab97b48d7ba4fd119646bd1bc : SUCCESS in 6m 43s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/1945f23126584e158b90fde0faacb665 : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":1},{"id":"8e0fcaa034abd455b604dba9b1588177045bbaa7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-18 18:36:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"27daa7ade9fc19592a7d5ded96e84eefe484ce5d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-18 18:37:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91777abe73722f77cda37871e4e83e75af56a62b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 19:39:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d6f4434abf14990a510bc55fa55d208\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7d074214c3b5490daf5e163eea31eb47 : SUCCESS in 46m 44s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/c4f532b124cd472f9930e02004c06505 : RETRY_LIMIT in 8m 43s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f451737abb184257962964316ace13a2 : SUCCESS in 52m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/731bbee041094fa7bc2ba60201efc66b : SUCCESS in 44m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2ee4e2f32b64462aa6b772cc0d3bcde0 : SUCCESS in 50m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/beaeaf33d8164bababf798e3e9d47c07 : SUCCESS in 50m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13d3562b9c10489383798feee049ccc8 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eb2d5eaa7454f88986ffe08010b88a1 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98461786744d4b4baece42a96869d74f : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d2e03661a4a4ca3928290641ca44705 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57ef4c05ea3342ffa47a7dd676860560 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f96a1bf83aca4125ad61902adb0c3115 : SUCCESS in 6m 23s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/496c1cd51df54f0cb451758f257d4034 : SUCCESS in 47m 40s","accounts_in_message":[],"_revision_number":1},{"id":"cf07f8b9e3fb3d067ba8755f051aba6fef2bb3df","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-04-19 13:32:34.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6917b479e963cc8be9b53c54bfd2e1f01d34cdf","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-20 19:26:06.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c50857ff1ea1fcb6780aa2006d51c6ece5f525cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 19:26:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"073ba831463400869cb43b33b11a4e66bf4248e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:17:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dd0be9ca07345f98e2f8b6625c83da0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6bc9e8bee6e646e3b6580a7fe70545a2 : SUCCESS in 50m 10s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6319c2ec17264bb3b5d29323769a86f3 : SUCCESS in 46m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d906353a52644fa8a9311969c08c2f62 : SUCCESS in 34m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c4dbbab5395140548447c8bbe58acb2d : SUCCESS in 53m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/075fe3d3059145539b804860a10583af : SUCCESS in 53m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bebcfabf7574d5fa3d190a4f5c6f48a : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cd8e3de502c41df82f9a8624bc7794b : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e177f3e531941e1a7495916f897b229 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e92f2d418391484fa844561f02be87ba : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":1},{"id":"383166da7626e9fce98ce77c1ea47259a7a67094","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:17:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5729757dde791b557d7b2f2d4ca0f49f65b5e440","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:19:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee646b5e967345ddad90bcce40639b48\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10241c4e058f4804852c0c59ea66e199 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0257040c186195ce13e4cfa8ee1b524a4451833a","revisions":{"0257040c186195ce13e4cfa8ee1b524a4451833a":{"kind":"REWORK","_number":1,"created":"2023-04-13 20:19:30.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/71/880371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/880371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/880371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/880371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/880371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/880371/1"}}},"commit":{"parents":[{"commit":"45801afaaf085c606fb84ea5059ef82ce1c94c52","subject":"Pin designate-tempest-plugin for stable/wallaby jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/45801afaaf085c606fb84ea5059ef82ce1c94c52"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-09 18:38:46.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-13 20:19:30.000000000","tz":0},"subject":"Use ids when removing ptr records","message":"Use ids when removing ptr records\n\nIf the record status changes during the removal process,\nthe current implementation will fail.\n\nChange-Id: I02b5d7499440154160c89ed63a2f70652fe72145\n(cherry picked from commit b3161ece234b5e2ee7658060aaa936af625bef38)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0257040c186195ce13e4cfa8ee1b524a4451833a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0257040c186195ce13e4cfa8ee1b524a4451833a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
