)]}'
{"id":"openstack%2Fdesignate~880372","triplet_id":"openstack%2Fdesignate~stable%2Fwallaby~I15864a76c8913f151154e2398ae86aea71883498","project":"openstack/designate","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-04-14 00:23:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I15864a76c8913f151154e2398ae86aea71883498","subject":"Removed unsupported test B309 from bandit","status":"MERGED","created":"2023-04-13 20:29:36.000000000","updated":"2023-04-14 00:24:18.000000000","submitted":"2023-04-14 00:23:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880372","cherry_pick_of_change":877150,"cherry_pick_of_patch_set":2,"meta_rev_id":"9a8723660b1b9d05ea690fbbe7c0c1e3f4b1e517","_number":880372,"virtual_id_number":880372,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-14 00:23:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-04-13 23:18:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-04-13 23:18:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-13 21:44:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 23:18:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"3c235288bc8f7d4a3ec51cd3b2aa4047fcdd9ca7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-13 20:29:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7c98dfd6aaad4f991a8069b4f48c3f1412a883b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 21:44:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0797e101d4484626b7600c1fa3a4ae8b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/496fd1c771764fcea0503807df341335 : SUCCESS in 44m 31s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/13fc38b13ce84269abdd149517f31b45 : RETRY_LIMIT in 18m 11s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/92d80a38410349d0ad8e48ffef1b9441 : SUCCESS in 54m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5c6bbf849f104b6880e6f8213e7d5cdc : SUCCESS in 47m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a8a9ac5139834b87b58e19b4ad9a0044 : SUCCESS in 55m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fccdcfd3db9a49c0ac7a352dcbde9c65 : SUCCESS in 56m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa5bd8a09b8748c68872eb1ff0dfce22 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61ad70cde4de4b2d94611c4992f95132 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3e2c71faf524608be25aba6a386eab1 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbd97360b6ac4f65b2e54d819f234a9f : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc7907426cb841159507ad6a4fc2a777 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b34508151c4e4d7486fe6b07a3241793 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d1fa000537843a881b62fb371ec6531 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"0c8b691710b492fdc104102db9f907f8bf17eddd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-13 23:18:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc2bbfcf1ddb5345615997892510d81689e7afa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 23:18:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7cce9800f15a2469d10f8480991faa64bb667897","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 00:23:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab9dd7e1e2d24b37b17617a79a9df0a9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6ea5ab1e620d49b79f63a0ed6966e528 : SUCCESS in 50m 45s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e7a199060f16430d89dd433188fb4dc3 : SUCCESS in 48m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9ce8871892e44d788f9c01f3754f82b5 : SUCCESS in 48m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/111c436ab23f4fd2949573ca42d688f7 : SUCCESS in 46m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/055f5d67b8c147609d93866b268e773a : SUCCESS in 58m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4b0db3ab9a34eee819fe99cea29c39e : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0abe255c755f42988e482fb73f768154 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c66e1f78030c4cc2807ae5dbc64b8403 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba360b077747495295396b0ddb842ec0 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5e706efc08c43629a5021d4191c7bac : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"588540ba5f1e3cd92fc2ca574ba926ad93d73414","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 00:23:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9a8723660b1b9d05ea690fbbe7c0c1e3f4b1e517","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 00:24:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd20ab2cf8b542968b1f3a012d464da7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6881e3b666c4f88a328465551b51948 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27bfcdc025e74314bbc650e71d1703e2 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef49a7614ccacc8729cfadaeda897dce7fd01a7c","revisions":{"ef49a7614ccacc8729cfadaeda897dce7fd01a7c":{"kind":"REWORK","_number":1,"created":"2023-04-13 20:29:36.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/72/880372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/72/880372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/880372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/880372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/880372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/72/880372/1"}}},"commit":{"parents":[{"commit":"45801afaaf085c606fb84ea5059ef82ce1c94c52","subject":"Pin designate-tempest-plugin for stable/wallaby jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/45801afaaf085c606fb84ea5059ef82ce1c94c52"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-03-11 00:47:19.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-13 20:29:36.000000000","tz":0},"subject":"Removed unsupported test B309 from bandit","message":"Removed unsupported test B309 from bandit\n\nB309 was removed with this patch and breaks the CI.\nhttps://github.com/PyCQA/bandit/commit/130a467434ff6b8f9bbd83100e409bc1fcb6f407\n\nChange-Id: I15864a76c8913f151154e2398ae86aea71883498\n(cherry picked from commit 758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ef49a7614ccacc8729cfadaeda897dce7fd01a7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ef49a7614ccacc8729cfadaeda897dce7fd01a7c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
