)]}'
{"id":"openstack%2Fdesignate~880374","triplet_id":"openstack%2Fdesignate~stable%2Fussuri~I15864a76c8913f151154e2398ae86aea71883498","project":"openstack/designate","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-04-20 18:20:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I15864a76c8913f151154e2398ae86aea71883498","subject":"Removed unsupported test B309 from bandit","status":"MERGED","created":"2023-04-13 20:29:49.000000000","updated":"2023-04-20 18:21:36.000000000","submitted":"2023-04-20 18:20:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880374","cherry_pick_of_change":877150,"cherry_pick_of_patch_set":2,"meta_rev_id":"7fbec73db1f8119eb4ddd2c8f6f8237b8d5c0e7d","_number":880374,"virtual_id_number":880374,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 18:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-04-20 17:04:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-04-20 17:04:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-13 21:15:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-20 17:04:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"f2302edae3e090be7a29de07ed0295fa7bb26931","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-13 20:29:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7d9d673d5a18ea35f2828d478ce4aab3eb9b9e75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 21:15:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d92103fe848a4491bc47d5066563ce54\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d4f7ed33b6ea46e8a7db6f2cb8f30b89 : SUCCESS in 44m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cd457680b3de454b82532abb5f2feec6 : SUCCESS in 42m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6330b5818cb447088077d0c9595956a2 : FAILURE in 17m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a510a00e344347818735f2c505a7cbbb : SUCCESS in 40m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2f738cb5541046d7b08166a6df7b615b : SUCCESS in 44m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9b89bf748b04ce5bda19baf0f290d38 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ecbe1c2683f4fe4aa32a88c0fb736f8 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d260bcc7a7949079f89b53c5c5736de : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44f21000f75e4304a6e9244edf1b7c98 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a241c815195b4265a3fb41236f47dd35 : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ed6215d4db54d83891537cbbe0d8143 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"aae1d8e7d49e507f2b93ab51ae78947810f0f760","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-13 21:41:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81c088bc0af8cfc11baf78a40180548b8c1559e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 22:09:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8cf5da887a4a44aea9656818c0ad635b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c76ccac45f874962a5f88537225a9416 : SUCCESS in 27m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5ac5c63bc51a41b9b0afb89cc5fa19dd : SUCCESS in 24m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c7c22e114c4f4fc084fa62e2059e68d8 : FAILURE in 6m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7cbac09bd086494a96019e748032488a : SUCCESS in 26m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/07e8b0c0feeb485089cd1bf5a9190324 : SUCCESS in 25m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/248d4e0a38b5463ca7d6c150290c1c86 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33f1f197dbb443a780d07ddccd179e61 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10a6795201644643a5e52557b7645181 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8966d137b3374eea98ae1b71ede94b1c : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc3aa34964a44a70aa8452c32a397d62 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67f601a142ae4848a0ba0c7bdcbd1d48 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"15565cfaf5282df80fed91c1cf710962f868ed28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-20 15:34:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9f250aacc5c6ca832f6bf29b3f6af3e32fee821a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 16:45:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da18c1280b61475b9cddfffdb8b803b9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7abc1951d20b4d3099d412597cf86c93 : SUCCESS in 32m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d95f665954f74519a2ae1878bdc043bb : SUCCESS in 28m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/904f5d0374d34775b28b280446bd67b0 : SUCCESS in 51m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d2894ac2aa574ca8a8b0414de55af11b : SUCCESS in 33m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/25dbb161fcdd41eea66a1b3d07d95863 : SUCCESS in 32m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6fd8c9ea8c94e3ebfd4b019ef0beca2 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db29a307280b48ec88595a3758363788 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79309687c28b4ab2b54062100418f25a : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/349b8041aeb44e3799122881cfe58abe : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ddcc09f9a0e4bd98e30bf2af4f71430 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b50d1125b9974b3dabbd3fbfed32a22e : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"aed13ddf7aec6f8dfacb7bc6747b62335ef792dd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-20 17:04:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f19fbdd2c63b74ce590a7cd8521744a5e287120f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 17:04:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b77e690d8eaf67555819be29533061cb2648f0ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:20:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f02fd38eae764832af34105064f0cb59\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/370a40f93a344ab5846229e7bde2d6c8 : SUCCESS in 44m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bd2a85c1155c4ac8a2d86d186c7c1ee9 : SUCCESS in 46m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4fde62ff456a4316b7c0a46809b7d5a9 : SUCCESS in 1h 05m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4d650bb4249b4478bb797f83de3fc52c : SUCCESS in 40m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/738cac5cf93241ddb78c730983653287 : SUCCESS in 34m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d44a2a0526e4741803b34e278cfbcf3 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f130b42578c4ea8a79e7f7985d9cf74 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d79960ecf64412f81bf4b233e879d6a : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b92c33f641cc4270875ea69f4d81a8cd : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/914a980f19a549daab5595cd70b935a5 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"cca58068530cdf6bbf7555e13c2fbe317679a727","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:20:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7fbec73db1f8119eb4ddd2c8f6f8237b8d5c0e7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:21:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc7fd3d7c206458986c49d1027051517\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3e2e0598079440bb35457cc74916bfa : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90edd49f8b2f4dec8af25cb4686b8cc3 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dcb5d6ca0f00a5d0f487ac6e0dbbf0b5ddf26b1a","revisions":{"5f224fadd700e2a7a23c90ac147eed423a4de567":{"kind":"REWORK","_number":1,"created":"2023-04-13 20:29:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/74/880374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/880374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/880374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/880374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/880374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/880374/1"}}},"commit":{"parents":[{"commit":"bdbbe6a396682f316304027845c818f94e6f4cd3","subject":"CI: Move queue configuration into project-template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bdbbe6a396682f316304027845c818f94e6f4cd3"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-03-11 00:47:19.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-13 20:29:49.000000000","tz":0},"subject":"Removed unsupported test B309 from bandit","message":"Removed unsupported test B309 from bandit\n\nB309 was removed with this patch and breaks the CI.\nhttps://github.com/PyCQA/bandit/commit/130a467434ff6b8f9bbd83100e409bc1fcb6f407\n\nChange-Id: I15864a76c8913f151154e2398ae86aea71883498\n(cherry picked from commit 758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f224fadd700e2a7a23c90ac147eed423a4de567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f224fadd700e2a7a23c90ac147eed423a4de567"}]},"branch":"refs/heads/stable/ussuri"},"dcb5d6ca0f00a5d0f487ac6e0dbbf0b5ddf26b1a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-20 15:34:45.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/74/880374/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/880374/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/880374/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/880374/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/880374/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/880374/2"}}},"commit":{"parents":[{"commit":"64b56187185a62780f79017a271d9962453bc442","subject":"Update Ussuri grenade job to use zuulv3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64b56187185a62780f79017a271d9962453bc442"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-03-11 00:47:19.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-20 15:34:45.000000000","tz":0},"subject":"Removed unsupported test B309 from bandit","message":"Removed unsupported test B309 from bandit\n\nB309 was removed with this patch and breaks the CI.\nhttps://github.com/PyCQA/bandit/commit/130a467434ff6b8f9bbd83100e409bc1fcb6f407\n\nChange-Id: I15864a76c8913f151154e2398ae86aea71883498\n(cherry picked from commit 758956d5b9321c72d63cbf3d8a1987d5dc6e6bb3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dcb5d6ca0f00a5d0f487ac6e0dbbf0b5ddf26b1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dcb5d6ca0f00a5d0f487ac6e0dbbf0b5ddf26b1a"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
