)]}'
{"id":"openstack%2Fdesignate~880377","triplet_id":"openstack%2Fdesignate~stable%2F2023.1~Iebe609e5f365d03e99f2a4580671175b4642763c","project":"openstack/designate","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-04-20 21:03:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iebe609e5f365d03e99f2a4580671175b4642763c","subject":"Use ids when removing sink managed records","status":"MERGED","created":"2023-04-14 02:00:36.000000000","updated":"2023-04-20 21:04:33.000000000","submitted":"2023-04-20 21:03:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880377","cherry_pick_of_change":879832,"cherry_pick_of_patch_set":13,"meta_rev_id":"4883d29b38bd7a8a3993e314dcffac51fef447a2","_number":880377,"virtual_id_number":880377,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 21:03:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-04-20 15:08:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2023-04-20 19:24:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2023-04-20 19:24:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-14 03:25:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-20 15:08:54.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2023-04-20 19:24:14.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"cd95815f7f5365451761a2fc0e2d5f97c145f9c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-14 02:00:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7b69e99c7ac967094281bfd6b016486b741dbac6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 03:25:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2fddd89afe1c4938885fff9bf8370435\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f950a62601994243bf08993ea4d03224 : SUCCESS in 50m 13s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c85e7d4fba864ad5bf2fd3449bd24f69 : SUCCESS in 51m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6bf98e699c5949ada738240090a71607 : SUCCESS in 39m 56s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/bc09cfa2b39a4de4987f078b7904998e : FAILURE in 49m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2208765d26bf44dcb084bf60d8b5d9c3 : SUCCESS in 50m 26s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/23f98ea9f479471082f338c475ca3861 : SUCCESS in 1h 13m 37s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bcf7fa81a6eb4ccfa91638fc9138dc45 : SUCCESS in 1h 07m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1346a83c4a724ef1af8fbad36675f9f2 : SUCCESS in 33m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2f75f2694e9a4d8d8046c88882c54fd0 : SUCCESS in 39m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b396c2a620614a86b2c275083aea9f71 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f509f0f293ad46b2b5f31c9c70f45536 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0712c23a7da8459e9b6fe57ecd826a2a : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/469cbde07b3a447492c577495ceb4504 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa9ff39f5454f0b98e08a3f934c9c14 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31158fdc95024ade8868bca10033a56f : SUCCESS in 5m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/101a24701aa04217ba917dcfe63d23f7 : SUCCESS in 36m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6f0150b1c8034385bfe6ab11ba67c55b : SUCCESS in 5m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc3d5c147c7276bfa5653400f7f249e59ba934de","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-14 17:28:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e79e741a6206117de997af6796b1e0a61ec25f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 18:42:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbcd3d65e6264c8f88ecbbf7f4602080\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dc4c6307716f443eb023eb17971f1815 : SUCCESS in 47m 15s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/525da2cdfe194da7a2e7e7d138e77f57 : SUCCESS in 41m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f05a61dc51c4448ba61dbabf875a6b49 : SUCCESS in 35m 59s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2718a7a91aaf46ca80901d3a1d3b8fdb : FAILURE in 38m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bfefd346a4c542b98aca0c95c5368b6c : SUCCESS in 40m 16s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8515b1543030456c836e71c094cbaf66 : SUCCESS in 1h 08m 11s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/22aa43fdc9f54b4db4a7ff42c33852cc : SUCCESS in 1h 05m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/590d9fd2f72d4207893989e92ec48d3f : SUCCESS in 46m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/421abb2c1e5a44469265b6f3c41831e0 : SUCCESS in 47m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29684de427214e459bb61e95edca2513 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45fbd08f2c2145d9ac1f1c1b520b2e3e : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c37f5e57bc0e4749a14712d4b27890e8 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b293c74728264beb89b155d6bd59c465 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fc02df37f6c464982b035c541804216 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eafc67ddfb184c1a983b5cfee3c0e4ff : SUCCESS in 4m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fb68f6f4af3b4f00b289206aafd0e02a : SUCCESS in 31m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f85024ee87a44a65b408c9d5f3b097fe : SUCCESS in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8d20869df7eb9c8dfd719c7990a10fe5187bf02e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-20 05:33:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b1352bc90cb66ae290a9ebc968ee3619307bea8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 06:41:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8aa3f7521404e15ba9f6126aec96de0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/739b5d8f514f4cf6a3316e0cb955145a : SUCCESS in 38m 49s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6896d5b07f9f407abed107071d833650 : SUCCESS in 50m 12s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b45713d8783143c0a79686c70db51da1 : SUCCESS in 47m 08s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0b68d43bca9c438eb3d321d8c1417527 : SUCCESS in 36m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/45ca30d772094e85b962e17dd38b4a32 : SUCCESS in 36m 29s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e493041086324014acf5f775863bb475 : SUCCESS in 56m 11s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/319aa870a20d48f1adfde6a504832f51 : SUCCESS in 1h 01m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c36560c1e8e64642ad34763baa573967 : SUCCESS in 35m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4eca00692b704377915d7bb21a8c9436 : SUCCESS in 36m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71b98e76637a48109d677d94d614edd7 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cb4dd4a04fa4bcdbf6958d5ec08e683 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ede4bdea5cd45c8adb5e41d0c01338c : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/658543ffb5764b1e92a4139a1a9e746f : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37d90ca0586646dcb9c55f07b641242d : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b90488f03b44623ab60db9dfa5871ec : SUCCESS in 5m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fc690f086bc74ef1b8476cd5c25be53d : SUCCESS in 38m 31s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1d04d313b7fa485f885a85699a5327fa : SUCCESS in 5m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"748c8e2730fe2f16d0c3e977e101c5d729230c45","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-04-20 15:08:54.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4cf93653a07ff045e9df5620cfb1b133316e1007","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-20 19:24:14.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ff44619964b11d577b4842226891bbb51e2d890","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-20 19:24:32.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d44a91f6c6a0cd65ed0d0998a93fa9e65a1c5eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 19:25:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f2a23933987a7ec8f5fdc3f8259b84b6070e9d54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:03:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c7f3c3636e4457ca0c64ef9613ccbac\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4af02e76680047099b25e7c542a607f0 : SUCCESS in 37m 04s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d1db31845fae4d13add4e2cce65dd6c5 : SUCCESS in 37m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b676d44e0bda41cd8c960054f80cba05 : SUCCESS in 42m 33s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1fba18322d184c9f9aede5fe49d439ab : SUCCESS in 57m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8f28db95860c4bb3849587fe56420e28 : SUCCESS in 36m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fc50f55c61284835b7edf0f20ef0bf3b : SUCCESS in 37m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97e11809bffd43bf8377234f9f0772e8 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e63ce7d0b894476b604e27c0f862637 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01f74e4784aa4e2281d21f0e4bc2d9b5 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1d1062b50354621b6bc8e0eac6b92cf : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2726ecb7e1be4e5a9737728e601c6bd3 : SUCCESS in 3m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dcafea1c744b4c50afa760a9271b5d4f : SUCCESS in 31m 28s","accounts_in_message":[],"_revision_number":1},{"id":"1814e476aa7b7876fbe12a666cb5b1c8ec6277f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:03:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4883d29b38bd7a8a3993e314dcffac51fef447a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:04:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58fa38d9382d4697b946ed635a879a6a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9ba54a6f9034486bf5e4aa096212148 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/720c6bb1b8774455accc36124b7ac9cb : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0721ced813c855d4edd2053e022a3c35d5e67566","revisions":{"0721ced813c855d4edd2053e022a3c35d5e67566":{"kind":"REWORK","_number":1,"created":"2023-04-14 02:00:36.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/880377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/880377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/880377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/880377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/880377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/880377/1"}}},"commit":{"parents":[{"commit":"4c08082bea6793334d3a215cdc0c3f051cdf6785","subject":"Fix sharing a zone with the zone owner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4c08082bea6793334d3a215cdc0c3f051cdf6785"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-05 05:04:56.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-14 02:00:36.000000000","tz":0},"subject":"Use ids when removing sink managed records","message":"Use ids when removing sink managed records\n\nIf the record status changes during the removal process,\nthe current implementation will fail.\n\nCloses-Bug: 2015762\nChange-Id: Iebe609e5f365d03e99f2a4580671175b4642763c\n(cherry picked from commit 2cb42ac9f6e5273ad64e697919b62e029466078f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0721ced813c855d4edd2053e022a3c35d5e67566"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0721ced813c855d4edd2053e022a3c35d5e67566"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
