)]}'
{"id":"openstack%2Fdesignate~880378","triplet_id":"openstack%2Fdesignate~stable%2Fzed~Iebe609e5f365d03e99f2a4580671175b4642763c","project":"openstack/designate","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-04-20 21:07:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iebe609e5f365d03e99f2a4580671175b4642763c","subject":"Use ids when removing sink managed records","status":"MERGED","created":"2023-04-14 02:35:13.000000000","updated":"2023-04-20 21:08:15.000000000","submitted":"2023-04-20 21:07:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880378","cherry_pick_of_change":879832,"cherry_pick_of_patch_set":13,"meta_rev_id":"5cf7d11bb11924bc8b672f39a5c7d1cfb0c37ed4","_number":880378,"virtual_id_number":880378,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 21:07:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-04-14 15:25:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2023-04-18 18:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2023-04-20 19:25:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-14 04:50:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-14 15:25:08.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2023-04-18 18:37:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"1e1c17ce659f2af2a31ac136c79ce5b0c23be598","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-14 02:35:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* designate/tests/test_notification_handler/test_nova.py","accounts_in_message":[],"_revision_number":1},{"id":"9773737617f7e181b65cf42c460e4f5a27f3f185","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-14 02:36:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c92176c70759e43f1d2cbce21b962569a63c736a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-14 02:37:27.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"8fbc80afd957c707ef5de7f537845dc8e0c86d0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 04:50:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66fbf13b1bb44be9a6e033013abbfce7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7c160ccd958845cd816ce2719db64dab : SUCCESS in 40m 12s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/de8afab55d804160be58acac77bf33c5 : SUCCESS in 51m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/03b9b402d80f4b5680577a4ca41cf489 : SUCCESS in 45m 04s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/79f927ab677644bfbcbf2917c7d67ae7 : SUCCESS in 52m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9a649552341b482e9ca7049fb0379e19 : SUCCESS in 1h 12m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5010e73c391546dea433291e115b5bd3 : SUCCESS in 1h 34m 23s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6cf75329a1b042a0a6c54bbf5a9cba86 : SUCCESS in 2h 04m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7190df625477482abbe9a3d32b417f2c : SUCCESS in 1h 41m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/be8d261e9e33405d8bff63f733e95307 : SUCCESS in 1h 09m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15448cf67f294516be77fcf5593b4605 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a51605c3d2c44058afe1f79d45e419c6 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2344636881134fe6bb46b772cde99044 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75fd0e3b4e92456d86929f2c2d8080c2 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce01ce94176d467fbefe0f294a64a9bb : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae750df95b1c41539cfb9866c81d69a7 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce3e05d94e5948de8cb93e4ce91be4e7 : SUCCESS in 4m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c9981d77cdb74042b53003e50cf3dd72 : SUCCESS in 27m 32s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a6e84bec1645446595d6f2b6f8e87fd3 : FAILURE in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8c973969f47ea1b7a4a2fa892ea1e06021fd32d7","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-04-14 15:25:08.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ef7171ef976b8acd9fc3904a82e9c0e606474b5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-18 18:37:57.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"77aa859e5b588560b8cc970a4e7a64425fb4398b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-20 19:25:29.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b9a8fe511bccbd0459cceb779a7bc749002be18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 19:26:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4853f37a828c180ead9d8dae4fa96a70a2a5c714","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:07:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1126e58f93c04c5b9a24eca15a8cbea1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7113999b5d3b483dae6ae4c6029c659b : SUCCESS in 41m 48s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4cd3f68c08d346b885f050f134354926 : SUCCESS in 41m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/da363922944349a1a2c96de6edf08a35 : SUCCESS in 38m 36s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1ffc31b66f924f9f84eaef1d9060f440 : SUCCESS in 57m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/afb18a9ad22844b98f28665e6e93acb5 : SUCCESS in 41m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/88ccd16a44094e7db3d2f5085794cfc0 : SUCCESS in 39m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7f86e4c5cb34ebca2a1a1b4aec6b85a : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47c33680e23f42459e003801e3653c46 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/832431c63cbc4ca6b4c6c1911bb1018a : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5f75756215647518f4ad57a05913afe : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/956c8056856d4922be4c9a5ebae85b30 : SUCCESS in 4m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0281a1061b9b446590bec9616ebe182b : SUCCESS in 34m 45s","accounts_in_message":[],"_revision_number":2},{"id":"1a054f5ac8fcf9feee4306d99f7981a6716d175e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:07:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5cf7d11bb11924bc8b672f39a5c7d1cfb0c37ed4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:08:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c4c28df8ca14b329849975427601e87\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2308b45f61a4196be4872a1ddca330c : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ceabea82a11422e9dacbdaa202bbc74 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b518dacca9ea997672ce0f8557bf79db6d8d190f","revisions":{"163d531768c149becf1beb97c71558b176b229e6":{"kind":"REWORK","_number":1,"created":"2023-04-14 02:35:13.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/78/880378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/880378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/880378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/880378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/880378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/880378/1"}}},"commit":{"parents":[{"commit":"be775801e33313305b5d90dca10bae166e6938c8","subject":"[coordination] backend_url should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/be775801e33313305b5d90dca10bae166e6938c8"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-05 05:04:56.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-14 02:35:13.000000000","tz":0},"subject":"Use ids when removing sink managed records","message":"Use ids when removing sink managed records\n\nIf the record status changes during the removal process,\nthe current implementation will fail.\n\nCloses-Bug: 2015762\nChange-Id: Iebe609e5f365d03e99f2a4580671175b4642763c\n(cherry picked from commit 2cb42ac9f6e5273ad64e697919b62e029466078f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/163d531768c149becf1beb97c71558b176b229e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/163d531768c149becf1beb97c71558b176b229e6"}]},"branch":"refs/heads/stable/zed"},"b518dacca9ea997672ce0f8557bf79db6d8d190f":{"kind":"REWORK","_number":2,"created":"2023-04-14 02:36:40.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/78/880378/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/880378/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/880378/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/880378/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/880378/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/880378/2"}}},"commit":{"parents":[{"commit":"be775801e33313305b5d90dca10bae166e6938c8","subject":"[coordination] backend_url should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/be775801e33313305b5d90dca10bae166e6938c8"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-05 05:04:56.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-14 02:36:28.000000000","tz":-420},"subject":"Use ids when removing sink managed records","message":"Use ids when removing sink managed records\n\nIf the record status changes during the removal process,\nthe current implementation will fail.\n\nCloses-Bug: 2015762\nChange-Id: Iebe609e5f365d03e99f2a4580671175b4642763c\n(cherry picked from commit 2cb42ac9f6e5273ad64e697919b62e029466078f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b518dacca9ea997672ce0f8557bf79db6d8d190f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b518dacca9ea997672ce0f8557bf79db6d8d190f"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
