)]}'
{"id":"openstack%2Fdesignate~880381","triplet_id":"openstack%2Fdesignate~stable%2Fwallaby~Iebe609e5f365d03e99f2a4580671175b4642763c","project":"openstack/designate","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-04-20 21:36:10.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Iebe609e5f365d03e99f2a4580671175b4642763c","subject":"Use ids when removing sink managed records","status":"MERGED","created":"2023-04-14 02:39:43.000000000","updated":"2023-04-20 23:27:52.000000000","submitted":"2023-04-20 23:26:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880381","cherry_pick_of_change":880380,"cherry_pick_of_patch_set":2,"meta_rev_id":"6eb1e940638ded40f3083bacdff0badc22279124","_number":880381,"virtual_id_number":880381,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 23:26:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-04-16 00:32:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2023-04-18 18:39:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2023-04-20 19:26:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-14 04:28:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-16 00:32:51.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2023-04-18 18:39:11.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"15a59681d12fb111cbdfdb11facde92c7d72209b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-14 02:39:43.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"582cc7317689c07babfd38627fa432a323d7e29f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 04:28:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac89cf25a097480f980e4a680c0d77f5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/540de9dea044481ea46dcae4ce70fd07 : SUCCESS in 52m 19s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/0c47094522f5493595cc85ba46838d34 : RETRY_LIMIT in 22m 36s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3e7bba480a6d4f68a29c834ddecce120 : SUCCESS in 52m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fd36a0b723454838a9ad20e3aa821112 : SUCCESS in 53m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f49b44c3314641de9ba5a2c5b3525651 : SUCCESS in 44m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bf689ac46a9e43779940eb9bbfef04c0 : SUCCESS in 1h 24m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/180804243f7744a384266e8793971213 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7af8440d113348f5af914e9c997375c5 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1691f886cdee4c1db445b111d42347c6 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5ad340c39ee4e7a9575c87f06d88213 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/123e8c67a81043048e2f42069f996588 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9779f2778ef4d2e9867cf2bb9795e96 : SUCCESS in 8m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e66831abb067468b8d40e41e8bc56ad6 : SUCCESS in 3m 53s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/91caba74bd44450386078836bef40681 : TIMED_OUT in 1h 02m 15s","accounts_in_message":[],"_revision_number":1},{"id":"7f614a6b61327aaa68ef5d804c654025a778a022","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-14 17:27:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"86a361967929b593444c4ca8dcef890232542a7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 18:31:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2667345a64944219af5588b641bc0aed\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/64bf4b8cdc2b48d28fbd68eedd06a8d0 : SUCCESS in 42m 21s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/b842c33ab91a468abfef51f084615c52 : RETRY_LIMIT in 8m 23s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/bf63ef2599f94503bf9e7e022dd68ddf : SUCCESS in 55m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a8e97c93caea4b77979a4daf1c401ab4 : SUCCESS in 50m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/811f78f940f0468fb642931237e643eb : SUCCESS in 48m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7e599da5301648a5b113eb2b43e32ac2 : SUCCESS in 49m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0f88f95395d404f97aa364004faa83a : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c1cfae2a7bb4694a5fa4ed785c9b624 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/489401876d3449ec88c95ee445fb41de : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90a3bbee5c6f487b9f97959270038beb : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02237396e1b04e658cdee1fab9e6f1e7 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/698bdb49f6a84565a81f72d70e05fc7b : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dd6c6a46d3a421fa96700d80777a6d4 : SUCCESS in 4m 08s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/5275a2dace2141589599ba99c79085f2 : SUCCESS in 42m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1262606504d6f44221593692bdfac519fe87bcf7","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-04-16 00:32:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a24ee4b98b030ea9f3064b637291956f9c6a095","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-18 18:39:11.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5796f12a5263e860847d94c77bb3325c36e698c0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-20 19:26:37.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84468fd6a00e15a8f4ff9f17ef689f9ed97794b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 19:27:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"066205c317530c3715f143a4ec41aea2ff5e92c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:17:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a8704a112db4890b1d1b4c2f01cd140\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/925df5d474194e9e91c9aa65aab3aa04 : SUCCESS in 41m 46s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3ee4916c9189467d9feca7e7babb8d2c : SUCCESS in 47m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/959d5498dab64695855e4ee7aa0f13cc : SUCCESS in 38m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3f78541ad7e14de2a934a93cde93a97a : SUCCESS in 43m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fa8416530d0943f49d7837208e204cc3 : FAILURE in 43m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/126cbff1c0034d23b58ba4f66aa36e65 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68e383a737b24645bff08850724fe3f3 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb4fae74d999479bbd13ae35e621d2f6 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2165ba58a2c44828a2c7a5860901f941 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/882351e736c448c080f0683cc7efead8 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ac578dd919ce0e0f01d545a3e4401b2d53c579d3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-20 21:36:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef84c22846a9014f1eea50ddc94cdc98ce94d55f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 22:38:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41dbf03f345d4d0ea7da9d2c928b208d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b836eaaaffa442cd96a32c7020e19478 : SUCCESS in 56m 49s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/569ea5ca68d84dab853957af37aa775e : RETRY_LIMIT in 18m 50s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8e46fd8169ef40228de7c7a48318d85b : SUCCESS in 50m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fb4dd219464b45798cf1ac228c1f7fbd : SUCCESS in 48m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/23b6822bb52b41ac8d75a248483977b6 : SUCCESS in 48m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d7900e331a7d4f3da3582d9b71a205ff : SUCCESS in 46m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/390bf9db3b1a4b998bf4421da9e34474 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b475bf37ec36482db7126f3309b0a7bb : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bf5e3fa2ae54a9e8a3629412669847e : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff3d1d7e38f74a85a62e80d0e9db196a : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/274c9f60334649eabbfd9203cb551b17 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de569b6f27eb4bff9688d2ceebc30d90 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc0bb2038b1c4747aa42104c46343d1d : SUCCESS in 3m 59s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/da6639b0e911433cb9c23d30ba5563a3 : SUCCESS in 42m 48s","accounts_in_message":[],"_revision_number":1},{"id":"05cdc03cefa7b3e8af424fc013d4d5eeecdb9bf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 22:38:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4219323eb6509fbbcd9d7fb38ebb6190c1f75f69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 23:26:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f80fb85c5d334f05bc267b8b810ad044\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fe7c5e4a56e24505aceffc4c24b10e34 : SUCCESS in 44m 11s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1b2fbf17d98c4f5bb108231d8e8b8c7c : SUCCESS in 42m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/49d930bad16a455fb7d6bd8c47934dd4 : SUCCESS in 40m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/52f59555fe694270acb7da16b70faa35 : SUCCESS in 43m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ecc96d6752be4f9887f48d32b9150362 : SUCCESS in 38m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/181d90c1d72f4afca6fe170a538793a3 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76a1190455e14815a58840eced83784f : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d19600817ee14383ac978d5f3d7950a5 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/704a4af4cd4d41d88f080615c828764a : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61f4c86bc7e54654b2702e1cd394d236 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"5be0f6c4481ac4c5eda8fa6c43b367b517a230c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 23:26:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6eb1e940638ded40f3083bacdff0badc22279124","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 23:27:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e83d4b344654d57b9a58efab9846ba3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb862095ed954572aa07b5f4fddb0a2c : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dae87f0385a49da8e39f5fcd7e88ef8 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"abb5263439eb6d95c82232e203a3ad2427dc6e19","revisions":{"abb5263439eb6d95c82232e203a3ad2427dc6e19":{"kind":"REWORK","_number":1,"created":"2023-04-14 02:39:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/81/880381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/81/880381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/880381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/880381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/880381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/81/880381/1"}}},"commit":{"parents":[{"commit":"ef49a7614ccacc8729cfadaeda897dce7fd01a7c","subject":"Removed unsupported test B309 from bandit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ef49a7614ccacc8729cfadaeda897dce7fd01a7c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-05 05:04:56.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-14 02:39:43.000000000","tz":0},"subject":"Use ids when removing sink managed records","message":"Use ids when removing sink managed records\n\nIf the record status changes during the removal process,\nthe current implementation will fail.\n\nCloses-Bug: 2015762\nChange-Id: Iebe609e5f365d03e99f2a4580671175b4642763c\n(cherry picked from commit 2cb42ac9f6e5273ad64e697919b62e029466078f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/abb5263439eb6d95c82232e203a3ad2427dc6e19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/abb5263439eb6d95c82232e203a3ad2427dc6e19"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
