)]}'
{"id":"openstack%2Fdesignate~880382","triplet_id":"openstack%2Fdesignate~stable%2Fvictoria~Iebe609e5f365d03e99f2a4580671175b4642763c","project":"openstack/designate","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-04-20 21:17:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iebe609e5f365d03e99f2a4580671175b4642763c","subject":"Use ids when removing sink managed records","status":"MERGED","created":"2023-04-14 02:39:51.000000000","updated":"2023-04-20 21:19:10.000000000","submitted":"2023-04-20 21:17:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880382","cherry_pick_of_change":880380,"cherry_pick_of_patch_set":2,"meta_rev_id":"8a5e8943b1610fc803a1df82ddc1ecb0ef1ee731","_number":880382,"virtual_id_number":880382,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 21:17:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-04-16 00:33:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2023-04-18 18:39:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2023-04-20 19:26:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-14 04:17:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-16 00:33:14.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2023-04-18 18:39:33.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"3f8c03af57f8549f6e4566f3d25864f958d18b24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-14 02:39:51.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"e54a0194236e85e8987b4653ad3ce70fbd2569dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 04:17:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c730c9a770a499f9f3d1123227348a8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2d6385fdca194ea791b9d849ea5be759 : TIMED_OUT in 1h 13m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/47997410ab8140b8850c1697eef39d61 : SUCCESS in 1h 12m 45s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/85d0640d26bd4474a21fab6f798fcf0d : SUCCESS in 1h 10m 38s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/83327d9d30f44e579d9d5808a2703f5d : SUCCESS in 41m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/24888d96ed0b44f09fb78ba896ba59b4 : SUCCESS in 45m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c334fee4e4fa415987e8764e28cea9ef : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b4aebbb68dc458c8384ad05f6725fc5 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdf74923189e43c797e2e9790f1b3c8e : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91fe3224652b41b5b031af022eae6312 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c468959a857489bb10eb1f76c055248 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/811e74325cb54ccca95decabaef4c3d4 : SUCCESS in 3m 38s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/bf903b0127154e6f88c930fc1738e78c : SUCCESS in 38m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2fad95637b0ea98d2508c157003610f272d7e61b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-14 17:28:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f8f04369d39e1035de0f852de24940a9d97ded0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 18:35:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d122c2d370f3479d9be48caad03aabd2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/70202fe1e07b485eb04d328d8d3cf821 : SUCCESS in 43m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/89b107687a1d42858d9154b77dbf91c7 : SUCCESS in 41m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/eb7cbad942964e44a1834de944a7cdaa : SUCCESS in 1h 01m 33s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ecb2bb86756947878336652e993f3779 : SUCCESS in 29m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/00cf5021784a4144aab6adcddf6f4c5f : SUCCESS in 44m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed0462ebc5ce4b02b725a4429a88976f : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3484336802b74086a26910f781441084 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3834f5eecf544822a5ed524208c7fc05 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a26566b299bd4746b9bad55090f4ad06 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/455cf1f7fcf34fe3a036d0ed56ad9a0c : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94ee347d00394b41ba60805e508c6783 : SUCCESS in 3m 59s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/53aef031f8bd4cb89bc8b25064107f27 : SUCCESS in 21m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0251ff2979e68f6d345913f499efd6bc9ea40ce4","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-04-16 00:33:14.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b006ddc45cd0092f0b3583c54e3b7bc5acb86128","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-04-16 00:33:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"250cf347a7e8a0d8860f18994beaae8619b942f2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-18 18:39:33.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5ea25505fa61ea4b4a83a4e066aa57d6e45827f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-20 19:26:23.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c0dff5ac98a2cea32aef1e57a87cde21f501615","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 19:27:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec60e4891b2ce0549bf3bcb76704aa1d1dfb711d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:17:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce8b347813a148e0a4be8aec1f2ca677\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ba08a2b9cee94207a0146723dd2d63b7 : SUCCESS in 26m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b99e8af993484bb2b06593b455ad2b20 : SUCCESS in 39m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/186dc4b08c614df586a039ff5dea444a : SUCCESS in 48m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1470e25c8fea4645ac6cfb95ea25e6cb : SUCCESS in 35m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac8ad49defc740d9885b2a0c3d72fcd9 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82c4aeed5fc243d6a119bb609df47073 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/395914266a2c4f96bbb75370c11c40a4 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57c78306dc584c9aa5108ef6fd9e4eb0 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce146d3a842d43e8a8f829750ab4e2aa : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8f444e40bba26ef787847eeb2aecc61e34a0d628","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:17:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8a5e8943b1610fc803a1df82ddc1ecb0ef1ee731","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:19:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14ade465e6e643d98498d8b428692e25\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbb899c1937e4e408bc4e6edf04cd283 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d291a1c6e6b4b2486238d3b80af8a4e : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7eb080039a663159f94f91844fdb238e8ec14c6","revisions":{"a7eb080039a663159f94f91844fdb238e8ec14c6":{"kind":"REWORK","_number":1,"created":"2023-04-14 02:39:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/880382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/880382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/880382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/880382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/880382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/880382/1"}}},"commit":{"parents":[{"commit":"7a03c36e263fe67f867ff6e722665754ba7e3cfe","subject":"Removed unsupported test B309 from bandit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a03c36e263fe67f867ff6e722665754ba7e3cfe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-05 05:04:56.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-14 02:39:51.000000000","tz":0},"subject":"Use ids when removing sink managed records","message":"Use ids when removing sink managed records\n\nIf the record status changes during the removal process,\nthe current implementation will fail.\n\nCloses-Bug: 2015762\nChange-Id: Iebe609e5f365d03e99f2a4580671175b4642763c\n(cherry picked from commit 2cb42ac9f6e5273ad64e697919b62e029466078f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a7eb080039a663159f94f91844fdb238e8ec14c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a7eb080039a663159f94f91844fdb238e8ec14c6"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
