)]}'
{"id":"openstack%2Fdesignate~880514","triplet_id":"openstack%2Fdesignate~master~I8f0508aedcc028b1a6e74c790c5d757a0d6c98d7","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"5572":{"account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"last_update":"2023-05-23 17:25:17.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-05-23 16:20:57.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I8f0508aedcc028b1a6e74c790c5d757a0d6c98d7","subject":"Restructured SQL code","status":"MERGED","created":"2023-04-16 02:53:28.000000000","updated":"2023-05-23 17:26:21.000000000","submitted":"2023-05-23 17:25:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"880514","meta_rev_id":"2e8d56c165d343dbc72e529213cb21944804491f","_number":880514,"virtual_id_number":880514,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-23 17:25:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-05-23 14:32:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-22 20:36:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-05-23 16:20:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-16 04:54:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-18 15:02:08.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2023-05-22 20:36:54.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"d7461bb55ddef57c13d97088d7dce92272d77120","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-16 02:53:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf2584512142b4703b90d24621a906f7086782f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-16 03:33:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ad8ad192ba71edac565db0aff4670146aabd1b60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-16 04:54:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61cb79cc8f544529876e467ba6cdca28\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/69ed46987de1451999667327f0742de8 : SUCCESS in 48m 26s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/eee45604fcb740ecaf4835c1a7621dbd : SUCCESS in 39m 37s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d8ffbba6d7b422a840d509072533374 : SUCCESS in 36m 18s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/581554e11520483c9a61b1f834cd1c64 : SUCCESS in 50m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7723b71f22b94e7d8d421a69981d9d8c : SUCCESS in 51m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5566cf8f44f445f8a3e1a900b4a0dee8 : SUCCESS in 1h 13m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6b02b080d14d46fd95305b2033d2a87d : SUCCESS in 1h 15m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6779cb74c605410a87ff63ea4be77b60 : SUCCESS in 55m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4c97a7781747451aaeaddc4f314c506d : SUCCESS in 48m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb8399f015d34a6abe4197bd944d31ec : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf531aa7a7644478921343c7c33bfdb9 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9a8d3c5219146ec8305d8911761f86c : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/adfa35dfc91a46e3bad7e1bebac23589 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/407b9e7ef2234222b1d1b78c6875b246 : SUCCESS in 8m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4948d3a523ab4287a5ba8df960f451d1 : SUCCESS in 37m 16s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c8b0cb00c8db452d8008948e9f743d5b : SUCCESS in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ed5b6a0d17bf2383b5d58002c9dc40f7ba3906e","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-04-18 15:02:08.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c66fd2194300bb9c99fa04f8f90d45197086baf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-20 20:25:34.000000000","message":"Topic new_sink_rpc removed","accounts_in_message":[],"_revision_number":2},{"id":"46ab334b2e2fda098541ba7f2134636b621e81b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-01 02:49:16.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3cbe5a4277883826d2d4da21c5e3a6fb8d1ce059","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-01 03:37:35.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"8a9c43e57f35779fc12f3c8a65382ec51f366c18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-01 04:49:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b79ed183995a4398aa0e1c8cf702692a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a6e5c752d64b44f8a4c4f8c74dae78c5 : SUCCESS in 52m 52s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/bf4ce2a02b6d4395bbdfcbceb24fc094 : SUCCESS in 46m 09s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/dca9ccbd4f6543458ae9c5b4ec6f05ee : SUCCESS in 52m 02s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/740d55ac8ae94724b1587f32e7ed7469 : SUCCESS in 41m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/23a5e7475a7447d5b4195d3ac5f96974 : SUCCESS in 34m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/389ed3837935442b9a9cb107720f6d7c : SUCCESS in 1h 04m 34s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e25461729fe4470f9cb31d343f783538 : SUCCESS in 42m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dee81f3a4fff4db88fd5a7f57468613f : SUCCESS in 33m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ffd9fdea3b36481e959b86d08e47cfb3 : SUCCESS in 30m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4dc897dcf1e4a37a1fea24d7ba85664 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e809626922348c998e0346f424064a2 : FAILURE in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d5df01b442642d4bb3d27b3768d6397 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7c2e1f73ef54239832a5c0a808288af : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/031a28a1f0c54227bd37062f663ef241 : SUCCESS in 8m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ef1c336657af46f28925802a1c8a1851 : SUCCESS in 39m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/485b8e6626944b7e9407cb59f34e3dad : SUCCESS in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d8db3ecedbecc438dcac668cac1a6b9513074472","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-01 06:31:49.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"3ea3b7d119c952c3826c34d8fc9cd118a28e9cf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-01 07:41:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e32d973b0f74446b2eb62d008658058\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2c8a2855f59743599bff3a8930cc3dd2 : SUCCESS in 41m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6433efcde40840c0be717139247299a0 : SUCCESS in 46m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9238a5001a1740e1a6951c0c5c52873a : SUCCESS in 39m 47s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b6fad1e7cc194b2991e7a5fd4fb55d03 : SUCCESS in 32m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e7b43238b5ef43b6b5a75d4a2b3f5984 : SUCCESS in 44m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/16021fc1f5e945ce94e41b4ad161ab3b : SUCCESS in 46m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d0681fa26fe64e3ab6ca5388784ae0cf : SUCCESS in 1h 04m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3190abaa044242f583f74872d9d56685 : SUCCESS in 31m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cef93296d0794ebd9416ce6e45e44102 : SUCCESS in 31m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc4248e2473a4138a6d696a16439e861 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58a436019db04125a799508479b73d7a : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec49367f47524e978bc2727e6964de06 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/665f18867f974bfcbc141f4f9090bf7f : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19be1c19fa294bf6898fe8e46555844f : SUCCESS in 6m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8224ced036cd43c9a74233887cedbab3 : SUCCESS in 37m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/70b6cf99c21c46fd882dcd2c35d9bfe7 : SUCCESS in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"81a9c6d4281fb959b6b15b0c973c340c979d9cfb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-04 22:26:49.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"c12594ea6eadccbb0b11ab14791dc781831eb3b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-06 23:44:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0d22ff2fa969e7091536cd47afab0b45d2c7f5ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-07 00:09:00.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"cb544c2a2fe35cd0bc5b9d9bc8f747d99912f5a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-07 00:31:06.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8e4c18fe8064e702014d5a74ea1d97db88ea4acc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-07 00:46:33.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"0388d230e9fe3f0ec2e7c8971a39765826a0aca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-07 02:02:09.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/986cc13428f34207a1a4b16952b44e53\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c1868bcc20054c0aa0939901f641e5d1 : SUCCESS in 1m 47s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1fa3b89471a5488fae7d9892de725c87 : SUCCESS in 38m 51s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/743f82fd99c64c478579b2aeda85cbfa : SUCCESS in 43m 59s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b83aee567e1e4f5a81278b405c316567 : SUCCESS in 38m 25s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a043125a8f2b446d873b1e7532dacd11 : SUCCESS in 31m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a032ee43f0a14b34915418346773f84e : SUCCESS in 34m 12s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ff02a0d4c98046b487448c4e0a273542 : SUCCESS in 45m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d33e2385e3444cba847515ab42090ce5 : SUCCESS in 1h 10m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cd09308b0b9e4f0ea32575a93d09ab26 : SUCCESS in 49m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8661e654c2354c8087cc7e21b10cdf15 : SUCCESS in 50m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5b5c632ed0d4ee8a7414f0978efd048 : FAILURE in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eda8746b65bb4bcca29f908286aaeded : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d10328790e564cb0b6d21668edee34e2 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95e9c15d7db149e6a0e7b1c17077df5f : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4541e8f50eb84507ac1e275212003be8 : SUCCESS in 7m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1656513ba94b4e23b6df83a37f8b7f60 : SUCCESS in 33m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/355affce72d44a14b50957ce4c60842c : SUCCESS in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d80e75daf1d961c90f5300fef147ca798af9eccf","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-07 02:13:27.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e76f1ecdf79ac3e94b1d968d7f66b0ee164e107c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-07 03:31:58.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1de2f064ed2b4d5dbf1eb0d8af6eb0cb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fe31bb1f6e1a499b98054591fa35bca0 : SUCCESS in 1m 43s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b2930b912e094ec18538cde5b469af5d : SUCCESS in 54m 13s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/07f6128b2c2845cbb2ffd5989f30625b : SUCCESS in 56m 51s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a929cf87208475298aa36649f43e37b : SUCCESS in 1h 04m 53s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8275e3ffb27247b0b5b2e96602b2fbea : SUCCESS in 47m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9e9d52539d54449d8522cb352e4dc586 : SUCCESS in 43m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b6833e7450384fe089356144a34eea62 : SUCCESS in 1h 08m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6d45960062134b1480b38c87107ca134 : SUCCESS in 1h 03m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3dfa97c505254795aa8dc8f138f778b6 : SUCCESS in 48m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8461e6d47e6c41a9b1a10863f0138608 : SUCCESS in 57m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76343394397f48649d7fb6acb8689f77 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fffd7f821f474cb28db3405af7c24969 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a3f6c940f0f478e9ed7f21f6c420587 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21ab602066c342adbc719de3503500d2 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b33f5ef0d544677becd9f05c1fa6acc : SUCCESS in 7m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f6d425ad807341a9a9ddaf4579d7d387 : SUCCESS in 39m 32s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5885c3f320624399a3b3ddb33bd2d1c0 : SUCCESS in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"316435c571f5b5abb8358bec5690c2ddb185f3ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-07 03:56:50.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"2d27f400035812c9739bc45ef15423c850eb85c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-07 04:59:35.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a15e388ae1c84c5f957470d12f8d076c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f662feabb1424740966b51d3114292fa : SUCCESS in 1m 47s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8e8068f8a88747189aa5fb916ba7e857 : SUCCESS in 36m 50s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/bd9e6837de314a16a71a6ef81b127e81 : SUCCESS in 43m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/68b932d7203546a1aa60b631527e18e8 : SUCCESS in 35m 26s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/aede2fd09c0b4b018e54edaa52ed66eb : SUCCESS in 35m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7e20185806ae47cc9ea1bd31f604376f : SUCCESS in 57m 25s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8e13be1f072c493da3dae17758772c92 : SUCCESS in 51m 16s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/efb6be0bdb2e4612bb48233816a650fc : SUCCESS in 46m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c9d8112a14b04f519f03c36b8dae1bcb : SUCCESS in 37m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f2e5855ef8c84a6e9c6f74fb1eb13f4e : SUCCESS in 52m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1471e29c2b854c83872d91883a391291 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88ee924cb1254292b6a17b452773c15c : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5d1c68d8aae442bb88bf2d8ec1192b0 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e60c9e662d5f4cb792156e780fd13ce1 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c44456087352434bab2a89264f68d122 : SUCCESS in 6m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5ecb31c5f07a44cfa35252b607026407 : SUCCESS in 31m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/88056a7f5baf486d89d921fbd181107f : SUCCESS in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"dd4452e0179abfce0f49db5b8ef73a45434b47b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-08 00:44:18.000000000","message":"Uploaded patch set 11: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"d8e9d1ebd72812ce51e2fd6271110958f0fda2d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 02:00:54.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8a4f3cbd25848d5a0299ecf9efe1354\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/19805fcacf414f4ea0cb10ddf405369d : SUCCESS in 1m 39s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8603b5689fcf4614b0d0aafb6f804317 : SUCCESS in 50m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8ce210ca489d49c6b8ff221222b49906 : SUCCESS in 1h 10m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/106e4dc8f36f473da5c156beeed30f4f : SUCCESS in 48m 39s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/fb0c221a06a14baebd867d278dfbe289 : SUCCESS in 39m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0b663f9e68ef4038be1f9a7705f52d37 : SUCCESS in 38m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9851f64526f0470dae30c42db710ea5a : SUCCESS in 51m 29s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f67ddf144dc3437c8e773d8ef33bee89 : SUCCESS in 45m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/30c8dbcd8c424e17b2ea1e3a2d91ab25 : SUCCESS in 50m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/44a737c27473423d8934d4c4d89c5153 : SUCCESS in 47m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27afa83e539c46d498bb14c23b9d6040 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a43c67ad02224ba6a97ad713c30c0963 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d8dff5e99904eb3b574f9f3543d3d95 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6904f711cf8e4ed2826654b97a98127d : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/004701ae0a57494885e2c79f54baf3f4 : SUCCESS in 7m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1da44bc74c9e48608c1222fde7cad5a9 : SUCCESS in 37m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c21593c9218c437aa467379f00a1525c : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"516b7e92af65cf14e0b7c0cc4c317ef11b52ea2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-09 16:59:05.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"062c0e9041d1d196e0c46b1e99b7ef581b0a5e16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 18:03:12.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa838e8f265944c7a4df54edb93f3e29\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1cee361fc3b04496bcdb415de2302e3c : SUCCESS in 2m 31s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a4e3d6a3a5e048569f718f59d19061c3 : SUCCESS in 40m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3bc071fda43d446c84acadd1f2a94111 : SUCCESS in 46m 54s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b01adb4f0ffd4a01a49139924592f6e3 : SUCCESS in 38m 48s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d2bf5f1fe82547e6a0d106df2e4805d7 : SUCCESS in 40m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5f4e699c1bbd4b73a14c8bc85879f65c : SUCCESS in 42m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6e1d8d187e1a4cd9ba2b07fd3f5cbb65 : SUCCESS in 53m 23s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/92da2d72034b43368220e17b81e1eac9 : SUCCESS in 48m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/78dc927cf1eb4370b1dcb75b009a8b40 : SUCCESS in 38m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d64055062f7449ddb2dffbe3e34b1043 : SUCCESS in 54m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/850e7207e0894078ace827b6ad3d2460 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63e1dadf6c3f49b7823726d3351f0b37 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce4647d221a04a23841f885d781b5696 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b5d28c380564a99b85f85cf367a0352 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da7233bcbde34fff954fe5b933cc549c : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a170f8f541841c3b43eb70d0bb3f1c5 : SUCCESS in 8m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/79f3604f782d40c1903b13ec4f09db18 : SUCCESS in 36m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/337c8e7c48e1486a816a4a3c28b6c82f : SUCCESS in 5m 46s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4f0a32460d8f9982f3659a23f84c7883939cc566","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-12 04:28:59.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"7a7a34fbe2726e973312148d0d17300870f0cb8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 05:47:40.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c23651ee1514f81b205efafb005ba08\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/125cdfc66d5a413dab400ce71dfd9eb8 : SUCCESS in 1m 50s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d7a4dbf273ff43748564422c57480b44 : SUCCESS in 38m 20s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/08023e540fe14bf2b89210cb6f071a11 : SUCCESS in 57m 20s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbe8510b944443568bf2cd8f5314fed5 : SUCCESS in 47m 16s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a2fe12f6ac0049348d76f8884f5094ca : SUCCESS in 51m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5a9325b8b8f3482697b5c956236d8412 : SUCCESS in 45m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/65da29a29d9e4df2a767378036607ad8 : SUCCESS in 1h 02m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/701793215d1f4895b5c19ace0806eea0 : SUCCESS in 1h 11m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fe9aea55f9f04b9e950975671acdd744 : SUCCESS in 37m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4d3a8209972a47bfb2d58962262d5849 : SUCCESS in 53m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15971ac296dc42b683ddf84ec30f2832 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d8d9602b56946dcbe665b20c1fb76df : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89112a85e2eb49f69e4faa6b30a909d9 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb7f7db8a40a427b8f16cf91581bb2c7 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24814ce78609482d90948d043cb39da3 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/982a2532e3794360b94e8d203d211d3d : SUCCESS in 7m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c2e86110af964bfc81b28fe55acd8527 : SUCCESS in 32m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/02eb1b3eb8254a2a9ed59837fac9a283 : SUCCESS in 4m 03s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"1d5361949ed36583a50b502a29dd02ac7c6ce816","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-22 20:36:54.000000000","message":"Patch Set 13: Code-Review+2\n\n(6 comments)","accounts_in_message":[],"_revision_number":13},{"id":"f97b482fc8210bd8df6f0e985e06a9a5af16d6e0","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-05-23 14:32:16.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"e24772e1330515fb05be63fbf9182c55578e28cb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-23 16:20:57.000000000","message":"Patch Set 13: Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"27a100cb1d6cbc85390577159fb63d6013ffd17d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 16:21:32.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"8f9f562f706fd9d5f61371e74f5bafee3b6d3c34","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-23 16:27:05.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"226e71b1bd568c6c93822d8640a06357929f81fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 17:25:17.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d864e3b7b8994a35ab30fae21f65c06b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c931b9a9daca438695ce7a79850a2f91 : SUCCESS in 1m 47s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6052462862054edca45a10ba9aae162c : SUCCESS in 34m 46s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b00aef6ed080405298e90c02b91f43f1 : SUCCESS in 39m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5116268785aa49e9bb80685de88a8ee2 : SUCCESS in 39m 55s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9a47ab07d38a43adb64939283c568195 : SUCCESS in 51m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2589f99173734224a3571b57c34c8e3c : SUCCESS in 38m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4d4b022efd354ffea1458e5bda6f9263 : SUCCESS in 57m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3894681c81042308b6f48444e8fb5ba : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23e9af71e82b43ee904aa0e6c4fe834b : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98eeeb098d9048038ce97afab5474e42 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da79b36ffeb44c2abaa4f363fb2aaa00 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/725ab91c997d40b2a06499a2ac4a5950 : SUCCESS in 7m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/658228a8e7cb430397624ce860d2b728 : SUCCESS in 45m 00s","accounts_in_message":[],"_revision_number":13},{"id":"1015a18b06e9e19e0f43847f4e594aad54c207de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 17:25:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"2e8d56c165d343dbc72e529213cb21944804491f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 17:26:21.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc4dbe6e35ee42f6bab4276f4ca43702\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/826b4df3157b4916af45051215789f6b : SUCCESS in 45s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"0c30fc491b0012da13a562c9107f4451790607dd","revisions":{"298f024ff575be7544c48244d3e89b1423b4204f":{"kind":"REWORK","_number":1,"created":"2023-04-16 02:53:28.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/880514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/880514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/880514/1"}}},"commit":{"parents":[{"commit":"f36906d57b14ff87f8128bedfa0d4b71b6ebe61d","subject":"Refactor managed record logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f36906d57b14ff87f8128bedfa0d4b71b6ebe61d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-16 02:53:11.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-16 02:53:11.000000000","tz":-420},"subject":"Clean up SQL code","message":"Clean up SQL code\n\nChange-Id: I8f0508aedcc028b1a6e74c790c5d757a0d6c98d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/298f024ff575be7544c48244d3e89b1423b4204f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/298f024ff575be7544c48244d3e89b1423b4204f"}]},"branch":"refs/heads/master"},"3c505fafd11f95a79ecfc75eb7e98b5d4437a5d4":{"kind":"REWORK","_number":2,"created":"2023-04-16 03:33:32.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/880514/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/880514/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/880514/2"}}},"commit":{"parents":[{"commit":"f36906d57b14ff87f8128bedfa0d4b71b6ebe61d","subject":"Refactor managed record logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f36906d57b14ff87f8128bedfa0d4b71b6ebe61d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-16 02:53:11.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-16 03:33:25.000000000","tz":-420},"subject":"Clean up SQL code","message":"Clean up SQL code\n\nChange-Id: I8f0508aedcc028b1a6e74c790c5d757a0d6c98d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3c505fafd11f95a79ecfc75eb7e98b5d4437a5d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3c505fafd11f95a79ecfc75eb7e98b5d4437a5d4"}]},"branch":"refs/heads/master"},"acdf812229ab60b37d4728498f1f3eed26d2826d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-05-01 02:49:16.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/880514/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/880514/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/880514/3"}}},"commit":{"parents":[{"commit":"a073f7a14891238167eeeb599fdeecaa8e9d6705","subject":"Refactor managed record logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a073f7a14891238167eeeb599fdeecaa8e9d6705"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-16 02:53:11.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-01 02:49:16.000000000","tz":0},"subject":"Clean up SQL code","message":"Clean up SQL code\n\nChange-Id: I8f0508aedcc028b1a6e74c790c5d757a0d6c98d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/acdf812229ab60b37d4728498f1f3eed26d2826d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/acdf812229ab60b37d4728498f1f3eed26d2826d"}]},"branch":"refs/heads/master","description":"Rebase"},"ffbbfb37cb2a3fc2deebcf3a12256c576de3d4f8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-05-01 03:37:35.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/880514/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/880514/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/880514/4"}}},"commit":{"parents":[{"commit":"3d8144e1fc7522444c3ee85c0ae04fd2afd276b1","subject":"Refactor managed record logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3d8144e1fc7522444c3ee85c0ae04fd2afd276b1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-16 02:53:11.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-01 03:37:35.000000000","tz":0},"subject":"Clean up SQL code","message":"Clean up SQL code\n\nChange-Id: I8f0508aedcc028b1a6e74c790c5d757a0d6c98d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffbbfb37cb2a3fc2deebcf3a12256c576de3d4f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffbbfb37cb2a3fc2deebcf3a12256c576de3d4f8"}]},"branch":"refs/heads/master","description":"Rebase"},"62668576033d54ef201b269b032f2c986701b4ba":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-05-01 06:31:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/880514/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/880514/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/880514/5"}}},"commit":{"parents":[{"commit":"ee911857d96d80f85dd9b7edca4233dbbf4f790d","subject":"Refactor managed record logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee911857d96d80f85dd9b7edca4233dbbf4f790d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-16 02:53:11.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-01 06:31:49.000000000","tz":0},"subject":"Clean up SQL code","message":"Clean up SQL code\n\nChange-Id: I8f0508aedcc028b1a6e74c790c5d757a0d6c98d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/62668576033d54ef201b269b032f2c986701b4ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/62668576033d54ef201b269b032f2c986701b4ba"}]},"branch":"refs/heads/master","description":"Rebase"},"d9c3dbf1a543aae040d58e81fb86cc3657458adf":{"kind":"REWORK","_number":6,"created":"2023-05-06 23:44:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/880514/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/880514/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/880514/6"}}},"commit":{"parents":[{"commit":"ee911857d96d80f85dd9b7edca4233dbbf4f790d","subject":"Refactor managed record logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee911857d96d80f85dd9b7edca4233dbbf4f790d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-16 02:53:11.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-06 23:43:12.000000000","tz":-420},"subject":"Restructured SQL code","message":"Restructured SQL code\n\nWe have a lot of abstraction that isn\u0027t really needed. This patch\nremoves some of that abstraction and reorganizes the code accordingly.\n\nChange-Id: I8f0508aedcc028b1a6e74c790c5d757a0d6c98d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d9c3dbf1a543aae040d58e81fb86cc3657458adf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d9c3dbf1a543aae040d58e81fb86cc3657458adf"}]},"branch":"refs/heads/master"},"d36166aac6271426de2383f847256a90bc9846a8":{"kind":"REWORK","_number":7,"created":"2023-05-07 00:09:00.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/880514/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/880514/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/880514/7"}}},"commit":{"parents":[{"commit":"ee911857d96d80f85dd9b7edca4233dbbf4f790d","subject":"Refactor managed record logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee911857d96d80f85dd9b7edca4233dbbf4f790d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-16 02:53:11.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-07 00:08:48.000000000","tz":-420},"subject":"Restructured SQL code","message":"Restructured SQL code\n\nWe have a lot of abstraction that isn\u0027t really needed. This patch\nremoves some of that abstraction and reorganizes the code accordingly.\n\nChange-Id: I8f0508aedcc028b1a6e74c790c5d757a0d6c98d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d36166aac6271426de2383f847256a90bc9846a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d36166aac6271426de2383f847256a90bc9846a8"}]},"branch":"refs/heads/master"},"8497f3aa44b3ce6150d307b643327ea5eb7337b2":{"kind":"REWORK","_number":8,"created":"2023-05-07 00:31:06.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/880514/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/880514/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/880514/8"}}},"commit":{"parents":[{"commit":"ee911857d96d80f85dd9b7edca4233dbbf4f790d","subject":"Refactor managed record logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee911857d96d80f85dd9b7edca4233dbbf4f790d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-16 02:53:11.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-07 00:31:00.000000000","tz":-420},"subject":"Restructured SQL code","message":"Restructured SQL code\n\nWe have a lot of abstraction that isn\u0027t really needed. This patch\nremoves some of that abstraction and reorganizes the code accordingly.\n\nChange-Id: I8f0508aedcc028b1a6e74c790c5d757a0d6c98d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8497f3aa44b3ce6150d307b643327ea5eb7337b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8497f3aa44b3ce6150d307b643327ea5eb7337b2"}]},"branch":"refs/heads/master"},"7b08880d4e6c49c3a28f0a5ccd75fa37c70ea5ef":{"kind":"REWORK","_number":9,"created":"2023-05-07 00:46:33.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/880514/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/880514/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/880514/9"}}},"commit":{"parents":[{"commit":"ee911857d96d80f85dd9b7edca4233dbbf4f790d","subject":"Refactor managed record logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee911857d96d80f85dd9b7edca4233dbbf4f790d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-16 02:53:11.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-07 00:46:24.000000000","tz":-420},"subject":"Restructured SQL code","message":"Restructured SQL code\n\nWe have a lot of abstraction that isn\u0027t really needed. This patch\nremoves some of that abstraction and reorganizes the code accordingly.\n\nChange-Id: I8f0508aedcc028b1a6e74c790c5d757a0d6c98d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7b08880d4e6c49c3a28f0a5ccd75fa37c70ea5ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7b08880d4e6c49c3a28f0a5ccd75fa37c70ea5ef"}]},"branch":"refs/heads/master"},"4bf021db460dd9fb514f9d91555622cb03a424ba":{"kind":"REWORK","_number":10,"created":"2023-05-07 03:56:50.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/880514/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/880514/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/880514/10"}}},"commit":{"parents":[{"commit":"ee911857d96d80f85dd9b7edca4233dbbf4f790d","subject":"Refactor managed record logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee911857d96d80f85dd9b7edca4233dbbf4f790d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-16 02:53:11.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-07 03:56:43.000000000","tz":-420},"subject":"Restructured SQL code","message":"Restructured SQL code\n\nWe have a lot of abstraction that isn\u0027t really needed. This patch\nremoves some of that abstraction and reorganizes the code accordingly.\n\nChange-Id: I8f0508aedcc028b1a6e74c790c5d757a0d6c98d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4bf021db460dd9fb514f9d91555622cb03a424ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4bf021db460dd9fb514f9d91555622cb03a424ba"}]},"branch":"refs/heads/master"},"56ab6785103082cdc6604cc1d1ecbcd9a17d2d99":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2023-05-08 00:44:18.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/880514/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/880514/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/880514/11"}}},"commit":{"parents":[{"commit":"ee911857d96d80f85dd9b7edca4233dbbf4f790d","subject":"Refactor managed record logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee911857d96d80f85dd9b7edca4233dbbf4f790d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-16 02:53:11.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-08 00:43:30.000000000","tz":-420},"subject":"Restructured SQL code","message":"Restructured SQL code\n\nWe have a lot of abstraction that isn\u0027t really needed. This patch\nremoves some of that abstraction and reorganizes the code accordingly.\n\n- Merged storage and sqlalchemy folders.\n- Deprecated [central].storage_driver.\n- Minor cleanup to SQL code.\n- Cleaned up comments and unit tests.\n\nChange-Id: I8f0508aedcc028b1a6e74c790c5d757a0d6c98d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/56ab6785103082cdc6604cc1d1ecbcd9a17d2d99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/56ab6785103082cdc6604cc1d1ecbcd9a17d2d99"}]},"branch":"refs/heads/master"},"15df619b16a198890cd63bb973a9dbeabd1cf55d":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2023-05-09 16:59:05.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/880514/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/880514/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/880514/12"}}},"commit":{"parents":[{"commit":"f70053f2aaa8b8d46690c94530bb2fb5f908347c","subject":"Merge \"Use monotonic time to protect from system time jumps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f70053f2aaa8b8d46690c94530bb2fb5f908347c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-16 02:53:11.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-09 16:59:05.000000000","tz":0},"subject":"Restructured SQL code","message":"Restructured SQL code\n\nWe have a lot of abstraction that isn\u0027t really needed. This patch\nremoves some of that abstraction and reorganizes the code accordingly.\n\n- Merged storage and sqlalchemy folders.\n- Deprecated [central].storage_driver.\n- Minor cleanup to SQL code.\n- Cleaned up comments and unit tests.\n\nChange-Id: I8f0508aedcc028b1a6e74c790c5d757a0d6c98d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15df619b16a198890cd63bb973a9dbeabd1cf55d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15df619b16a198890cd63bb973a9dbeabd1cf55d"}]},"branch":"refs/heads/master","description":"Rebase"},"0c30fc491b0012da13a562c9107f4451790607dd":{"kind":"REWORK","_number":13,"created":"2023-05-12 04:28:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/880514/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/880514/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/880514/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/880514/13"}}},"commit":{"parents":[{"commit":"590064e8d986660e294d469e39e95241857778b5","subject":"Merge \"Remove unused test code for coordination\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/590064e8d986660e294d469e39e95241857778b5"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-04-16 02:53:11.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-12 04:28:36.000000000","tz":-420},"subject":"Restructured SQL code","message":"Restructured SQL code\n\nWe have a lot of abstraction that isn\u0027t really needed. This patch\nremoves some of that abstraction and reorganizes the code accordingly.\n\n- Merged storage and sqlalchemy folders.\n- Deprecated [central].storage_driver.\n- Minor cleanup to SQL code.\n- Cleaned up comments and unit tests.\n\nChange-Id: I8f0508aedcc028b1a6e74c790c5d757a0d6c98d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0c30fc491b0012da13a562c9107f4451790607dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0c30fc491b0012da13a562c9107f4451790607dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
