)]}'
{"id":"openstack%2Fdesignate~880691","triplet_id":"openstack%2Fdesignate~master~Ibeb7154a4c300f59dfa25737736f59609b938bcc","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-05-10 01:13:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibeb7154a4c300f59dfa25737736f59609b938bcc","subject":"Fix shared zones missing project ID error","status":"MERGED","created":"2023-04-17 23:21:30.000000000","updated":"2023-05-10 01:14:54.000000000","submitted":"2023-05-10 01:13:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880691","meta_rev_id":"7e323e93b9269ccf023284a6c1e5c482da857bc9","_number":880691,"virtual_id_number":880691,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-10 01:13:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-05-09 22:55:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-19 04:32:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-09 23:53:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-18 00:19:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-19 04:32:37.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2023-04-19 13:31:35.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"CC"},{"updated":"2023-05-09 22:55:50.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"809b9baeb3612b8f4173c8522bc1a8dc96bc8e13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-17 23:21:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47520590a4120c0330d1f43117f9cbf96cc9442c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 00:19:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76bfe4d3add349b7bb7831f5fdaf5981\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/876b1e5667ce40a4b272ef214f82f7e4 : SUCCESS in 39m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7e654f641ea4450fadfb8554cbb54809 : SUCCESS in 48m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/51cc13c04bef44c9acda376a42eea3db : SUCCESS in 41m 02s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/fd561c1add6d4fdaa9c81d5ffc8c3c1d : SUCCESS in 40m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/acb80b3605f244b8ab5fcac5367992b2 : SUCCESS in 43m 52s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5d95290a155d48248e109f3c4abc33fb : SUCCESS in 57m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d448572b0f47484c8bb3bbefe669dc5f : SUCCESS in 56m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fc8039282dcb49da9e1aea281ff3fe00 : SUCCESS in 36m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bb37cc4907d44ac18b26666af433177d : SUCCESS in 48m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c4a6b7626d8427da2314da6695c0dc4 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db66264295744a2fb19f59e7b6dcd335 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6ae111c7bef4337b4cb9fc2559e5e71 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d810b17ca7d24db5bd7691133a00a842 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d3886a8a0c34f8a940bf159500dc3ab : SUCCESS in 8m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/34ebc2d4955248f29230572344101265 : SUCCESS in 38m 07s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/39caf7a0677f4073a3916dc53f4d7e6a : SUCCESS in 4m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ec9923934abc5df345234cb259c904f3a14cbbd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-19 04:32:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cb28a2ff509a229736899a90635b0d2521369b7e","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-04-19 13:31:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79867a64afe26fab14d2d7dec0a97cdd62771212","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-05-09 22:55:50.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5812e69f3415d7173bfb46331d410842b9c74223","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-09 23:53:09.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4181dfa85a3f3a878c8b408123bbc00e9573c4bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 23:53:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f553eecd2c3ffd73f90aac3a2640b38f522fc301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 01:13:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0044112302a84beda63ff24a6fba9ba4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/043468ce64f3436e97299216c2c67ea7 : SUCCESS in 54m 54s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3b5ea4197fd44980b4110e6e55674452 : SUCCESS in 52m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9417ec33a6ae453c99d391f4763e1d76 : SUCCESS in 48m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f0d27eefe2cb451f90b6eec6ca2bf332 : SUCCESS in 1h 15m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4614054291b141bc8ae856a8f95c847a : SUCCESS in 40m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a1192bb9f07f4e5a8a6ecef38b05630d : SUCCESS in 48m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb5a5c3317fe407a86d90d7cf9f61f1b : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ee32162c6454cea970d883da1824a70 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6420a4165834521b3465e0e3a575c39 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7f5d3a0b24b44c0916c97b10f613654 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d54326dbdd594523b44205c4261b26d4 : SUCCESS in 7m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1d63356430f1438891922a2a36a2f00b : SUCCESS in 39m 09s","accounts_in_message":[],"_revision_number":1},{"id":"18dc33f71ef5eef5842022cf445e85545eea743b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 01:13:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7e323e93b9269ccf023284a6c1e5c482da857bc9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 01:14:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82b204c5e05c47628c8de966f484cdf8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8521022f52484bacae7ca56951784785 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f37baf808295699847ca9b897cd485bf1f5ea1a","revisions":{"0f37baf808295699847ca9b897cd485bf1f5ea1a":{"kind":"REWORK","_number":1,"created":"2023-04-17 23:21:30.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/91/880691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/880691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/880691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/880691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/880691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/880691/1"}}},"commit":{"parents":[{"commit":"ae3bb0ddf359cd0a0d7b5b4705e331d104c6fce0","subject":"Merge \"Use ids when removing sink managed records\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ae3bb0ddf359cd0a0d7b5b4705e331d104c6fce0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-17 23:21:05.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-17 23:21:28.000000000","tz":0},"subject":"Fix shared zones missing project ID error","message":"Fix shared zones missing project ID error\n\nPreviously if you tried to create a zone share with a scoped token that\ndoes not have a project ID, you would receive a 500 error. This patch\ncorrects that to return a proper error describing that a project ID is\nrequired when creating zone shares.\n\nChange-Id: Ibeb7154a4c300f59dfa25737736f59609b938bcc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0f37baf808295699847ca9b897cd485bf1f5ea1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0f37baf808295699847ca9b897cd485bf1f5ea1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
