)]}'
{"id":"openstack%2Fdesignate~880863","triplet_id":"openstack%2Fdesignate~stable%2Fussuri~I3fe5895eeee727f53d1c9202d4d0930963435aaa","project":"openstack/designate","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-04-20 15:28:48.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I3fe5895eeee727f53d1c9202d4d0930963435aaa","subject":"Update Ussuri grenade job to use zuulv3","status":"MERGED","created":"2023-04-19 15:30:07.000000000","updated":"2023-04-20 16:54:56.000000000","submitted":"2023-04-20 16:53:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880863","meta_rev_id":"886d7b8d222574ee8195311670cca4bee25b24b2","_number":880863,"virtual_id_number":880863,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 16:53:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-04-20 15:09:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-20 15:28:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-04-20 02:14:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-20 15:28:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-19 18:20:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-20 02:14:36.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2023-04-20 15:09:34.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"751e44b6355d219d7d01e8766c26d0676050c23b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-19 15:30:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40079831a11f7a436b606447644536fb5b83b2a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-19 15:48:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"78ee905d597b5f9b75329a416aeb376ee1c5e71c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-19 16:36:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d51901710598a43a38bb6603b56aa95b74ef6a84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-19 16:58:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"005c1d295920b3e9515d9a4dcbe5ff02e048eb9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-19 17:11:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"07942e6607c72f3d95dd42415d63224941c02c6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 18:20:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c67d272d28d444c99ce291d92643fd3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/49308e580b0c402a88107886c9aba98b : SUCCESS in 36m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4d1f4735c8834512abac549be66b586d : SUCCESS in 35m 50s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b59306fa0f7e4dd3a2990e214eb40768 : POST_FAILURE in 5m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7c5abcdd0ab6455d92c504ada7eb811b : SUCCESS in 43m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/68c8c78409484d21bd625784731b6663 : SUCCESS in 42m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29341cac5192414db495ac460c0e4262 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ee9a3b6530a4f9ea628aea6acdaf7d0 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8488b3dd34944a11aae16386cc159741 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1000baf3d95343c394f8595a10367d65 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10d08338690b40cfaaa64d6b379ebc27 : SUCCESS in 8m 26s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/08adf29503a344d89fec5a1fcab2d330 : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":5},{"id":"1a215f5da5986e205b1abb932a48267b1246b5bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-19 19:29:17.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0f0a5e16d425d1bbb2cba708d8e94b10abeb0f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 20:24:31.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c9900ca50f34321969c30bca4fab8e8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/781035962e8b4128b77704b761dfc5b6 : SUCCESS in 32m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d80b724d6b794107bf033183fc899b56 : SUCCESS in 33m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a966d323e8bc45859fb6c0afd151ba9b : FAILURE in 49m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9bd86de98045448cba0a34d7d44e1a19 : SUCCESS in 43m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e5620239c85543c080d3ccc085bc63ad : SUCCESS in 41m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b75a238843774a14a49efc3d29570585 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5cf1ad1590b44abbc36d8212f1e9146 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d1bfbc44bd94827af6d3ad0361357f1 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a1ebb66559d4510bb6442a864640af5 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13528d1053714b8c9c88268552ec723a : SUCCESS in 8m 37s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/6150c59f616e4206ab0a56c54e219611 : SUCCESS in 21m 24s","accounts_in_message":[],"_revision_number":6},{"id":"eaa64d7e914dba1811b14d1638948b67b20291d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-19 22:54:51.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"73cfb32bf76bcd0b342e68ea6da196eb1978b186","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 23:49:43.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fccdd2d05bcc4c80b3a1ede7fac8f78e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d5abe18672d04999a556e10e8e0d890b : SUCCESS in 42m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/367429141c1a4cabb968781bfc6e7c5b : SUCCESS in 41m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0e2efc72320c4e58b0e90fafc8198a0d : FAILURE in 54m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7cf26c27d6534c2b8c0b6f85e2d9ddf2 : SUCCESS in 41m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6b801224a3194e8a90c9484deef786ca : SUCCESS in 41m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48c1f231b980448bb21b8b9c5b045fab : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26e1b2d025a44ae9a4c7be8d1a2dbed7 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a71a2b7b09df4b3cbcf23e8434c1d75e : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afa0516cb7fb479cbb3369e4acccdfa6 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/850173b7cb0d47759b41bcafeb299bce : SUCCESS in 8m 27s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/24702b1f31c142bda44e8ae1257179ea : SUCCESS in 37m 43s","accounts_in_message":[],"_revision_number":7},{"id":"30e3d4620450ffaae2239822031371c07a80712d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-20 00:03:25.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"a175ab75b84225f5bd8e6ffaac6747161ec77f40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 01:11:32.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e7a19c5563e4494a044f3d77acfcb4d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b9c8d71efaff4c9b8d70df4b7214724d : SUCCESS in 43m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8b6118318bed4a788fe868d345b9b46b : SUCCESS in 39m 16s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c261436166a24d9991924714cb00e218 : SUCCESS in 1h 07m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3b8793cad96e477d873238fb4eefff76 : SUCCESS in 42m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b480970f7c034aac9dcc12a2beede7f2 : SUCCESS in 46m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/197588e6933a4c72a6cc543cbe796cfb : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0342e5ca2067411aa7a3a7df803f249a : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cae47be5a7584078929d9255d322bc9d : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4efe07ba9cf8473aa656481d8e9285c5 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/169f49d1144f45c6b0889b1c04e0570a : SUCCESS in 8m 01s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/9f4831350b994762b1e24ab457765a51 : SUCCESS in 26m 44s","accounts_in_message":[],"_revision_number":8},{"id":"7165df2353cd4ef7677b0a3210c0df9ee96a9540","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-20 02:14:36.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"4c54c16e90bbb365b0fb67e89023ff3f35e16dc1","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-04-20 15:09:34.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"22316935d8c47df5a7b3af0254885ac77c49d849","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-20 15:28:48.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ea42ae8d745a48a4ab2dc9ccc571403ae8cf0a81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 15:29:27.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"1594014494b73057af5ecd709ff02e1c314af1c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 16:53:05.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b37de917e75e4e83a9aba94f966b4fe4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/749619c64e164ed089e4925971754ab9 : SUCCESS in 44m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8eaf0cc3126d47e2a24b59377b95821f : SUCCESS in 44m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1728f4edc0b64b46863251e8ffedf9c7 : SUCCESS in 1h 16m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4c7e775407644225b7bc38cdf41c5c77 : SUCCESS in 46m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3a1ee2f2fe3b4b70b3ff78befe0f1c3d : SUCCESS in 44m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d9989d737104ed7aa216861067bfeca : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c3519c2c2674c6f8a880d39ae84184b : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2878262b8e94b3883793e2778eb74a8 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9897a4c36aa840a19ad0bd2d147992b4 : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":8},{"id":"8693db54d20c4c8ef3b44e66a41e3e804d61eb8e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 16:53:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"886d7b8d222574ee8195311670cca4bee25b24b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 16:54:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d5c54ef4aba47a7bd4c83d4c2b78da7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1eea2c1ac7243388052997807345c9b : SUCCESS in 36s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"64b56187185a62780f79017a271d9962453bc442","revisions":{"301378c0dcfb1d8587a1f90bdc3df47a77bfe468":{"kind":"REWORK","_number":1,"created":"2023-04-19 15:30:07.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/63/880863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/880863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/880863/1"}}},"commit":{"parents":[{"commit":"bdbbe6a396682f316304027845c818f94e6f4cd3","subject":"CI: Move queue configuration into project-template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bdbbe6a396682f316304027845c818f94e6f4cd3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-19 15:29:42.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-19 15:30:05.000000000","tz":0},"subject":"DNM Testing ussuri grenade fix","message":"DNM Testing ussuri grenade fix\n\nChange-Id: I3fe5895eeee727f53d1c9202d4d0930963435aaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/301378c0dcfb1d8587a1f90bdc3df47a77bfe468"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/301378c0dcfb1d8587a1f90bdc3df47a77bfe468"}]},"branch":"refs/heads/stable/ussuri"},"15c0aa88805a83fe913b6c9a9b8cce56bd6e76d4":{"kind":"REWORK","_number":2,"created":"2023-04-19 15:48:36.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/63/880863/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/880863/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/880863/2"}}},"commit":{"parents":[{"commit":"bdbbe6a396682f316304027845c818f94e6f4cd3","subject":"CI: Move queue configuration into project-template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bdbbe6a396682f316304027845c818f94e6f4cd3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-19 15:29:42.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-19 15:48:29.000000000","tz":0},"subject":"DNM Testing ussuri grenade fix","message":"DNM Testing ussuri grenade fix\n\nChange-Id: I3fe5895eeee727f53d1c9202d4d0930963435aaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15c0aa88805a83fe913b6c9a9b8cce56bd6e76d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15c0aa88805a83fe913b6c9a9b8cce56bd6e76d4"}]},"branch":"refs/heads/stable/ussuri"},"8dae951640735707e2d176a6ff2f586d00c18fc6":{"kind":"REWORK","_number":3,"created":"2023-04-19 16:36:52.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/63/880863/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/880863/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/880863/3"}}},"commit":{"parents":[{"commit":"bdbbe6a396682f316304027845c818f94e6f4cd3","subject":"CI: Move queue configuration into project-template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bdbbe6a396682f316304027845c818f94e6f4cd3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-19 15:29:42.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-19 16:36:45.000000000","tz":0},"subject":"DNM Testing ussuri grenade fix","message":"DNM Testing ussuri grenade fix\n\nChange-Id: I3fe5895eeee727f53d1c9202d4d0930963435aaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8dae951640735707e2d176a6ff2f586d00c18fc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8dae951640735707e2d176a6ff2f586d00c18fc6"}]},"branch":"refs/heads/stable/ussuri"},"2d9c564bff62d493d40d92884efe724dccdfe712":{"kind":"REWORK","_number":4,"created":"2023-04-19 16:58:58.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/63/880863/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/880863/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/880863/4"}}},"commit":{"parents":[{"commit":"bdbbe6a396682f316304027845c818f94e6f4cd3","subject":"CI: Move queue configuration into project-template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bdbbe6a396682f316304027845c818f94e6f4cd3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-19 15:29:42.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-19 16:58:51.000000000","tz":0},"subject":"DNM Testing ussuri grenade fix","message":"DNM Testing ussuri grenade fix\n\nChange-Id: I3fe5895eeee727f53d1c9202d4d0930963435aaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2d9c564bff62d493d40d92884efe724dccdfe712"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2d9c564bff62d493d40d92884efe724dccdfe712"}]},"branch":"refs/heads/stable/ussuri"},"8e072b01c82c4c4be3579bd7b6bfc35fe16128e4":{"kind":"REWORK","_number":5,"created":"2023-04-19 17:11:24.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/63/880863/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/880863/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/880863/5"}}},"commit":{"parents":[{"commit":"bdbbe6a396682f316304027845c818f94e6f4cd3","subject":"CI: Move queue configuration into project-template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bdbbe6a396682f316304027845c818f94e6f4cd3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-19 15:29:42.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-19 17:11:18.000000000","tz":0},"subject":"DNM Testing ussuri grenade fix","message":"DNM Testing ussuri grenade fix\n\nChange-Id: I3fe5895eeee727f53d1c9202d4d0930963435aaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8e072b01c82c4c4be3579bd7b6bfc35fe16128e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8e072b01c82c4c4be3579bd7b6bfc35fe16128e4"}]},"branch":"refs/heads/stable/ussuri"},"73874e52fd168384c13853b7213e2085cc23f5ec":{"kind":"REWORK","_number":6,"created":"2023-04-19 19:29:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/63/880863/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/880863/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/880863/6"}}},"commit":{"parents":[{"commit":"bdbbe6a396682f316304027845c818f94e6f4cd3","subject":"CI: Move queue configuration into project-template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bdbbe6a396682f316304027845c818f94e6f4cd3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-19 15:29:42.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-19 19:29:11.000000000","tz":0},"subject":"DNM Testing ussuri grenade fix","message":"DNM Testing ussuri grenade fix\n\nChange-Id: I3fe5895eeee727f53d1c9202d4d0930963435aaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/73874e52fd168384c13853b7213e2085cc23f5ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/73874e52fd168384c13853b7213e2085cc23f5ec"}]},"branch":"refs/heads/stable/ussuri"},"e056637f35cb2bac8868b66a6295df6d1023dc0e":{"kind":"REWORK","_number":7,"created":"2023-04-19 22:54:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/63/880863/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/880863/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/880863/7"}}},"commit":{"parents":[{"commit":"bdbbe6a396682f316304027845c818f94e6f4cd3","subject":"CI: Move queue configuration into project-template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bdbbe6a396682f316304027845c818f94e6f4cd3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-19 15:29:42.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-19 22:54:21.000000000","tz":0},"subject":"Update Ussuri grenade job to use zuulv3","message":"Update Ussuri grenade job to use zuulv3\n\nChange-Id: I3fe5895eeee727f53d1c9202d4d0930963435aaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e056637f35cb2bac8868b66a6295df6d1023dc0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e056637f35cb2bac8868b66a6295df6d1023dc0e"}]},"branch":"refs/heads/stable/ussuri"},"64b56187185a62780f79017a271d9962453bc442":{"kind":"REWORK","_number":8,"created":"2023-04-20 00:03:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/63/880863/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/880863/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/880863/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/880863/8"}}},"commit":{"parents":[{"commit":"bdbbe6a396682f316304027845c818f94e6f4cd3","subject":"CI: Move queue configuration into project-template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bdbbe6a396682f316304027845c818f94e6f4cd3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-19 15:29:42.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-20 00:03:18.000000000","tz":0},"subject":"Update Ussuri grenade job to use zuulv3","message":"Update Ussuri grenade job to use zuulv3\n\nChange-Id: I3fe5895eeee727f53d1c9202d4d0930963435aaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64b56187185a62780f79017a271d9962453bc442"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64b56187185a62780f79017a271d9962453bc442"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
