)]}'
{"id":"openstack%2Fdesignate~882260","triplet_id":"openstack%2Fdesignate~master~I382250784be7dd9c20245cb90b4a19a9e7e313a5","project":"openstack/designate","branch":"master","topic":"test_cleanup","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-05-11 01:13:52.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I382250784be7dd9c20245cb90b4a19a9e7e313a5","subject":"Remove unused test code for coordination","status":"MERGED","created":"2023-05-07 03:59:31.000000000","updated":"2023-05-11 02:34:24.000000000","submitted":"2023-05-11 02:31:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882260-test_cleanup","meta_rev_id":"9669fdf0ba301bc341dc4eb6d8ab3cdd97803c9e","_number":882260,"virtual_id_number":882260,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-11 02:31:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-05-09 22:54:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2023-05-11 00:54:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-11 01:13:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-07 05:09:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-09 22:54:18.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2023-05-11 00:54:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"9d5c3700cdec8d5c7376358b2754336639730f25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-07 03:59:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3828641c54c348447de863fe74799361560e4c12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-07 05:09:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7b019a5f4d742cda566d5260000e21f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/97a0c0a690294f308e08542380f1ab78 : SUCCESS in 34m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c6045e1a1fc74a61a97c7198bb18cec7 : SUCCESS in 47m 48s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5aad170789442099695cb3d76992723 : FAILURE in 43m 53s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c3ea95eaa98d47ebbd6b1d0587441069 : SUCCESS in 49m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/391089b7a5f8437888775a307752e780 : SUCCESS in 54m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5ad0141ec8fd4e2daeadaa58a910b93b : SUCCESS in 1h 04m 27s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/96eeaea6ddaa4362bfe01cc330e38835 : SUCCESS in 47m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6f591021cf1b43fc92007f7f3545efdd : SUCCESS in 31m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/edf76f91308d490ab29de8049b6c0382 : SUCCESS in 36m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daf0b49b5a154405a525c737a6fb4b3d : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0081a9529eea4337a447777100b905d7 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47bae9986c6c47c2970f2d3662b249a3 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b766279fb7e479d882dac28ff3e3e49 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e6bd20a8e841f6addc2138118666bc : SUCCESS in 6m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e67484a6407249a8b39ce0d9e2373648 : SUCCESS in 34m 07s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e672a6d0487241dfabb6da22606d457c : SUCCESS in 3m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8d70fd37a2bcf948eae419b43b8ecf5d845683d","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-05-09 22:54:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a8a6697781dd93e45d1c5773254d1c5f213fd2a7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-11 00:54:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"82e52b06a6d97c66fcfaee795a437f90e30f39c1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-11 01:13:52.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a6c1a7dd4689d3178e100bc2663b75d84cd5a216","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 01:14:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ce57e0dc240f11eb83aa9f7dbaf36213a5e3bbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 02:31:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c79335d1b3f4291b126767ee12b2b79\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/34157c22f74b42e99f9951a7523c3a96 : SUCCESS in 40m 09s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ceb4d4ebad824d07bb531b4e418ff2eb : SUCCESS in 34m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6ba19bbb501947149ed3e203490014ef : SUCCESS in 31m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/36a8cfe8da724992aff74d33b04a5f09 : SUCCESS in 59m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e101113ec2b84afe88fcb5fcb36a1c6d : SUCCESS in 58m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/47b4c832be274d92897290336f0e697c : SUCCESS in 54m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dc1fca5d25e47afb3cd21dec79f40e6 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2951e5dd8dbc44b6836b4b87fcbf4d21 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0efc00f1ace43e583c3bfd5ab7b854d : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/848ba972c87f4dda8324590445c74093 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97b73e512f5c446cb2cfe7e03cd6becd : SUCCESS in 7m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2a0af786baed4033a2402c6ace95abc5 : SUCCESS in 35m 59s","accounts_in_message":[],"_revision_number":1},{"id":"3de49feb5e0686ef8f479809e0121862bb6b2e79","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 02:31:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9669fdf0ba301bc341dc4eb6d8ab3cdd97803c9e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 02:34:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fea622b5f62745ffa4982ba3af4153bb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/001900c5b8654ccbb4157d04b9ee40df : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"accecda40409e4a66c8d622cdfcd37d2c9d1fa4c","revisions":{"accecda40409e4a66c8d622cdfcd37d2c9d1fa4c":{"kind":"REWORK","_number":1,"created":"2023-05-07 03:59:31.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/60/882260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/882260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/882260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/882260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/882260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/882260/1"}}},"commit":{"parents":[{"commit":"118294ac1443507ad8ac9a18d99102f02850c2b4","subject":"Fix invalid fallback value for BaseAddressHandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/118294ac1443507ad8ac9a18d99102f02850c2b4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-07 03:59:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-07 03:59:09.000000000","tz":-420},"subject":"Remove unused test code for coordination","message":"Remove unused test code for coordination\n\nChange-Id: I382250784be7dd9c20245cb90b4a19a9e7e313a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/accecda40409e4a66c8d622cdfcd37d2c9d1fa4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/accecda40409e4a66c8d622cdfcd37d2c9d1fa4c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
