)]}'
{"id":"openstack%2Fdesignate~882262","triplet_id":"openstack%2Fdesignate~master~I665f9d3a5bb8ceeace0bdedac2b64c9dc6205668","project":"openstack/designate","branch":"master","topic":"simplify","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-05-11 01:13:48.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I665f9d3a5bb8ceeace0bdedac2b64c9dc6205668","subject":"Clean up test imports","status":"MERGED","created":"2023-05-08 00:34:22.000000000","updated":"2023-05-11 02:33:13.000000000","submitted":"2023-05-11 02:31:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882262-simplify","meta_rev_id":"7545002f2af04a95e2c07fb9ed8f9dcbe3758141","_number":882262,"virtual_id_number":882262,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-11 02:31:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-05-11 00:55:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-11 01:13:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-08 01:31:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-11 00:55:58.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"66686265a7ea03df76663822b7ae8fd38962e376","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-08 00:34:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bfac48abfec82215dc42242826cf2d8be108efb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 01:31:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5561a53d2e754ee792740c9f11475c8d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7ef586b7008b4a7ab566b3645c973596 : SUCCESS in 33m 08s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2f79705147b541908fd50122ef44ffc0 : SUCCESS in 43m 08s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/66d6c91c96f843c7b44b1cea3653ffa1 : SUCCESS in 36m 55s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/dca0632c6fc54d33b3da3c62bb1b348a : SUCCESS in 31m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2d40029d26f14ec39e73cb0fdc3b29f4 : SUCCESS in 33m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0c497207ded742638be22996674d7f17 : SUCCESS in 47m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c04ca4a7fba9418c85cad42a7b08c408 : SUCCESS in 47m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1b55ac70e5334eebaf8ad1e871b45be0 : SUCCESS in 33m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/354a532c74e94f00a1726806cbf4ac8a : SUCCESS in 51m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55054a62956f4577be0961d18079931f : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc464aa60ee845e882dcd9bbd9e9c78b : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0beb4a3db2bd43af8fa02b46c0b50bd4 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f79378020dc446e1b52e398e41e994f8 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9be98c4a13845148daefab9d5b7eb00 : SUCCESS in 7m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/acbfcbb507504d16997a334ec1678e5c : SUCCESS in 40m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/684c1d5c293b4959a63dc4b64d5b2841 : SUCCESS in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ccd9ef4f872ab311571f33c8e38204257a30932","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-11 00:55:58.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7fe9d51048f6cb1dbdd6b5092fa0409e96a390a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-11 01:13:48.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ab8a8dd499cc3db433a0130675a416871cc9b969","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 01:14:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8322df5885ad163ce8b1d816abab9e1b89e39c03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 02:31:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2139c3aea5b4d94989a997eef14c707\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/12b0eedf2dd34bcfa148bfc2e285c426 : SUCCESS in 51m 46s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b45e240f6a9c4f4fa8c53730bfe7981e : SUCCESS in 51m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/07ce7f339be048d2a191d46df7f2fae3 : SUCCESS in 49m 18s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/44e3f8e7f7154d0197bb67a7c494b946 : SUCCESS in 1h 10m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6b6fcb112fb946e0a525ed628e9c8b65 : SUCCESS in 49m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2a91db5917fe4195a31fcf66c31ef9b2 : SUCCESS in 37m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69569beb61d9470e9f17c9c786eccc73 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ef5f7f777804f269d3cf1760d8919e8 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/339285fb99ab4d62b5aed67096254f57 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6361dd5ee0f411da0066e97c291579b : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f104487136294600a21d4657b8b93695 : SUCCESS in 6m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d64169b406884de898260a2d32a2c178 : SUCCESS in 42m 28s","accounts_in_message":[],"_revision_number":1},{"id":"c0cef1cfaa94ab677bc02734d807526a0e24af86","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 02:31:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7545002f2af04a95e2c07fb9ed8f9dcbe3758141","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 02:33:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b67487f55be4c69aa83551c877b599d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7b8e28406b043e0b4bbf6db59edfde7 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7aed84431559b201a559696915373d3dad14733","revisions":{"e7aed84431559b201a559696915373d3dad14733":{"kind":"REWORK","_number":1,"created":"2023-05-08 00:34:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/62/882262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/62/882262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/882262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/882262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/882262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/62/882262/1"}}},"commit":{"parents":[{"commit":"118294ac1443507ad8ac9a18d99102f02850c2b4","subject":"Fix invalid fallback value for BaseAddressHandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/118294ac1443507ad8ac9a18d99102f02850c2b4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-08 00:21:06.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-08 00:34:02.000000000","tz":-420},"subject":"Clean up test imports","message":"Clean up test imports\n\nChange-Id: I665f9d3a5bb8ceeace0bdedac2b64c9dc6205668\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7aed84431559b201a559696915373d3dad14733"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7aed84431559b201a559696915373d3dad14733"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
