)]}'
{"id":"openstack%2Fdesignate~883822","triplet_id":"openstack%2Fdesignate~master~I91b2c2e4bcb83aea74dc531b0dc4b95842e6b4a0","project":"openstack/designate","branch":"master","topic":"wsgi_tests","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-05-24 19:39:43.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I91b2c2e4bcb83aea74dc531b0dc4b95842e6b4a0","subject":"Add basic api wsgi tests","status":"MERGED","created":"2023-05-24 05:52:51.000000000","updated":"2023-05-24 21:03:55.000000000","submitted":"2023-05-24 21:02:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883822-wsgi_tests","meta_rev_id":"e9d2452b9f66e16b3a0d979f5966dd8be712546d","_number":883822,"virtual_id_number":883822,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-24 21:02:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-05-24 17:48:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-24 19:39:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-24 07:09:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-24 17:48:47.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"7275687ee554883a85e97a2a268bf0319b257d05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-24 05:52:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5559323a506b37e700a932194078aa6d62fe3de0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 07:09:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c132942d27164efc9ee79a9719337047\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/69acbd94802a4a8fbecc80a2a1a3a6af : SUCCESS in 37m 59s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8b77c21f18374d5c992c1051c7b065ca : FAILURE in 10m 32s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f72560c37aee4863915f5f92da12ceaa : SUCCESS in 39m 53s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e013ebca8a5347f78392552ec4112dae : SUCCESS in 55m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/da664d6415f14c8e9fc37a77ca3aca31 : SUCCESS in 48m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9dead040999542e5a0922d9af84e80f0 : SUCCESS in 54m 30s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0457a0c689ac47339add66b10b3af942 : SUCCESS in 1h 09m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d82d6d48594144b8a911bd5db19f88dc : SUCCESS in 35m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ce89d667a4d94e5e8fe430036cca3bad : SUCCESS in 54m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b44b14f6d36143b58860405d3727099d : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1330429631df46fda1b16fbd1c71b79a : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a169cefd48244349c3d89c42ebba12b : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/822fa255f19d4d14ac9d1b473607464d : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/205ec8caba564054879e0aa3d4d4d38a : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/897b5ecf07eb4c4a82978ecb8cd20a13 : SUCCESS in 6m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3454383b266f4261aafe22fe7c12feb4 : SUCCESS in 42m 53s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/eea8b48087764e1aa7e72345b959f31b : SUCCESS in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ccfcdbc6fd8715908cc1c49a875f58d5d9d76f5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-24 17:48:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"961c71fecced5249a1c628fc9f58f029d6d1ac2a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-24 19:39:43.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e05f2f5ac808ca47ab5d1756267a9da71cddb7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 19:40:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5371af433240cf8f71a68fb61f64323c23b91216","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 21:02:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06ac38b021004cf78052833936e0effa\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/023f83244e9f43a285374a667441bb29 : SUCCESS in 44m 09s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/53b1d8b60c5e482e96fbad965fbad43a : SUCCESS in 52m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/652c56f4a8574430b07b63cb73d327fb : SUCCESS in 51m 16s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6917c34d639149efb415698c297905ee : SUCCESS in 1h 11m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5e1d20c2a76943d1a88c97be77e9dc63 : SUCCESS in 55m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2a98e8e08be6481fb7fe9a6ab3409609 : SUCCESS in 45m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da1dbaf8b396427c9ea48768d29c432d : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbe9b6e752b54e49b8975c2df3cd0948 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/257eeaad3abd4816b758cde04ae5c2a5 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe9134a7c1ab4f1fb360f723aa6c2a7f : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c810d754cdc348f68470c7a71d062855 : SUCCESS in 7m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8433afb74a014d5bb0f8cfa400479448 : SUCCESS in 32m 13s","accounts_in_message":[],"_revision_number":1},{"id":"defca94af2571b467f25b033a891eed4ee571940","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 21:02:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e9d2452b9f66e16b3a0d979f5966dd8be712546d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 21:03:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e67b21c91902481db9e9e1aee10115a2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1659b09ca4744eeb60c7af40e3312de : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46e72847472fb4d1c8daa03ce95b0fe78a54f61f","revisions":{"46e72847472fb4d1c8daa03ce95b0fe78a54f61f":{"kind":"REWORK","_number":1,"created":"2023-05-24 05:52:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/22/883822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/22/883822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/883822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/883822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/883822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/22/883822/1"}}},"commit":{"parents":[{"commit":"757aa5e5d7c896fd0eb862331a59c2b645ae65a5","subject":"Merge \"Restructured SQL code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/757aa5e5d7c896fd0eb862331a59c2b645ae65a5"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-24 05:51:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-24 05:52:30.000000000","tz":-420},"subject":"Add basic api wsgi tests","message":"Add basic api wsgi tests\n\nChange-Id: I91b2c2e4bcb83aea74dc531b0dc4b95842e6b4a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/46e72847472fb4d1c8daa03ce95b0fe78a54f61f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/46e72847472fb4d1c8daa03ce95b0fe78a54f61f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
