)]}'
{"id":"openstack%2Fdesignate~884345","triplet_id":"openstack%2Fdesignate~master~Idc21c18bb7689b857f3892634e030930a68f4765","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-05-30 19:13:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idc21c18bb7689b857f3892634e030930a68f4765","subject":"Add manage cmd tests","status":"MERGED","created":"2023-05-27 00:47:26.000000000","updated":"2023-05-30 19:14:15.000000000","submitted":"2023-05-30 19:13:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884345","meta_rev_id":"b5892c71b9cf92b93c450375ff1119182d7c20d8","_number":884345,"virtual_id_number":884345,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-30 19:13:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-05-30 17:50:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-05-30 17:50:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-27 03:36:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-30 17:50:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"429a53be08322aaf5ee9be5fa82b1ccb3620fc28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-27 00:47:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ecb57bf1a4203a59d94b26a8c171125167b70d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-27 00:49:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"08a841ec6833a91633c7b5d2f7334aa4c8cbed53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-27 01:01:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"154da27b6d50dee9a132e670c5c056cdd8936c52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-27 01:09:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"14a30c09579c9debe62a3ff6ec1868d0f30dd60a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-27 01:37:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"89b1c5d9f393810f121ae9a34cc7c5cf2f6df6c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-27 02:02:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"96043e2a2fe28788ac5ab2ab5867bb48836d4043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 03:36:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc5d89d4d9ee4941989bd4060c3f0dd5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/89eb63439692452281a5ffc867e0af19 : SUCCESS in 48m 35s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/09259a4f897c4002b26f67305631a5f9 : FAILURE in 11m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0951f130f9bb4c658194cc744595b2e4 : SUCCESS in 1h 04m 44s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8e4a3a5c5e8548ac8967e9d45a4ccff5 : SUCCESS in 51m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6253f4a0b74d46b69397ce7a9c821b35 : SUCCESS in 50m 45s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c4ad6e68e0364469b0ae72ba10f921fe : SUCCESS in 1h 00m 29s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b4f31c1a15db4d539d9c90fe85f79352 : SUCCESS in 1h 16m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e703dacc966d451b80599e48688650cf : SUCCESS in 49m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ecee2f4f14df4c2484037e4fa5a44385 : SUCCESS in 41m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c5079f272f944e6a0185eb28ce5dc8a : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/760e9f3c80fa4c9397ad5e2ebd6abf49 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bd6d358b30944f792c131aedb00ca5b : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb900906fdaf48c3aa5d2c307d189cf6 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0ac5c5737014cf1b2dc1b4edfe42621 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28910c0e713c441a9d214b84f42197e0 : SUCCESS in 10m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b747a7a1ac6e41698419a6fd67cb9833 : SUCCESS in 34m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5ac95389dbbc48dd8e5e8b77fd86a59e : SUCCESS in 4m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"42a5fe0350c81d5dba4b2162777e954dcdd49a28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-28 04:42:03.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"f4923f26bc9a4b4978029e0b58344e80fdc7f1fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-28 05:36:29.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c791ebc69a9a4d7a8afe293ca139903c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/001730156181499aa526eea7ffaabf15 : SUCCESS in 32m 51s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/01f432648b45448da2084b191b5a9a48 : FAILURE in 24m 13s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/75a328c249be42f0a1c24c1ba1fa3e4d : SUCCESS in 34m 42s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/32f0a44dc5f7489f946d6fc4088b930a : SUCCESS in 49m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b2664d2d248946d0b190d5514e1f74c1 : SUCCESS in 30m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/554857de9b154f11a3582a8c0fd19c9f : SUCCESS in 45m 30s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c77ac19e6bc74fa8a3a8c577ac9ebcb9 : SUCCESS in 44m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2d64d9091516434f9728d2bb9b19a4b0 : SUCCESS in 38m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8d1f7de5d4244c9f850a6cf872e7f376 : SUCCESS in 33m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4aa0f08037164525bbf0b5fb2dfd7cb5 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/067d9137d8c5495fb99c672bdaa6f335 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30846bd9eb8246b4a8c61c8975619f61 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d25b7920f0ab4bde9c500ca625a7ee13 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d085a2926ab442f9632b05dc6d3e9ed : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fac9c86783c4f36839fc922e8925ccd : SUCCESS in 8m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1f1a7537cd2a4265a817304a7b56052a : SUCCESS in 35m 12s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d59e249bde1445949184a98e993b24b8 : SUCCESS in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0cfc3ecce9f2192f018efffa06df98d073e01543","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-29 05:17:59.000000000","message":"Topic test_manage removed","accounts_in_message":[],"_revision_number":7},{"id":"494f6f9166e823563cee21540a0f73ed7a5d88bc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-30 17:50:49.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"801a134574026a48162596c2f77b7d5b890e2068","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 17:51:28.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"3eb9576f17599a39272c96bf3bab88503a066d29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 19:13:09.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3578c3805a7049f09e96523f45463eb4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1e7c3f7679d8421d929931403cb8ac85 : SUCCESS in 50m 52s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/94bb6c36e21543b7a617f7b34514b137 : SUCCESS in 55m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1cfe3152f2494906bc1ede63d683ac83 : SUCCESS in 53m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e8d8a2dd76614904beb5a86494508a0c : SUCCESS in 1h 13m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6c2cdf35cfd4467391b3dd4c944dd201 : SUCCESS in 58m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/023aaee29aa24758b56216a13d4264b3 : SUCCESS in 56m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d279c7e092c43ef99686a04e8b54a16 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77bcee3bb43649fc8b2314ffb84c33c4 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d0a8110616741308ee7f14484897520 : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25e33d130f414bc48612ea301d6abce3 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecca17e0e067410b86eed26a3fd34914 : SUCCESS in 8m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/65a2cc4194014c95ae576a38a06dc317 : SUCCESS in 31m 16s","accounts_in_message":[],"_revision_number":7},{"id":"a434ea40d5a3c2f1bca6caa24b98c7a8bc445fd7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 19:13:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"b5892c71b9cf92b93c450375ff1119182d7c20d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 19:14:15.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ee3a29962174ae58a410cddb40cc376\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/964588fe628547dd915fcf99360e8618 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"661f0dd5175fb0334a7f0f6f1ff4482164d09b91","revisions":{"c38f982c0ccfa5b40c04c721bfe5aa5cebea3e94":{"kind":"REWORK","_number":1,"created":"2023-05-27 00:47:26.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/45/884345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/45/884345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/45/884345/1"}}},"commit":{"parents":[{"commit":"b9bd8e9bade38c4d6af72b482ce85b3e7a8cba6e","subject":"Add cmd tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9bd8e9bade38c4d6af72b482ce85b3e7a8cba6e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 00:47:19.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 00:47:19.000000000","tz":-420},"subject":"Add manage cmd tests","message":"Add manage cmd tests\n\nChange-Id: Idc21c18bb7689b857f3892634e030930a68f4765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c38f982c0ccfa5b40c04c721bfe5aa5cebea3e94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c38f982c0ccfa5b40c04c721bfe5aa5cebea3e94"}]},"branch":"refs/heads/master"},"4380da565ad4e88cb36e2199077f60731aa99f71":{"kind":"REWORK","_number":2,"created":"2023-05-27 00:49:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/45/884345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/45/884345/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/45/884345/2"}}},"commit":{"parents":[{"commit":"b9bd8e9bade38c4d6af72b482ce85b3e7a8cba6e","subject":"Add cmd tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9bd8e9bade38c4d6af72b482ce85b3e7a8cba6e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 00:47:19.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 00:49:51.000000000","tz":-420},"subject":"Add manage cmd tests","message":"Add manage cmd tests\n\nChange-Id: Idc21c18bb7689b857f3892634e030930a68f4765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4380da565ad4e88cb36e2199077f60731aa99f71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4380da565ad4e88cb36e2199077f60731aa99f71"}]},"branch":"refs/heads/master"},"89372c39191a4f285b1419cd9def2471034417de":{"kind":"REWORK","_number":3,"created":"2023-05-27 01:01:10.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/45/884345/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/45/884345/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/45/884345/3"}}},"commit":{"parents":[{"commit":"b9bd8e9bade38c4d6af72b482ce85b3e7a8cba6e","subject":"Add cmd tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9bd8e9bade38c4d6af72b482ce85b3e7a8cba6e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 00:47:19.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 01:00:59.000000000","tz":-420},"subject":"Add manage cmd tests","message":"Add manage cmd tests\n\nChange-Id: Idc21c18bb7689b857f3892634e030930a68f4765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/89372c39191a4f285b1419cd9def2471034417de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/89372c39191a4f285b1419cd9def2471034417de"}]},"branch":"refs/heads/master"},"ed3e341ebd52c62fd95b5f277fddd4b35603ed5d":{"kind":"REWORK","_number":4,"created":"2023-05-27 01:09:57.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/45/884345/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/45/884345/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/45/884345/4"}}},"commit":{"parents":[{"commit":"b9bd8e9bade38c4d6af72b482ce85b3e7a8cba6e","subject":"Add cmd tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9bd8e9bade38c4d6af72b482ce85b3e7a8cba6e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 00:47:19.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 01:09:52.000000000","tz":-420},"subject":"Add manage cmd tests","message":"Add manage cmd tests\n\nChange-Id: Idc21c18bb7689b857f3892634e030930a68f4765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ed3e341ebd52c62fd95b5f277fddd4b35603ed5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ed3e341ebd52c62fd95b5f277fddd4b35603ed5d"}]},"branch":"refs/heads/master"},"f4088cf0d69d7800c53ff0d7d4742ae64c47b604":{"kind":"REWORK","_number":5,"created":"2023-05-27 01:37:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/45/884345/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/45/884345/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/45/884345/5"}}},"commit":{"parents":[{"commit":"b9bd8e9bade38c4d6af72b482ce85b3e7a8cba6e","subject":"Add cmd tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9bd8e9bade38c4d6af72b482ce85b3e7a8cba6e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 00:47:19.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 01:37:42.000000000","tz":-420},"subject":"Add manage cmd tests","message":"Add manage cmd tests\n\nChange-Id: Idc21c18bb7689b857f3892634e030930a68f4765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f4088cf0d69d7800c53ff0d7d4742ae64c47b604"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f4088cf0d69d7800c53ff0d7d4742ae64c47b604"}]},"branch":"refs/heads/master"},"fa715224bd2dcf327305f83dd9bd93a63c53781d":{"kind":"REWORK","_number":6,"created":"2023-05-27 02:02:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/45/884345/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/45/884345/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/45/884345/6"}}},"commit":{"parents":[{"commit":"b9bd8e9bade38c4d6af72b482ce85b3e7a8cba6e","subject":"Add cmd tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9bd8e9bade38c4d6af72b482ce85b3e7a8cba6e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 00:47:19.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 02:02:53.000000000","tz":-420},"subject":"Add manage cmd tests","message":"Add manage cmd tests\n\nChange-Id: Idc21c18bb7689b857f3892634e030930a68f4765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa715224bd2dcf327305f83dd9bd93a63c53781d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa715224bd2dcf327305f83dd9bd93a63c53781d"}]},"branch":"refs/heads/master"},"661f0dd5175fb0334a7f0f6f1ff4482164d09b91":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-05-28 04:42:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/45/884345/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/45/884345/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/884345/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/45/884345/7"}}},"commit":{"parents":[{"commit":"48826ffa96b0ed358ea814657d31fb5ba28966ee","subject":"Fix create zone export RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/48826ffa96b0ed358ea814657d31fb5ba28966ee"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 00:47:19.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-28 04:42:03.000000000","tz":0},"subject":"Add manage cmd tests","message":"Add manage cmd tests\n\nChange-Id: Idc21c18bb7689b857f3892634e030930a68f4765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/661f0dd5175fb0334a7f0f6f1ff4482164d09b91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/661f0dd5175fb0334a7f0f6f1ff4482164d09b91"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
