)]}'
{"id":"openstack%2Fdesignate~884346","triplet_id":"openstack%2Fdesignate~master~I1b00300118340faecc81c33f3a77e46e151e0a1c","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-05-30 22:05:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1b00300118340faecc81c33f3a77e46e151e0a1c","subject":"Improve central unit test coverage","status":"MERGED","created":"2023-05-27 02:55:29.000000000","updated":"2023-05-30 22:06:56.000000000","submitted":"2023-05-30 22:05:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884346","meta_rev_id":"25fe28627bfae5bd11d48996e8622a457b331c68","_number":884346,"virtual_id_number":884346,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-30 22:05:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-05-30 20:38:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-05-30 20:38:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-27 04:31:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-30 20:38:27.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"a3e41d93df26de3206479a138c7ef843b9f2295b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-27 02:55:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46d5b0cb32987e8caec7e5f4bff164f3e59b7da8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 04:31:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/617a226998574ae8822d9f9963187bb4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fbd2768fb6644b6a87b09d18b09ce9d1 : SUCCESS in 42m 57s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c0cd95e5711949a1bab0813f03c0eb8e : FAILURE in 29m 23s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e776f3f566674398a3f64eacdf749a40 : SUCCESS in 57m 35s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0edbead982804d5ab46637ff730b53cd : SUCCESS in 45m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3f1459d1e44243c29d9b54821f8479ff : SUCCESS in 37m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/26d0c16e29bc445c9db31442f1d407f2 : SUCCESS in 1h 26m 54s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/67b467c15557472eb50ad9c6e67384bf : SUCCESS in 1h 10m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5a1bb3292b8e4792a262713051f34905 : SUCCESS in 49m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3f52ea04fa01465796568e434700d5c5 : SUCCESS in 55m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f532b596e804f6e930065265f7c3739 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6a0efc562994b69a89724c006d51519 : FAILURE in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e674cab9e034466bbaa470c571f034bf : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/654f767f7fe048ee9b2cda62ec18d442 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14a06d813ae142deb7753610c96ae50d : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ca872b95f0647bbb6075eb3ea63d87c : SUCCESS in 6m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c666b6e61c2942b28b773913cae795a5 : SUCCESS in 32m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6e79cea14ba64470949feb104803d5ad : SUCCESS in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb21d6eaf47451bfc27754b7675453fdcaf248ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-27 05:08:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d9cccca82757e078e2179d71d032e52d614811e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 06:25:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a468d967d65e4b74af7ac7a6e38e46ed\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9119dbceb9e84f32ae6ba8d18a55a222 : SUCCESS in 37m 31s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/09ab01a709fa40d8ae0db8b6bbc72bb2 : FAILURE in 9m 16s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/62b809ce2fe846a48f2274fac18c26ad : SUCCESS in 34m 34s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ac1ad77e4e9841bab74170c033582159 : SUCCESS in 38m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1363bbbbac204de098f0fb4386b9a59a : SUCCESS in 34m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c73bb62fde334d5cb93a07f871a10fe8 : SUCCESS in 1h 11m 10s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cc51a865c1614ce79297bc0a9ea5ea9d : SUCCESS in 44m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4f9040b57b0b432b8bb910ad32f2ca0b : SUCCESS in 32m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f210bfee6b384072b138187f6896c95c : SUCCESS in 35m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/245be8342bd249cfb0b31998d9040fa0 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3307209e2f664e07b29a0254f6925cb3 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/150c606d07fa4ce2a6a0b9d5107aef88 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4bb6a157670476590a5fe1a3786cee0 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6a4f529c9464d0fa7ceb83e02d03d49 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ade48f0bfb4845e69a293f8927b584e3 : SUCCESS in 7m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ffcf782e6d6d455fa1428d121e1485ec : SUCCESS in 34m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/64e6aecda1c54be3abee65050d196bdc : SUCCESS in 5m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7232e1dae0e7dee9434f24f68e6722e91dc7b9f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-27 19:41:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"86631531707447adaef0c50f326fdde41f72442b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-27 19:47:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d140377a70894f9271c76d411c3c97f3392d5b2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 21:07:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d007f5afa634703921ff8c6746e2a2f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ecfde117da1a4c64b9fc9df74717a59a : SUCCESS in 53m 47s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ed77cf8b49864ea9a1dcfdd43fe43ba6 : FAILURE in 9m 27s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f4f2b2e098c43f69198bffb23ad3dfa : SUCCESS in 36m 44s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6982b691441f46b3a94b1565603bc068 : SUCCESS in 36m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9e72f06aa49441aeb82ac88dd8103af2 : SUCCESS in 33m 38s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/32392ac5c0bc42bfb3f6391abf57e4d7 : SUCCESS in 53m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/862b29866f5e4f5494b5c383917c1838 : SUCCESS in 1h 12m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c3afcc694cde4157916ad866686e0f1b : SUCCESS in 35m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/acb68ce794154792aa4b1453a66a2986 : SUCCESS in 37m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d388117b77984e2b86a344fa44c8e6a5 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70f128d42d2c4c279b3d0851506acd18 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1390f5b9a4ed4ffe9f8b85fe0292dfba : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d25d683911214e1db00cff875ff477ad : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1bdf543c90494c1984cbcc98370b29e5 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95380782e5ce4356ad10d1646b544e45 : SUCCESS in 7m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8e2efcbb61a94a069eb175cdfa521105 : SUCCESS in 31m 58s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/43ce6a77f0034e5eb524adf4b64da6c2 : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"841acdce56f8cb3636d5d8b0bf483c3aa4bf4f8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-27 21:10:24.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"13c1eae815e34f8f49c39bc84f8f4a91d0d9b2a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-27 22:07:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d271179503fe9c55492d585553443d0558a2b091","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-27 22:08:18.000000000","message":"Topic more removed","accounts_in_message":[],"_revision_number":6},{"id":"a4b89f28a9b027bb955fddbbfbf2d84e671b5576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 23:04:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8203ca7c617148b9b7472d0677cf21fb\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0b1d7166c2044b42a3874195af2fcff8 : SUCCESS in 51m 26s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5f1d02e305c341edb87b388b6ed9e246 : FAILURE in 9m 31s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5ed523f489043508e08dfc9a8b952f6 : SUCCESS in 32m 48s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5f0263958e29424e8e2e5d81427e166e : SUCCESS in 32m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/196214b290654a99b1d4c4e5d788b8a5 : SUCCESS in 34m 16s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/44d8ad415c23458996a8176c849fe634 : SUCCESS in 45m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/48f8d528301a440a8712563f33adc709 : SUCCESS in 50m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/885d3173c4be44aeae0311cf0c1729a7 : SUCCESS in 28m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1fc5745b66194ad380310e26dee0a544 : SUCCESS in 34m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdf09d259f8f40b494a3f3d9d7b8d8c6 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/786a6b537c3a420f82666d383a2ed1e9 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec28b52d613049da99a0bf8aa5991509 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89bb136a1319497caff6b9eb78b002fd : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/319599a502fd470d99f5a01b1fecd391 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79f8d37804b14ad985cf5d6305ff791d : SUCCESS in 6m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/40ac7d5dcb4f4cfeb34e7a07f8cfa09d : SUCCESS in 30m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0a426744bbc94382bf80cd3623ea861c : SUCCESS in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ed5a6492bc1c7689078da01ddedead82c8d86eac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-28 04:42:07.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"0dc4d8c45f95ae2d1278da5de78200509ded393d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-28 05:44:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2692d28aaa54a8cb29a1a2d82726de0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9bac172281444011b9f97a367943ad33 : SUCCESS in 37m 09s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3681e30c743d45d1a0e4e0e7598adacf : FAILURE in 23m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1310d44540bd4b188fc680cb2d3c5343 : SUCCESS in 35m 09s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/273aeb3d823a4e239957d3532760d3e5 : SUCCESS in 35m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f8628ed649494d0093e04fc13a6f32a2 : SUCCESS in 40m 45s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/dfd5423756c14bf697700eb8d0fa53ba : SUCCESS in 52m 22s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0a7dc3f85e9b43f8a4aa292ca0112868 : SUCCESS in 53m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b4541058124546fba506d7378804dc53 : SUCCESS in 38m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b9d1880e86d842b8b00134ea67ce1cc0 : SUCCESS in 50m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9f96f50d65845888b08aa628f52c8cd : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7002d05356c462498246a693bbc2c61 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9c3e0b13c75494e817e41e93b73d83f : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f0e828c366843f7b5b1e29109709c95 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63b492e4e7a44e418b66b67b0f6cc147 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/785bb8c5036e45498c180247aeebf4cc : SUCCESS in 6m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3be7cfbf71f2466b9381ae61cfc62a82 : SUCCESS in 34m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ab26cf6feb6c4ddbb74c9c4e67f4edb6 : SUCCESS in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"02deddee46548df6a829f40f5d110ff4cb701cb3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-30 20:38:27.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"74bd2bbfd55278d0c52c6aab96d3babcfd3e2c82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 20:39:00.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"725489b58016bff329a7245bc7e8273fb6696e00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 22:05:25.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdc500c0f013410e92deae3497e0412f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dfd71da3e386474eb1a0ae4be88ce947 : SUCCESS in 56m 08s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/10db2175e8a74a968b7e476bd5b360f0 : SUCCESS in 56m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a09ea9442f9b4b588598a29b445c3617 : SUCCESS in 48m 20s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/321d176d2b6f41078ff713f70fb4e3a8 : SUCCESS in 1h 16m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/822398084bae44f4bc0958249c227fea : SUCCESS in 48m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a025a0ca8266415185441e65aa995b9b : SUCCESS in 48m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f2afa694280485abc8808d42185ad00 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fec4f4e1d7a44e2a83ec7e3d51e721a3 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/706c9153913d423e80bd82903f5963a0 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08ff9e1d8cc34574a243d4e3a6b22323 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a671dcf1a95b4c51942ad3c698abe2ea : SUCCESS in 8m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e67c7e52f5a04f1e97868c03a53e8d57 : SUCCESS in 32m 19s","accounts_in_message":[],"_revision_number":7},{"id":"0428d4ffd1cc66d4cf6282e9be521db96f236ad5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 22:05:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"25fe28627bfae5bd11d48996e8622a457b331c68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 22:06:56.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/509151f55471468ca8c38693b8da9e26\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9d3b38a5bbf484bb946da4851f45fa4 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"9ba81579acd6d583109ea6ed50dcc9c18ba4412a","revisions":{"0630253b98c1c027a1e0098f73d5fc4cf0ed65a1":{"kind":"REWORK","_number":1,"created":"2023-05-27 02:55:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/884346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/884346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/884346/1"}}},"commit":{"parents":[{"commit":"fa715224bd2dcf327305f83dd9bd93a63c53781d","subject":"Add manage cmd tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa715224bd2dcf327305f83dd9bd93a63c53781d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 02:54:24.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 02:54:24.000000000","tz":-420},"subject":"Improve Zone/Record central status coverage","message":"Improve Zone/Record central status coverage\n\nChange-Id: I1b00300118340faecc81c33f3a77e46e151e0a1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0630253b98c1c027a1e0098f73d5fc4cf0ed65a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0630253b98c1c027a1e0098f73d5fc4cf0ed65a1"}]},"branch":"refs/heads/master"},"5166614eb57fc33de22ca0190f15b83a8c441227":{"kind":"REWORK","_number":2,"created":"2023-05-27 05:08:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/884346/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/884346/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/884346/2"}}},"commit":{"parents":[{"commit":"fa715224bd2dcf327305f83dd9bd93a63c53781d","subject":"Add manage cmd tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa715224bd2dcf327305f83dd9bd93a63c53781d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 02:54:24.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 05:08:35.000000000","tz":-420},"subject":"Improve Zone/Record central status coverage","message":"Improve Zone/Record central status coverage\n\nChange-Id: I1b00300118340faecc81c33f3a77e46e151e0a1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5166614eb57fc33de22ca0190f15b83a8c441227"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5166614eb57fc33de22ca0190f15b83a8c441227"}]},"branch":"refs/heads/master"},"d15be460004a7af53812857bc7c42596d5ccc9cd":{"kind":"REWORK","_number":3,"created":"2023-05-27 19:41:26.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/884346/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/884346/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/884346/3"}}},"commit":{"parents":[{"commit":"fa715224bd2dcf327305f83dd9bd93a63c53781d","subject":"Add manage cmd tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa715224bd2dcf327305f83dd9bd93a63c53781d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 02:54:24.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 19:41:16.000000000","tz":-420},"subject":"Improve Zone/Record central status coverage","message":"Improve Zone/Record central status coverage\n\nChange-Id: I1b00300118340faecc81c33f3a77e46e151e0a1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d15be460004a7af53812857bc7c42596d5ccc9cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d15be460004a7af53812857bc7c42596d5ccc9cd"}]},"branch":"refs/heads/master"},"87dbe16a092c40751dc21f4c6ed02338ba8e6f79":{"kind":"REWORK","_number":4,"created":"2023-05-27 19:47:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/884346/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/884346/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/884346/4"}}},"commit":{"parents":[{"commit":"fa715224bd2dcf327305f83dd9bd93a63c53781d","subject":"Add manage cmd tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa715224bd2dcf327305f83dd9bd93a63c53781d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 02:54:24.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 19:47:37.000000000","tz":-420},"subject":"Improve Zone/Record central status coverage","message":"Improve Zone/Record central status coverage\n\nChange-Id: I1b00300118340faecc81c33f3a77e46e151e0a1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/87dbe16a092c40751dc21f4c6ed02338ba8e6f79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/87dbe16a092c40751dc21f4c6ed02338ba8e6f79"}]},"branch":"refs/heads/master"},"8501fd7cc349ff9317a1ff922b4666cf84b0afbe":{"kind":"REWORK","_number":5,"created":"2023-05-27 21:10:24.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/884346/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/884346/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/884346/5"}}},"commit":{"parents":[{"commit":"fa715224bd2dcf327305f83dd9bd93a63c53781d","subject":"Add manage cmd tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa715224bd2dcf327305f83dd9bd93a63c53781d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 02:54:24.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 21:10:05.000000000","tz":-420},"subject":"Improve central unit test coverage","message":"Improve central unit test coverage\n\nChange-Id: I1b00300118340faecc81c33f3a77e46e151e0a1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8501fd7cc349ff9317a1ff922b4666cf84b0afbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8501fd7cc349ff9317a1ff922b4666cf84b0afbe"}]},"branch":"refs/heads/master"},"24b8c0e7fb74f8a13319a0152741d06b6551eff0":{"kind":"REWORK","_number":6,"created":"2023-05-27 22:07:31.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/884346/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/884346/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/884346/6"}}},"commit":{"parents":[{"commit":"fa715224bd2dcf327305f83dd9bd93a63c53781d","subject":"Add manage cmd tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa715224bd2dcf327305f83dd9bd93a63c53781d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 02:54:24.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 22:06:54.000000000","tz":-420},"subject":"Improve central unit test coverage","message":"Improve central unit test coverage\n\n- Minor changes to fix code that wasn\u0027t unit testable.\n\nChange-Id: I1b00300118340faecc81c33f3a77e46e151e0a1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/24b8c0e7fb74f8a13319a0152741d06b6551eff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/24b8c0e7fb74f8a13319a0152741d06b6551eff0"}]},"branch":"refs/heads/master"},"9ba81579acd6d583109ea6ed50dcc9c18ba4412a":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-05-28 04:42:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/884346/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/884346/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/884346/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/884346/7"}}},"commit":{"parents":[{"commit":"661f0dd5175fb0334a7f0f6f1ff4482164d09b91","subject":"Add manage cmd tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/661f0dd5175fb0334a7f0f6f1ff4482164d09b91"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-27 02:54:24.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-28 04:42:07.000000000","tz":0},"subject":"Improve central unit test coverage","message":"Improve central unit test coverage\n\n- Minor changes to fix code that wasn\u0027t unit testable.\n\nChange-Id: I1b00300118340faecc81c33f3a77e46e151e0a1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ba81579acd6d583109ea6ed50dcc9c18ba4412a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ba81579acd6d583109ea6ed50dcc9c18ba4412a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
