)]}'
{"id":"openstack%2Fdesignate~884347","triplet_id":"openstack%2Fdesignate~master~I235ac441456016c9bd9caf04637ec633812dc007","project":"openstack/designate","branch":"master","topic":"remove_find_zone_share","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-05-30 18:44:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I235ac441456016c9bd9caf04637ec633812dc007","subject":"Remove unused find_zone_share code","status":"MERGED","created":"2023-05-29 03:19:03.000000000","updated":"2023-05-30 18:45:10.000000000","submitted":"2023-05-30 18:44:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884347-remove_find_zone_share","meta_rev_id":"4c69661592714ab3bf58c48c98d615d409db7f96","_number":884347,"virtual_id_number":884347,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-30 18:44:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-05-30 17:46:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-05-30 17:46:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-29 04:34:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-30 17:46:30.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"262b1c22e9075c3cf7aa5ef925d7192996d601fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-29 03:19:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"889f97dd2f6512d00d09310ba640c0dd1c29c009","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 04:34:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8636d69181e64f4ab8e6025e338851b9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d1424819dafd4c308d54d5791ddce70f : SUCCESS in 40m 10s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/61c1fbf1e7184d178609ccbae6f701ff : FAILURE in 27m 55s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd30616c98b441e08c359e62ff777197 : SUCCESS in 1h 08m 18s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/646f845f3377477e90f1a308ae526018 : SUCCESS in 49m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fc18ba3bb99b4a858f1176266a53a2f4 : SUCCESS in 36m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0cc11cf55b5041b585f0237c22498d93 : SUCCESS in 52m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fd79d12b032a466db7d2ad4536c62a88 : SUCCESS in 51m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2f60a1795b9f401abe2ff1876ff2035d : SUCCESS in 33m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/243043a44c634327b5c221309d3da69e : SUCCESS in 37m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18e4c57013794d1189318fed8052130a : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09e739c4fe69403e880ba143b292a751 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e3b9db0fc3a47c6b2df741efa2abd88 : SUCCESS in 14m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b878e8a9212e4539b40092af29a9646a : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a7d7c4446084175be71f9e1caab8c09 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c1d65a5ab4845b48919c4e033cb0495 : SUCCESS in 23m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4bb1327941b84714bf55f2066bc0f6c1 : SUCCESS in 32m 44s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/faaa595b49874d44b963647501c99ddb : SUCCESS in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e1a2e9c76a94cf637c05870f1ddac0ae99901f7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-30 17:46:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b95cc0e8d0235b12fb67194dc836fbe26d8cee8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 17:47:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fddf94cde58bdbf2f7ad5884801bb50c2fc66e59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 18:44:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cc581560a9c4c7292bac4b468533046\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f82a2393d60643e9bf18b25bcfd466e2 : SUCCESS in 40m 49s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/69a2641a2b724ea587c4c75d7f0c8e4f : SUCCESS in 41m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ecc3115eedae4d448c0d7bc1d9375f34 : SUCCESS in 34m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0e27810dd4ba4ac68de87b6cbf0124a2 : SUCCESS in 56m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/113e4c4bfd874af891ec0561fd0fe484 : SUCCESS in 49m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/082b37384fb24d75a174d335fb623ad5 : SUCCESS in 45m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63578b4951864f30930b860a83ec9427 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f00ac44dc8a6490891b744b3cc00950b : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b638584567af461ea3e1e10176aafe50 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b2256c9c5dc4692a7ded2d4d2ea966c : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1b281aeb422444f97804a13f6cd6e2a : SUCCESS in 8m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/19f61b47e72640d9ac91d8b204454a37 : SUCCESS in 30m 45s","accounts_in_message":[],"_revision_number":1},{"id":"370f3d5cd1ccbad7747dfd288889c856214f78a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 18:44:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4c69661592714ab3bf58c48c98d615d409db7f96","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 18:45:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaa7e53615c447adb4adfbab2d8154ba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a6a2d54e3a145588de1c36b61bc2b30 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9620686529b38ff6c862f42c4183d035ea3156bf","revisions":{"9620686529b38ff6c862f42c4183d035ea3156bf":{"kind":"REWORK","_number":1,"created":"2023-05-29 03:19:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/884347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/884347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/884347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/884347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/884347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/884347/1"}}},"commit":{"parents":[{"commit":"48826ffa96b0ed358ea814657d31fb5ba28966ee","subject":"Fix create zone export RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/48826ffa96b0ed358ea814657d31fb5ba28966ee"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-29 03:18:55.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-29 03:18:55.000000000","tz":-420},"subject":"Remove unused find_zone_share code","message":"Remove unused find_zone_share code\n\nChange-Id: I235ac441456016c9bd9caf04637ec633812dc007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9620686529b38ff6c862f42c4183d035ea3156bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9620686529b38ff6c862f42c4183d035ea3156bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
