)]}'
{"id":"openstack%2Fdesignate~884348","triplet_id":"openstack%2Fdesignate~master~I519d9e648e684238ca19d00df6abf9b472e3c666","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-05-30 22:30:53.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-05-31 00:09:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I519d9e648e684238ca19d00df6abf9b472e3c666","subject":"Improve storage unit test coverage","status":"MERGED","created":"2023-05-29 04:51:28.000000000","updated":"2023-05-31 00:10:33.000000000","submitted":"2023-05-31 00:09:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884348","meta_rev_id":"3dce471ebc6ead4959045d7634ce2219a9d55548","_number":884348,"virtual_id_number":884348,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-31 00:09:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-30 22:30:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-30 22:30:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-29 06:21:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-30 20:44:09.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"ce3177a372cbd77dfe675f843251d5c0b57957e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-29 04:51:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3193f4cd54bbb8372acc9b1f4e550e4a0070e36d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-29 05:13:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4527048552360477b1e4d8ddf37e140e09fce3ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-29 05:17:30.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c440a46594ef3b85d3147c83ba17d0487678d9c0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-29 05:17:57.000000000","message":"Topic storage removed","accounts_in_message":[],"_revision_number":3},{"id":"06ee7801d0fb8bfdfff42491b70f0ec8a53d55a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 06:21:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/056a69bd5b8a4295bc4d3a1f7e722dff\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c176cd8ad28547b6aa3dddc81c0cc159 : SUCCESS in 51m 00s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c4f6d749f65b4539acb090764a3895e9 : FAILURE in 23m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7249e1b9c15f4064ad648c49400119ef : SUCCESS in 59m 16s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/862369ce847541c392526c50b81b5934 : SUCCESS in 48m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/624dca6fee6543dcb6e8d3a2077e041a : SUCCESS in 37m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/97416ae972cb4384b6aba561b9e5d4f7 : SUCCESS in 51m 35s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/41b7054b8247488994b9fbe1afdc1e59 : SUCCESS in 41m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2bd51b9581054d93a03c6d02e1d3d21a : SUCCESS in 33m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6aacf550c8ad4f82b5734b34bc8b6626 : SUCCESS in 36m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d5c72eeed684abaa0e90feede87aff0 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd02081e8d1d45a3bd2c5c1b3e089195 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53ceeb6994d8424b80a819ffc55ee011 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d42d001747d4974a5d11915bbc0fefb : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea2cd7ca99e246b8abe999f3cfc4139e : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a53d041dadd45ad9558e6e3c36c05ac : SUCCESS in 7m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f2e369613e874c899c0bb58e538883f0 : SUCCESS in 30m 38s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a39482f9540f42c2820ca6f419057ee4 : SUCCESS in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c6c2fb0f9e856ed30bbc55c447867c4376d21f3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-30 06:26:45.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"958deda701fa66e2b688c3a6f209e5aede3920aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 07:41:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee6446c7641746cf87a6a99fa520328c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/158ae9ecbaf745cbb86d9d26317e607b : SUCCESS in 51m 09s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/21a4a60d6a2f474d84457b82e996a1d5 : SUCCESS in 1h 09m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b41e14dc40a747f8871b56f2a2be2ac3 : SUCCESS in 58m 38s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6d7af204355c47cc8c2d119c596f8b11 : SUCCESS in 44m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/88755a63c1fd497db6da61158f30f75a : SUCCESS in 39m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bf495eed63224c32b7ef6dda3484cfbe : SUCCESS in 58m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d99b0b0576684abfba40146ee0dd9987 : SUCCESS in 54m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/90e76a568a494d5898cd93f5b7a3de3c : SUCCESS in 45m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/837e9889f8004bd898b7c14f41555097 : SUCCESS in 56m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62da1f7721d84688921edda6d90976c8 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7ef38332edd43209d399840137fc5c7 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a3a90b6b6814dc3a8c4d34d1d32f0a7 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/304272dc46794cf58fa339637398cfa7 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9d5cf647e034232912b804447ead3cd : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a5b5a3d0664212889490c3c7a1e565 : SUCCESS in 8m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9c0cfd16e8a14b74a3f17bfac863466d : SUCCESS in 40m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/db7db944a7e347b0917c71e708031dde : SUCCESS in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f22a77e7fa7472012b75fbe42a7fe4ffa02198ff","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-30 20:44:09.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f1e375201d34314482a20976f1946f73c2b06e35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-30 20:56:17.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"6875aa5c73a16ad03fb4dbcaed53b26402508aad","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-30 20:56:32.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"aebfdd002763475ffd21e0bf7e3834609dd2be5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-30 21:51:54.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"80e360206ba60ef63aecf5d83c29da6af615f165","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-30 22:30:53.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b6716569dfb7b3406335506a0541116a9714eaad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 22:47:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79e7ed8df567416ea6a911e8814799e8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/adc28d0d004041b1a928f89d2aec7399 : SUCCESS in 40m 49s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4c63cdfbf7ba4cfb91a32ebe2d34b1b1 : SUCCESS in 50m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e510ff94de9f4b559dbfdb19bf123d6a : SUCCESS in 43m 02s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6c60016eedb443f99481b238387b1289 : SUCCESS in 47m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/91beedd692c04743965f12e71ae005a8 : SUCCESS in 41m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/563ef4dfe43c44a4966f66626336684f : SUCCESS in 54m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7d2ae7d522d643518834359d00cdd443 : SUCCESS in 52m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/df0d08cb03df4a6c9fcb1a646e40ed2f : SUCCESS in 42m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a35e5b0eee18431cb09a1b13800a24b3 : SUCCESS in 44m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/baa967b1a6a942d981b7e1a1658a7d3f : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a73504c7c3144aa7abd123bd6bf3c644 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9618b3aea320483896b712131772f83e : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd2cff4a59294560b087160e28950eda : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47d8cb8361d142a0888388ac856cdb8e : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7291752730d944f5b8311adde24775a9 : SUCCESS in 8m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a3c38fc72b5841068c3516b0feae5f21 : SUCCESS in 32m 29s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/718ea2de16264db2aa420879164dc383 : SUCCESS in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"475e5c0745dc29583887f5667958f4046f56f207","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 22:48:25.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d95c1456dd0e13c2323298fd05276e1033e8a408","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 00:09:16.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/935fb81e9a344f0e950ae82e8ae12439\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/09745d70a8b7425fbaba295b48aa3ad6 : SUCCESS in 48m 38s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/bf9cbb95032f40a796e5e8fded87fba7 : SUCCESS in 52m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/deb296f0bf494f649b4ad81018d6d005 : SUCCESS in 47m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c3efe735c3944a8490458b2b206939f3 : SUCCESS in 1h 20m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b52af588c3664e57b224ef3f30a6cc72 : SUCCESS in 42m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c2cfa6075d064d22991cd527e02f6133 : SUCCESS in 35m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79b32d250cea479a83794ed51f5f14da : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5fd6da1f86d426f90b442b4c9485dda : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccdb7b5fba9047adb1fdcb38c67e6e71 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40155b2aa1ee4115b5ead128d023e0e9 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01ecc5bb8ac94b56b4cc95f5bbdbeaa9 : SUCCESS in 7m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/09b3e05f532b45ceb41b44e92c9b330b : SUCCESS in 31m 21s","accounts_in_message":[],"_revision_number":6},{"id":"752db87c0c8cc55df00f32a551a535a6585d4e06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 00:09:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"3dce471ebc6ead4959045d7634ce2219a9d55548","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 00:10:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b97f721484314af5a45e9b17e4e55a87\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d5862d79f2c43a68230da87661c3e2b : SUCCESS in 59s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a7cbcc0770d09da7daca6ba34a8aa3985f0cc2ba","revisions":{"bd776b1274282058092f74326632de3e6139d77f":{"kind":"REWORK","_number":1,"created":"2023-05-29 04:51:28.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/884348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/884348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/884348/1"}}},"commit":{"parents":[{"commit":"9ba81579acd6d583109ea6ed50dcc9c18ba4412a","subject":"Improve central unit test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ba81579acd6d583109ea6ed50dcc9c18ba4412a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-29 03:16:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-29 04:50:48.000000000","tz":-420},"subject":"Improve storage test coverage","message":"Improve storage test coverage\n\n- Minor cleanup to tests.\n- Use assertRaisesRegex for exception testing.\n\nChange-Id: I519d9e648e684238ca19d00df6abf9b472e3c666\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bd776b1274282058092f74326632de3e6139d77f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bd776b1274282058092f74326632de3e6139d77f"}]},"branch":"refs/heads/master"},"0be88cb88658737d39de9d7a9935f024a2f8ca8e":{"kind":"REWORK","_number":2,"created":"2023-05-29 05:13:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/884348/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/884348/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/884348/2"}}},"commit":{"parents":[{"commit":"9ba81579acd6d583109ea6ed50dcc9c18ba4412a","subject":"Improve central unit test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ba81579acd6d583109ea6ed50dcc9c18ba4412a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-29 03:16:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-29 05:13:00.000000000","tz":-420},"subject":"Improve storage test coverage","message":"Improve storage test coverage\n\n- Minor cleanup to tests.\n- Use assertRaisesRegex for exception testing.\n\nChange-Id: I519d9e648e684238ca19d00df6abf9b472e3c666\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0be88cb88658737d39de9d7a9935f024a2f8ca8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0be88cb88658737d39de9d7a9935f024a2f8ca8e"}]},"branch":"refs/heads/master"},"a8120ed651e742f0271f975b1a815d74c755e60d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-05-29 05:17:30.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/884348/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/884348/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/884348/3"}}},"commit":{"parents":[{"commit":"9ba81579acd6d583109ea6ed50dcc9c18ba4412a","subject":"Improve central unit test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ba81579acd6d583109ea6ed50dcc9c18ba4412a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-29 03:16:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-29 05:17:30.000000000","tz":0},"subject":"Improve storage unit test coverage","message":"Improve storage unit test coverage\n\n- Minor cleanup to tests.\n- Use assertRaisesRegex for exception testing.\n\nChange-Id: I519d9e648e684238ca19d00df6abf9b472e3c666\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a8120ed651e742f0271f975b1a815d74c755e60d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a8120ed651e742f0271f975b1a815d74c755e60d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"13bbc7934d4f27557b8aac1c98ce987f0fdefd00":{"kind":"REWORK","_number":4,"created":"2023-05-30 06:26:45.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/884348/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/884348/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/884348/4"}}},"commit":{"parents":[{"commit":"9ba81579acd6d583109ea6ed50dcc9c18ba4412a","subject":"Improve central unit test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ba81579acd6d583109ea6ed50dcc9c18ba4412a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-29 03:16:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 06:26:37.000000000","tz":-420},"subject":"Improve storage unit test coverage","message":"Improve storage unit test coverage\n\n- Minor cleanup to tests.\n- Use assertRaisesRegex for exception testing.\n\nChange-Id: I519d9e648e684238ca19d00df6abf9b472e3c666\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/13bbc7934d4f27557b8aac1c98ce987f0fdefd00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/13bbc7934d4f27557b8aac1c98ce987f0fdefd00"}]},"branch":"refs/heads/master"},"7a1ac70026db93174d675276ca82dfc87a5aaf76":{"kind":"REWORK","_number":5,"created":"2023-05-30 20:56:17.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/884348/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/884348/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/884348/5"}}},"commit":{"parents":[{"commit":"9ba81579acd6d583109ea6ed50dcc9c18ba4412a","subject":"Improve central unit test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ba81579acd6d583109ea6ed50dcc9c18ba4412a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-29 03:16:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 20:56:17.000000000","tz":0},"subject":"Improve storage unit test coverage","message":"Improve storage unit test coverage\n\n- Minor cleanup to tests.\n- Use assertRaisesRegex for exception testing.\n\nChange-Id: I519d9e648e684238ca19d00df6abf9b472e3c666\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a1ac70026db93174d675276ca82dfc87a5aaf76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a1ac70026db93174d675276ca82dfc87a5aaf76"}]},"branch":"refs/heads/master"},"a7cbcc0770d09da7daca6ba34a8aa3985f0cc2ba":{"kind":"REWORK","_number":6,"created":"2023-05-30 21:51:54.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/884348/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/884348/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/884348/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/884348/6"}}},"commit":{"parents":[{"commit":"9ba81579acd6d583109ea6ed50dcc9c18ba4412a","subject":"Improve central unit test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ba81579acd6d583109ea6ed50dcc9c18ba4412a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-29 03:16:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 21:51:54.000000000","tz":0},"subject":"Improve storage unit test coverage","message":"Improve storage unit test coverage\n\n- Minor cleanup to tests.\n- Use assertRaisesRegex for exception testing.\n\nChange-Id: I519d9e648e684238ca19d00df6abf9b472e3c666\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a7cbcc0770d09da7daca6ba34a8aa3985f0cc2ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a7cbcc0770d09da7daca6ba34a8aa3985f0cc2ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
