)]}'
{"id":"openstack%2Fdesignate~884349","triplet_id":"openstack%2Fdesignate~master~I3dc80fac13e60016e0611410b269bc487d93a39b","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-05-30 22:31:11.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-05-31 00:28:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3dc80fac13e60016e0611410b269bc487d93a39b","subject":"Increase minimum acceptable test coverage","status":"MERGED","created":"2023-05-29 05:25:04.000000000","updated":"2023-05-31 00:29:38.000000000","submitted":"2023-05-31 00:28:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884349","meta_rev_id":"c24a482d350a3f253ae4a0ed89757893758198b0","_number":884349,"virtual_id_number":884349,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-31 00:28:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-30 22:31:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-30 22:31:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-29 06:39:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-30 20:45:15.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"82d0cf9d4be2f144f81f458a1c2810caefb58946","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-29 05:25:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1846d30127fc102b0013c6543a297a4f7c3aa004","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-29 05:25:19.000000000","message":"Topic coverage removed","accounts_in_message":[],"_revision_number":1},{"id":"6fd299328bc9665bd38be937b29bdc970a7d2d35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 06:39:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7249f6df898141e692195b521c1a24d7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4033654205434bf38e9b7f35e58e5c69 : SUCCESS in 3m 36s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e8976e83f05f449a81b634bf32aad6d1 : SUCCESS in 49m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/33ec7770e8b747d5b9a129f02b94b8c5 : FAILURE in 26m 49s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/47e1c22016c9413d9810b709587cd7c5 : SUCCESS in 1h 01m 19s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0ac3b331b2664117bc05608db07402e4 : SUCCESS in 48m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/985bf99f3af749f8a97758cf7a2c34f5 : SUCCESS in 39m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/50e0f5cd7e534a04ac4c5cb4177373c1 : SUCCESS in 1h 09m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bed4ea2b30a649de8118563be353a8c6 : SUCCESS in 1h 07m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3d65c5085585457ca84ac2ad372e1385 : SUCCESS in 36m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/324453bdedad428792c222110497fc05 : SUCCESS in 48m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e04a50ee4f6b4f9fbd4756aecb45bc2f : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e8dc5198e5d43259cd2844f6ee45cc3 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5e1906aa52d4c38a9f8c9e2642445db : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd8517db794244a9b78e20a71eb37162 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44dd9827c6ae49ec81a6ce6bd3823aa3 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/404e19ad0e264ce182615a23f989c0c0 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca44e327884045cab12505631e11c1cc : SUCCESS in 3m 48s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6764eeecd21c4a3c95e394cb7a98d002 : SUCCESS in 3m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a3a970e042e1e70615b3670004bdadda736f032","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-30 06:26:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b619aa13ba56e4c5193507c5a6a7d67e3765abec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 07:28:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aff820e384fb4313a4825451e60bc7d6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4fd5940ceb2049598d0a558f5f69bf42 : SUCCESS in 3m 42s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/090980f0254e492eb7daaa2efc1c8b54 : SUCCESS in 53m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/561cb90cd8eb4be8908b831d3f8057a6 : SUCCESS in 51m 01s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/65f992cd36434d0db206b2547c52abdd : FAILURE in 47m 50s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/685636536e9f48d9a86b61209ea55264 : SUCCESS in 43m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/801eded47b8c4935aa45605185c62ecf : SUCCESS in 39m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4116d2dd05b5441e98b7b71c7ba033ba : SUCCESS in 58m 58s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d458f6fa89294de588bfa19da254bf1b : SUCCESS in 58m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2bab5f0a346241efa7bc61fe7b545399 : SUCCESS in 39m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0ae7b34c63fe4817ab6c55c5405b42b0 : SUCCESS in 43m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b5cef0061d344a6b8eb243af67f2f54 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5405e011eb6c44e786b79069abb7757d : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ff0d6ea8b814831ad2ba1ad7086164b : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c1f17e49ffc451881c69ad502ca1619 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36b20e1348194b35b50b1acff85bdb7f : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/052e6779d697468e8553d1b9569caf98 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/171ceb5e6e9a4eca860a30ea80032f56 : SUCCESS in 4m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/38508a3b91ba49fdb5d40944e526198f : SUCCESS in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"24936cf229a7aa67bb4b76626aa4044b95175b6a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-30 20:45:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2f0c4af8d87250ddeb4e74e1f27f113dc6948946","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-30 20:56:36.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c43e2db33d698e2d465faef794c419bf68f20cf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-30 21:52:16.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"8e930794ebd4ce875d970b2ef808ab5f0d7186dc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-30 22:31:11.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5ff89525930c8697a141aff5475d940b63b50e79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 23:13:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/031a32e2351e4c439cf13c8284e35da7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1244d2df85c84cd4953139c2ac908206 : SUCCESS in 3m 41s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a8b48c85452a4d2397dbfe5ff6bc1526 : SUCCESS in 42m 33s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/94c39ad8423d42edb9b289492cff1910 : SUCCESS in 47m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf144fe8682e4b7787de4477b28efdbe : SUCCESS in 45m 01s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6cce5b43578246d6a1ab591a20df1b80 : SUCCESS in 38m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2ba0f13272af42d3841900fe970042a5 : SUCCESS in 37m 24s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6517fa7a77ba491ebaf6d17fc5720b2d : SUCCESS in 1h 05m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3dfab43655524356ac2ead23e653f2c3 : SUCCESS in 1h 13m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/66cd921c8cd2445799c6521d5360bf6a : SUCCESS in 57m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4fd9699b07a345a2b5a3dc5fecf24cbc : SUCCESS in 43m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2901b78f814143cebd618009bbd884ef : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ae8a58b47354f6a8dabe6a2dc391239 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93f9601e47b949828e1b332ff95fd923 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36023938956a4bdcb6e0f4ea54112f5e : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bcc36ca5a76d428f928a19b8ced3d2ac : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51edb14583864c9e9a61d3a470a1558f : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d4045bc55664426bb8ba699c27dab3d : SUCCESS in 4m 29s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7359f4fb390c47eaa65186d9b7eb3b11 : SUCCESS in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2e81b99dde180df450a2be774422cc0500eff96d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 23:14:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2c70ac4a4ffc336e560b9357ee526394b08d8763","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 00:28:36.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58b649afe017468aaea266f8d110442d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bea1e774924f4eb1943808449c9f0d09 : SUCCESS in 4m 00s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/99abec2e04b44d50a8dd074435bd12e0 : SUCCESS in 43m 15s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8e232ba4d21c4670aab8256f76378701 : SUCCESS in 32m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/14790f2424f34fffa7c7dae76940054d : SUCCESS in 42m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/51b6914812654e06a7c15a156d6fbc37 : SUCCESS in 1h 09m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3bf18078c50345b8925be6ec094d0922 : SUCCESS in 56m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f310f62a5eaf4ff494da7cdfa4d59b57 : SUCCESS in 46m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72a2a47de51f46bfb2aa28603b815707 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/599af9f98858404e980edf67485d0487 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c243cb26d3804d118c7072417b48ccb7 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee0d0787ea1a4702a167c35bd3c1de9c : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e714f37650604e6eafc745745447f3cf : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3279650a7d814d4e93b74e88ece5c28d : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":4},{"id":"c71fb7e7a803cf8989fcf3ff588ba8a3be67f12c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 00:28:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c24a482d350a3f253ae4a0ed89757893758198b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 00:29:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20066771d1b741809230f551213f59c2\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/9790fdc63c5344fe8ea988464801e641 : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c7dbe675eb64fa9837d2ec2a36d0065 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ae89bd70d6d4d6685e0675f325d62f7 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a53c685adfd8bf1db57443d1fa2327a47cc339e1","revisions":{"c032a82e71a2b2c087c251ce3f80f61bef8f379a":{"kind":"REWORK","_number":1,"created":"2023-05-29 05:25:04.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/49/884349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/49/884349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/884349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/884349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/884349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/49/884349/1"}}},"commit":{"parents":[{"commit":"a8120ed651e742f0271f975b1a815d74c755e60d","subject":"Improve storage unit test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a8120ed651e742f0271f975b1a815d74c755e60d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-29 05:24:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-29 05:24:52.000000000","tz":-420},"subject":"Increase minimum acceptable test coverage","message":"Increase minimum acceptable test coverage\n\nChange-Id: I3dc80fac13e60016e0611410b269bc487d93a39b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c032a82e71a2b2c087c251ce3f80f61bef8f379a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c032a82e71a2b2c087c251ce3f80f61bef8f379a"}]},"branch":"refs/heads/master"},"74ace85d32202089da3222f5e67a86f2178008be":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-05-30 06:26:58.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/49/884349/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/49/884349/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/884349/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/884349/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/884349/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/49/884349/2"}}},"commit":{"parents":[{"commit":"13bbc7934d4f27557b8aac1c98ce987f0fdefd00","subject":"Improve storage unit test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/13bbc7934d4f27557b8aac1c98ce987f0fdefd00"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-29 05:24:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 06:26:58.000000000","tz":0},"subject":"Increase minimum acceptable test coverage","message":"Increase minimum acceptable test coverage\n\nChange-Id: I3dc80fac13e60016e0611410b269bc487d93a39b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/74ace85d32202089da3222f5e67a86f2178008be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/74ace85d32202089da3222f5e67a86f2178008be"}]},"branch":"refs/heads/master","description":"Rebase"},"1608ff347876860e9c780d5bacdaf8743ebcb940":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-05-30 20:56:36.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/49/884349/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/49/884349/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/884349/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/884349/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/884349/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/49/884349/3"}}},"commit":{"parents":[{"commit":"7a1ac70026db93174d675276ca82dfc87a5aaf76","subject":"Improve storage unit test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a1ac70026db93174d675276ca82dfc87a5aaf76"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-29 05:24:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 20:56:36.000000000","tz":0},"subject":"Increase minimum acceptable test coverage","message":"Increase minimum acceptable test coverage\n\nChange-Id: I3dc80fac13e60016e0611410b269bc487d93a39b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1608ff347876860e9c780d5bacdaf8743ebcb940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1608ff347876860e9c780d5bacdaf8743ebcb940"}]},"branch":"refs/heads/master","description":"Rebase"},"a53c685adfd8bf1db57443d1fa2327a47cc339e1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-05-30 21:52:16.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/49/884349/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/49/884349/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/884349/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/884349/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/884349/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/49/884349/4"}}},"commit":{"parents":[{"commit":"a7cbcc0770d09da7daca6ba34a8aa3985f0cc2ba","subject":"Improve storage unit test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a7cbcc0770d09da7daca6ba34a8aa3985f0cc2ba"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-29 05:24:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 21:52:16.000000000","tz":0},"subject":"Increase minimum acceptable test coverage","message":"Increase minimum acceptable test coverage\n\nChange-Id: I3dc80fac13e60016e0611410b269bc487d93a39b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a53c685adfd8bf1db57443d1fa2327a47cc339e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a53c685adfd8bf1db57443d1fa2327a47cc339e1"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
