)]}'
{"id":"openstack%2Fdesignate~884560","triplet_id":"openstack%2Fdesignate~master~I2ef28eeb285462dc12b13de80cdab06b5cfb929b","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-05-27 07:12:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2ef28eeb285462dc12b13de80cdab06b5cfb929b","subject":"Fix create zone export RBAC","status":"MERGED","created":"2023-05-26 23:59:36.000000000","updated":"2023-05-27 07:13:15.000000000","submitted":"2023-05-27 07:12:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884560","meta_rev_id":"f9d348bb82c6ce2f4aeeba228453bb21db598096","_number":884560,"virtual_id_number":884560,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-27 07:12:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-05-27 03:16:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-27 00:56:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-05-27 03:16:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-27 00:56:57.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2023-05-27 01:09:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0e95980aa54986122e23e60058e7424f7b52914","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-26 23:59:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fb93877da627ed0c85e1863679ceaf731c5fa65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-27 00:11:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2dc23774d951be74eb466f5131fb5c46a78e8028","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-27 00:56:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e20675582e141c4daeee085ed426348f1f1d5e4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 01:09:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c43206a9673b43c49b56ff926596b972\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/76ab297cd272440a96fbd45180cc5b0b : SUCCESS in 48m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/faad43f2bc834663976392f5bd53e87c : FAILURE in 12m 51s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/389988ae4fe34d8f9ceeded41a392842 : SUCCESS in 35m 35s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/979aa35df9ba49e0948159d2fc233c0b : SUCCESS in 48m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c9e0e4487d0e4f699743e31568f9b99c : SUCCESS in 40m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9cd9399ed1e3454991d4f3e0efc28d2c : SUCCESS in 57m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4e21b2a70f2e475bb17f92aacdd0247e : SUCCESS in 51m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f18a1c85639a4b25a5377edd27005a5f : SUCCESS in 37m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/191ab95db781458ebb31dae2e0c9aa07 : SUCCESS in 34m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82a6b8449abc4c30af2c49090b730622 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57803f6ea1ff4c87996f7cd5d09eda68 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe22fd78091c4113965620d17083fc70 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c327ced249b494494e5c82887a5ea56 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1292b2f99ebe452e9d605c46cf393b95 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e5343a5134d47d887486ef0307f0599 : SUCCESS in 8m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e82e1fcbeb4a4110829e369752326b5e : SUCCESS in 39m 10s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/298bf2c09c144642b2c205bff9ea6d0e : SUCCESS in 3m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8fdfadd3026174a52ca4b48f75b0031eb29d6646","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-27 03:16:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ef443c3ac2d65737cf13121af7f28f65bb3bd8ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 03:16:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9d8ad3014ca37d3f2b6434055edf528960bf63ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 03:52:28.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0eebeb3b67b9422ab8dac06459d98a7a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/544401136d8e4ff794589928691baa9e : CANCELED\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/fcae725706794fc18bdb9272f49a5718 : CANCELED\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f26da44bf88541a3aea6b57ef78e3d80 : FAILURE in 34m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/26ee401e526d455fa50bf57b35acf026 : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/aa934d088a2f4aa1aa8752e886b712a0 : CANCELED\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bf0a3c71d5fd47feb86f6260e0c1ebcc : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac6bef3d67fe4f2186cfd8111d9d9cb3 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bd82468586243ae860dc783fc473bf8 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fc167de396e4454bff1368ee3aef0c6 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89a0d7fefa3d444bb227c3d7aa2af2d8 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e023c59d75eb40cc8239f1710a4eae10 : SUCCESS in 7m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/06671f2a7c2e4d33b419a4cf9f6d08f8 : CANCELED","accounts_in_message":[],"_revision_number":2},{"id":"8bde31602ae259dd8bd42232b75e96748f24f5b6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-27 05:12:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"03d837879c388e40d59aec0fabb83c51535762de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 06:25:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da4a3fd3616d40708ed9db04a82284a2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e121c41dce3446a3a538a02720896719 : SUCCESS in 33m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6740d00c9fae42ecba56b718eecc596f : FAILURE in 13m 04s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ace61d1ce0dd4cc590d0d8fff6cd33dd : SUCCESS in 43m 42s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a93fd37179e24ad8a086088955e91db9 : SUCCESS in 33m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/89f037a484b044b1bb71766cc99bcecb : SUCCESS in 42m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f557e9bd7cce4bde87ec70acdfc0ae4b : SUCCESS in 46m 30s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e1525434ec8a463ca540dfa875e3b32f : SUCCESS in 1h 07m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1a68fe608c1640b0a780252a0e34e202 : SUCCESS in 35m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fb526d80d44c44cb81d87fa1ffa9bcf9 : SUCCESS in 55m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a815dbd878c4f6cbadda8b634507638 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/144d3fceb5ea4fc3a275ebb6bbe7721f : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f250b7b6da224a5a83facd792659a2d0 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87832cbabc414ae1852b2b0d6b645a27 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/909d4bfe7c774182b53489add27e4d1d : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be792c87b62f4a97a2c1c8ac94dfda11 : SUCCESS in 6m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6d80090381a44015b7eeb672127805e7 : SUCCESS in 43m 58s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7e84ccc5c6e242458f2669cff37ac67f : SUCCESS in 4m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab4312565cc1715a2fca63fb607ddb9dab965b07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 06:26:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"86e04b00291e5d015f2a854b68e8a1547caa56c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 07:12:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6298a9c0ba441abb04044ffd164250b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5e69ba1293a240d0b0da7c6b56fb9798 : SUCCESS in 33m 45s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/edfba08078d749ed9dbbcb079011c6db : SUCCESS in 35m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b144d4cbee6d4987a178b515d16a552b : SUCCESS in 33m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9d400cfe857542c8a1ceb68c09ec39be : SUCCESS in 45m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7cd42daeab5540719e1f6787a6dfff49 : SUCCESS in 34m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c858ea968a414ad38025cd31cd6cbe74 : SUCCESS in 32m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/721863d0244d489eb09674df5850e668 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a01693b5b4c741b095040472bc65953d : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdba680763c644eb8f96f288471d0852 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b78424df9524ef9829314731c3fd23d : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82a864b5ea774ae6a90e4cf8a811bb40 : SUCCESS in 6m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e845644ca3e64eb799c30c90bb43e7ca : SUCCESS in 30m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d6cfd8d7f30a4515e27a9290e11fd6cda3c430c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 07:12:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f9d348bb82c6ce2f4aeeba228453bb21db598096","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 07:13:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4e08f6e3b4c448485db36a124237a59\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f172e9745894b889e164ed47434e6ce : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"48826ffa96b0ed358ea814657d31fb5ba28966ee","revisions":{"77dc28fa705ee4abfd473105063ec94e8aedcd42":{"kind":"REWORK","_number":1,"created":"2023-05-26 23:59:36.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/60/884560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/884560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/884560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/884560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/884560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/884560/1"}}},"commit":{"parents":[{"commit":"b9bd8e9bade38c4d6af72b482ce85b3e7a8cba6e","subject":"Add cmd tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9bd8e9bade38c4d6af72b482ce85b3e7a8cba6e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-05-26 23:56:20.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-05-26 23:59:34.000000000","tz":0},"subject":"Fix create zone export RBAC","message":"Fix create zone export RBAC\n\nPreviously the zone export create API relied on the zone lookup to fail if the\nproject did not have permission to export the zone. Because of this the RBAC\ntest was simply if the project_id was valid.\nWith shared zones, administrators may not want to allow exports of a zone that\nis shared with a project. This patch corrects the RBAC to check for zone\nownership before creating the zone export. This way a policy can be created to\nallow zone exports for projects that have a zone share.\n\nChange-Id: I2ef28eeb285462dc12b13de80cdab06b5cfb929b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77dc28fa705ee4abfd473105063ec94e8aedcd42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77dc28fa705ee4abfd473105063ec94e8aedcd42"}]},"branch":"refs/heads/master"},"48826ffa96b0ed358ea814657d31fb5ba28966ee":{"kind":"REWORK","_number":2,"created":"2023-05-27 00:11:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/60/884560/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/884560/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/884560/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/884560/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/884560/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/884560/2"}}},"commit":{"parents":[{"commit":"b9bd8e9bade38c4d6af72b482ce85b3e7a8cba6e","subject":"Add cmd tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9bd8e9bade38c4d6af72b482ce85b3e7a8cba6e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-05-26 23:56:20.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-05-27 00:11:07.000000000","tz":0},"subject":"Fix create zone export RBAC","message":"Fix create zone export RBAC\n\nPreviously the zone export create API relied on the zone lookup to fail if the\nproject did not have permission to export the zone. Because of this the RBAC\ntest was simply if the project_id was valid.\nWith shared zones, administrators may not want to allow exports of a zone that\nis shared with a project. This patch corrects the RBAC to check for zone\nownership before creating the zone export. This way a policy can be created to\nallow zone exports for projects that have a zone share.\n\nChange-Id: I2ef28eeb285462dc12b13de80cdab06b5cfb929b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/48826ffa96b0ed358ea814657d31fb5ba28966ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/48826ffa96b0ed358ea814657d31fb5ba28966ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
