)]}'
{"id":"openstack%2Fdesignate~884593","triplet_id":"openstack%2Fdesignate~master~Ia12b530730921db5235fd35f4cabf80b916c216f","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-05-30 22:31:48.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-05-31 00:28:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia12b530730921db5235fd35f4cabf80b916c216f","subject":"Fix bug in update_zone with recordsets","status":"MERGED","created":"2023-05-30 05:35:34.000000000","updated":"2023-05-31 00:30:25.000000000","submitted":"2023-05-31 00:28:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884593","meta_rev_id":"80a1f548ad4425fcb3a2727d38c6bcfae087bfa7","_number":884593,"virtual_id_number":884593,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-31 00:28:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-30 22:31:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-30 22:31:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-30 07:34:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-30 20:47:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"bd5dc4e6b6d2d327c53430d5c8166c02d107fbaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-30 05:35:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb3dc707097e3b12b99bc9126de742f90ae557f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-30 05:59:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3af8c37f5f8696bf681a70941114221c7a0920ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-30 06:27:12.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"6884a1d1508fabf6e71bd8993c383071ac0f45e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-30 06:31:20.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"c29bb8df33a281f7b85269e4bafd126dc204ade0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-30 06:31:29.000000000","message":"Topic bug/update_recordset removed","accounts_in_message":[],"_revision_number":4},{"id":"a0144696f6ed356ac26c66b9243f4c094be02bde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 07:34:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/687b9cfb6ec04cf28f2bb288b56ecae0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f9c800e17d444d71a3bd9747575dfced : SUCCESS in 42m 24s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0e12a75f28d441dc9d24f9873eaa2275 : SUCCESS in 59m 10s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/176f66c4d764482f8913f347681de3ac : SUCCESS in 57m 13s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/738d065917d146b9a65cecac65434867 : SUCCESS in 50m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/14adc90193f34d19b8bf459b12520c36 : SUCCESS in 40m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7330d94f98a848878756a35f1a348def : SUCCESS in 1h 00m 19s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c61c083194e84717bcad1657db17914f : SUCCESS in 52m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/90b29543c1e84c90b2c576a617bc8b5c : SUCCESS in 41m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3e5fd09a5c8045839a5597d25bd6c1ee : SUCCESS in 43m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f16f2093fd544258c6a0dbb3654dcc8 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34e2afeee4264e5a862c75256e536236 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d155909d9d946f6b9e137f9337fdea8 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/410e1baf3cc74e51bdc222f89c2fa7a0 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ccf8f7644ece4ea3a390e9dfbbda389f : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf091c25868849ee90c15aeaaadfd3c6 : SUCCESS in 6m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/01034a491c9545c9840f1f0e5e3d095c : SUCCESS in 31m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/845be7c8e85e473797744df382a6e9e9 : SUCCESS in 3m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"98c9be51a490fbbd8fa3bf201f08f66f83af57b5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-30 20:47:39.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2495c41b2917a8510fec26558d10aae34605e9ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-30 20:53:47.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"c96caa25bb408c554d0012becaf9de516826780b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-30 20:54:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b399dc1086557f3588899d9269f773591624d81a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-30 20:56:41.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"e1acf1e053479cda3acba92895edfeb628835ab2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-30 21:52:21.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"8869edbb2b4bc1d28ba521b65a36f706de761f15","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-30 22:31:48.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f39d84508d0ce01dee512c14bb02f0cc02412d67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 23:06:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/991063e31dcd444c8176bc19fed0aa04\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/147a9993627b40c98398cf87aaaef7da : SUCCESS in 41m 07s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/78fee91a8b9e43efb10d62c142393577 : SUCCESS in 52m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2bd983a56a2480aab2595fd6ad603b2 : SUCCESS in 49m 38s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b745797d8fac483eb5cf6028751a67a0 : SUCCESS in 43m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4aac509bfbe540cc949ad72da1648884 : SUCCESS in 49m 25s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4dfd98bf51e64ddd9c60eaac6dbad6de : SUCCESS in 1h 03m 34s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9516b6222818470b9e7b41ddbd9bfb53 : SUCCESS in 1h 09m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/41376df7a49c4fa48f575dd50f92ae31 : SUCCESS in 52m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ad64679a349d42c5817bb112068cea39 : SUCCESS in 39m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/644216af04854d35b6b2f24be9831cb3 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/490a48b9ae384034998ce2ee1759bf57 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/957a0218196e4fccb61eaa36fd902c72 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f209b7af83f41f79f055acdc4c641e6 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/884ab789e1d34bf884112ae90f746e31 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/254a35430a3b4f2f98bb55b2623e917c : SUCCESS in 7m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c911af27b9264672b374aa7e81b71ba3 : SUCCESS in 40m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8f45eb8ba380423c9e6729066320d914 : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3ec2cfad0a4d364b4d20efa5ee867a63276ee6fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 23:14:38.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"7ea8f84fa87e1dab3edfd5015af574ea8272f6d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 00:28:38.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b961028cb9ae4480932ce931edec9de4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/98a932b154674bb7882c70d141ee2edc : SUCCESS in 40m 37s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b1ac58d127d44f4a811d00af13653721 : SUCCESS in 37m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b7b2b129e8cf4b5aa6e09525d3eac87c : SUCCESS in 57m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/37274371f1474e729c135f9ca98c781b : SUCCESS in 47m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c97f2b67e66849998b5a047ed302770d : SUCCESS in 33m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e41dc6ceda8b46d482f66beb1318858c : SUCCESS in 33m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fab2529c231649d5b5a73e7397a091cc : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d25b7277168f47cb8919ab71df313254 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0dc2d45e1a4e4a5cb29839cc3ef6e83b : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7b456d3f6a842d994f95220ab4dd826 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ef86bab838a41d9b70e451be0e98e0d : SUCCESS in 6m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e117b99b6fb942ebbec097c32d1d0b3f : SUCCESS in 39m 34s","accounts_in_message":[],"_revision_number":7},{"id":"5aa35c5352c6f0759f2ba643054bd02dff539d2f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 00:28:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"80a1f548ad4425fcb3a2727d38c6bcfae087bfa7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 00:30:25.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af6669ff32d84e3bb8ceb828c3977b00\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7860afa197fd4384a28f7480880aa34b : SUCCESS in 41s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"b45a7b5167fcd5798da35c3bb3ef5aaffc66828f","revisions":{"b827db1633e76c7180ec11bd221127175e48029d":{"kind":"REWORK","_number":1,"created":"2023-05-30 05:35:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/93/884593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/884593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/884593/1"}}},"commit":{"parents":[{"commit":"c032a82e71a2b2c087c251ce3f80f61bef8f379a","subject":"Increase minimum acceptable test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c032a82e71a2b2c087c251ce3f80f61bef8f379a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 05:33:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 05:33:51.000000000","tz":-420},"subject":"Fix bug in update_zone","message":"Fix bug in update_zone\n\nWhen trying to write unit tests I found that the previous code path\ndid not work and would throw an sql error. This fixes up the code\npath and adds some basic test coverage.\n\nChange-Id: Ia12b530730921db5235fd35f4cabf80b916c216f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b827db1633e76c7180ec11bd221127175e48029d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b827db1633e76c7180ec11bd221127175e48029d"}]},"branch":"refs/heads/master"},"ab86e22ce3c0083034d861ae2034ac467b9c0a1d":{"kind":"REWORK","_number":2,"created":"2023-05-30 05:59:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/93/884593/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/884593/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/884593/2"}}},"commit":{"parents":[{"commit":"c032a82e71a2b2c087c251ce3f80f61bef8f379a","subject":"Increase minimum acceptable test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c032a82e71a2b2c087c251ce3f80f61bef8f379a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 05:33:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 05:59:36.000000000","tz":-420},"subject":"Fix bug in update_zone","message":"Fix bug in update_zone\n\nWhen trying to write unit tests I found that the previous code path\ndid not work and would throw an sql error. This fixes up the code\npath and adds some basic test coverage.\n\nChange-Id: Ia12b530730921db5235fd35f4cabf80b916c216f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab86e22ce3c0083034d861ae2034ac467b9c0a1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab86e22ce3c0083034d861ae2034ac467b9c0a1d"}]},"branch":"refs/heads/master"},"3140b991ea625d7340acc42b091434ffbfcff0e3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-05-30 06:27:12.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/93/884593/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/884593/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/884593/3"}}},"commit":{"parents":[{"commit":"74ace85d32202089da3222f5e67a86f2178008be","subject":"Increase minimum acceptable test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/74ace85d32202089da3222f5e67a86f2178008be"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 05:33:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 06:27:12.000000000","tz":0},"subject":"Fix bug in update_zone","message":"Fix bug in update_zone\n\nWhen trying to write unit tests I found that the previous code path\ndid not work and would throw an sql error. This fixes up the code\npath and adds some basic test coverage.\n\nChange-Id: Ia12b530730921db5235fd35f4cabf80b916c216f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3140b991ea625d7340acc42b091434ffbfcff0e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3140b991ea625d7340acc42b091434ffbfcff0e3"}]},"branch":"refs/heads/master","description":"Rebase"},"e6bffe76517a2f9dc48faf3bae29eb7e15da759a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-05-30 06:31:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/93/884593/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/884593/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/884593/4"}}},"commit":{"parents":[{"commit":"74ace85d32202089da3222f5e67a86f2178008be","subject":"Increase minimum acceptable test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/74ace85d32202089da3222f5e67a86f2178008be"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 05:33:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 06:31:20.000000000","tz":0},"subject":"Fix bug in update_zone with recordsets","message":"Fix bug in update_zone with recordsets\n\nWhen trying to write unit tests I found that when updating a zone\nwith recordsets there are scenarios where it would throw an sql\nerror. \n\u003e sqlalchemy.exc.CompileError: Unconsumed column names: zone_name\n\nThis fixes up the code path and adds some basic test coverage.\n\nChange-Id: Ia12b530730921db5235fd35f4cabf80b916c216f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e6bffe76517a2f9dc48faf3bae29eb7e15da759a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e6bffe76517a2f9dc48faf3bae29eb7e15da759a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c953f8e36c64132d44361130c94a839c7da35d15":{"kind":"REWORK","_number":5,"created":"2023-05-30 20:53:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/93/884593/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/884593/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/884593/5"}}},"commit":{"parents":[{"commit":"74ace85d32202089da3222f5e67a86f2178008be","subject":"Increase minimum acceptable test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/74ace85d32202089da3222f5e67a86f2178008be"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 05:33:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 20:53:42.000000000","tz":0},"subject":"Fix bug in update_zone with recordsets","message":"Fix bug in update_zone with recordsets\n\nWhen trying to write unit tests I found that when updating a zone\nwith recordsets there are scenarios where it would throw an sql\nerror. \n\u003e sqlalchemy.exc.CompileError: Unconsumed column names: zone_name\n\nThis fixes up the code path and adds some basic test coverage.\n\nChange-Id: Ia12b530730921db5235fd35f4cabf80b916c216f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c953f8e36c64132d44361130c94a839c7da35d15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c953f8e36c64132d44361130c94a839c7da35d15"}]},"branch":"refs/heads/master"},"9b157dcc6de6472f259fe28117c8fedaaa16162d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-05-30 20:56:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/93/884593/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/884593/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/884593/6"}}},"commit":{"parents":[{"commit":"1608ff347876860e9c780d5bacdaf8743ebcb940","subject":"Increase minimum acceptable test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1608ff347876860e9c780d5bacdaf8743ebcb940"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 05:33:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 20:56:41.000000000","tz":0},"subject":"Fix bug in update_zone with recordsets","message":"Fix bug in update_zone with recordsets\n\nWhen trying to write unit tests I found that when updating a zone\nwith recordsets there are scenarios where it would throw an sql\nerror. \n\u003e sqlalchemy.exc.CompileError: Unconsumed column names: zone_name\n\nThis fixes up the code path and adds some basic test coverage.\n\nChange-Id: Ia12b530730921db5235fd35f4cabf80b916c216f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9b157dcc6de6472f259fe28117c8fedaaa16162d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9b157dcc6de6472f259fe28117c8fedaaa16162d"}]},"branch":"refs/heads/master","description":"Rebase"},"b45a7b5167fcd5798da35c3bb3ef5aaffc66828f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-05-30 21:52:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/93/884593/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/884593/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/884593/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/884593/7"}}},"commit":{"parents":[{"commit":"a53c685adfd8bf1db57443d1fa2327a47cc339e1","subject":"Increase minimum acceptable test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a53c685adfd8bf1db57443d1fa2327a47cc339e1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 05:33:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-05-30 21:52:21.000000000","tz":0},"subject":"Fix bug in update_zone with recordsets","message":"Fix bug in update_zone with recordsets\n\nWhen trying to write unit tests I found that when updating a zone\nwith recordsets there are scenarios where it would throw an sql\nerror. \n\u003e sqlalchemy.exc.CompileError: Unconsumed column names: zone_name\n\nThis fixes up the code path and adds some basic test coverage.\n\nChange-Id: Ia12b530730921db5235fd35f4cabf80b916c216f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b45a7b5167fcd5798da35c3bb3ef5aaffc66828f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b45a7b5167fcd5798da35c3bb3ef5aaffc66828f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
