)]}'
{"id":"openstack%2Fdesignate~884606","triplet_id":"openstack%2Fdesignate~master~I267ecfe3288ae5ad7a7b5a5fdca0ab5f5e56b082","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-06-05 17:47:13.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I267ecfe3288ae5ad7a7b5a5fdca0ab5f5e56b082","subject":"Removing old archive code","status":"MERGED","created":"2023-06-05 04:55:59.000000000","updated":"2023-06-05 18:45:11.000000000","submitted":"2023-06-05 18:44:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884606","meta_rev_id":"e9eb8f5621d0fe66db7a72881f46f043c26b5c3c","_number":884606,"virtual_id_number":884606,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-05 18:44:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-06-05 16:21:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-05 17:47:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-05 06:19:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-05 16:21:34.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"21e96566412b8ea4f057476ee0e847b24181d765","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-06-05 04:55:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f85bff564c6dbc163051445ea6e059c2a2d4a04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-06-05 05:03:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"56c51f254340785cb49e8ce19b28bad1105aa457","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 06:19:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29d35e83c5a843ffa674191b007fa526\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b399d833f68844fc88d170cc6c450f88 : SUCCESS in 57m 55s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/430ba2383ae644f584a1b3245985b5df : SUCCESS in 1h 01m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/86f554c4ffef416a87a6e619a7bea62f : SUCCESS in 37m 45s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d267d70fa9064cb08012c11202ab3951 : FAILURE in 37m 17s (non-voting)\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/acad7f194b2e4c5b84a71f4d3cdc8d45 : SUCCESS in 34m 48s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/809fad54898b4eadb1dddb0955d2d8cf : SUCCESS in 1h 15m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c7de80118a284fc59e6e81d9f090db87 : SUCCESS in 1h 07m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/14473b72650547a0a2c6946a6d8ddc8d : SUCCESS in 35m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2352bc0706f54a3baef8c0a542bfabcb : SUCCESS in 54m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04c0fc62cdac4a25bc41d19e38a39608 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8db5fa708fa041f7afdf1956e748d97b : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c1c4256fa2349ffb01c155b0049af53 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/460384e0cd6443eaafcb4926e3583014 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a33b64e08c74054a48b5c4d592015c2 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d233b884cd43f482eac17b8851e68c : SUCCESS in 6m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e218a5eb7f0e4531acde502e22efe713 : SUCCESS in 1m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/405c378e9d4645539b406068227c5ece : SUCCESS in 30m 04s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1e645fc157d840e1bdf6728c6557db4e : SUCCESS in 3m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"77cee72444dd9925d876e3e0b9c990a6fb7c0e70","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-06-05 16:21:34.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"582023d12487d3908a23030cca23509c8d0247ee","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-06-05 17:47:13.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"895da0bc5f5847069520bd53e2937e8fce9ca945","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 17:47:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6e468eb55e8ead1acc37f86d6ed4c16f98f88671","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 18:44:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/296c29787c8b42f28e6bde631391343f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a2ee4357ad8547eab9e0576889aae878 : SUCCESS in 47m 31s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7e10011cc6f04721a4b8e13ee0f6ad04 : SUCCESS in 36m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/34095d5c50684451a945898a83f142b4 : SUCCESS in 34m 21s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b371f954813b4aae9ee8e1e27161b644 : SUCCESS in 55m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4ffb1bc6478d4f6d87cdcd676f9e401a : SUCCESS in 43m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4ef60f877aa04a9aa6465732714c12c7 : SUCCESS in 43m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d23e8796b34b45369e1454b161992048 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7255ed36344c4048973fa8c4a82911fc : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d5762402b0e4388b1f509ad790262f4 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d102fa0be384b5db0ec9db1f24bbbb4 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90502cd3ac024812ba903a04ddea1aa2 : SUCCESS in 8m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/335df835c9bc43198880692b42eb6976 : SUCCESS in 1m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5a1a6ee039f74f14812a0f4cf9280609 : SUCCESS in 41m 07s","accounts_in_message":[],"_revision_number":2},{"id":"b4d23744556f9e7fde0902196784f9f5d7084aab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 18:44:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e9eb8f5621d0fe66db7a72881f46f043c26b5c3c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 18:45:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd7857663d01458086b18c537567cda4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/334edc9f78f84c6aba48ed5e05aa55a1 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bf45ec737bbd48f62aef672f3580712faeff9be6","revisions":{"9308d728dac620f602fbab16cd9fd957ff148a78":{"kind":"REWORK","_number":1,"created":"2023-06-05 04:55:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/06/884606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/884606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/884606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/884606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/884606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/884606/1"}}},"commit":{"parents":[{"commit":"aacea51715035690958afa79c5f69b4fea025994","subject":"Re-enable test jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aacea51715035690958afa79c5f69b4fea025994"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-06-05 04:55:44.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-06-05 04:55:44.000000000","tz":-420},"subject":"Removing old archive code","message":"Removing old archive code\n\nChange-Id: I267ecfe3288ae5ad7a7b5a5fdca0ab5f5e56b082\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9308d728dac620f602fbab16cd9fd957ff148a78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9308d728dac620f602fbab16cd9fd957ff148a78"}]},"branch":"refs/heads/master"},"bf45ec737bbd48f62aef672f3580712faeff9be6":{"kind":"REWORK","_number":2,"created":"2023-06-05 05:03:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/06/884606/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/884606/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/884606/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/884606/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/884606/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/884606/2"}}},"commit":{"parents":[{"commit":"aacea51715035690958afa79c5f69b4fea025994","subject":"Re-enable test jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aacea51715035690958afa79c5f69b4fea025994"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-06-05 04:55:44.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-06-05 05:03:37.000000000","tz":-420},"subject":"Removing old archive code","message":"Removing old archive code\n\nChange-Id: I267ecfe3288ae5ad7a7b5a5fdca0ab5f5e56b082\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bf45ec737bbd48f62aef672f3580712faeff9be6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bf45ec737bbd48f62aef672f3580712faeff9be6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
