)]}'
{"id":"openstack%2Fdesignate~884609","triplet_id":"openstack%2Fdesignate~master~If67bd7121b98fddce77b3ae116dee48bd6464c63","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-06-05 20:47:04.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"If67bd7121b98fddce77b3ae116dee48bd6464c63","subject":"Add basic managed records coverage","status":"MERGED","created":"2023-06-05 06:04:48.000000000","updated":"2023-06-05 22:45:34.000000000","submitted":"2023-06-05 22:44:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884609","meta_rev_id":"9248a60efa2b09bf76a32e4c28aac3c23de9e3ed","_number":884609,"virtual_id_number":884609,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-05 22:44:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-06-05 16:24:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-05 18:41:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-05 07:21:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-05 16:24:06.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"1001e591db5fc076d088e03da10dcc0ed8df9fc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-06-05 06:04:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ab244cb25faebcc5d04d33fdbe59e3897784687","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-06-05 06:05:05.000000000","message":"Topic central_tests removed","accounts_in_message":[],"_revision_number":1},{"id":"0d97bc0d95ea8dcfe51f0a170c3a5a26a4cc0bf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 07:21:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c443397075e44c9ea262a27c29e2faea\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/92dc2c247d684fc882115dab2568287b : SUCCESS in 40m 21s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b4d020c627a34b20a2280f87f495ff95 : SUCCESS in 58m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/921146a0cd7c464dbf05774106161950 : SUCCESS in 47m 59s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/afac7857424e40609f7656b972859915 : FAILURE in 40m 46s (non-voting)\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3eb4eb91be0a421ea8aa318f3dc7fe88 : SUCCESS in 34m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b828ef409bba4172a2f5d1db994a627e : SUCCESS in 1h 02m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8db29d0b8085405da0c6790ce602524e : SUCCESS in 52m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/44771ed31fe44c03ada2c00e5d376243 : SUCCESS in 47m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c2ffcbcd45c442429d2564a0e5fb3dd2 : SUCCESS in 37m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d9379941fec4e75a651c9b76cd16a90 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2ee407ecbcb4b62a871a269df774331 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/218d74c5448c4c159eb8ab702478cc03 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7bc84807d8d141c889050f7aa06de6f7 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00ed46eb26114e55b8d5dee07e8e641c : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f51cf2bc1b34597925dfad9515c2912 : SUCCESS in 6m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/99cf8f0a76ce48bebe98ce53cb9950ec : SUCCESS in 31m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/976b7225374a42c28e210a471ccec0fb : SUCCESS in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"decd559ff226f445ea1053a665a3f810d2de8e85","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-06-05 16:24:06.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7dac04b90a4aeed43da462d3bb3a3b326901eb6f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-06-05 18:41:38.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bfaafb20b052a0bba16dbe5b7c420128dbefaa69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 18:42:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"177357e6f23d8ca39fc758ca6fd38681859d85bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 19:45:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9308292902784b10b85fc248b31747f2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7d53ab01706e4d74b3bb447b592c745e : CANCELED\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/66b182e19a3a494a9dd1210f293d4738 : SUCCESS in 56m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/650b37df79f5440892436792852055c9 : SUCCESS in 42m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/121d5eab1a8f43e49d20f85fa7997ed6 : FAILURE in 1h 02m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4febafa5635d4f21877e00839012748b : SUCCESS in 41m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f2f26803c2e54f89b4b86aa0c21e9637 : SUCCESS in 42m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9841d55af5749fe8afda381b6b1d560 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6de2e1eb8734c72a2023e9a022cd355 : SUCCESS in 10m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40cca02a908d47569c33437d64076c85 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31926cbba7994dab86e19bb9c362b898 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f295c2597a23473eb3b67d5f2806085f : SUCCESS in 8m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a29fae0500944bcfaad6b673cb9cd966 : SUCCESS in 46m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c5584d69d77a32a29fc141f4227add5bd6fbbbb4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-06-05 20:47:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"458b2c2760770c482bbfebf42ab8642a9d24bdbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 21:46:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f158e275727641839587896460dfeb10\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1b0c245bd405403381576dd99bdf8cd2 : SUCCESS in 38m 13s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/567ff4d61b10435085fb85584e572dbd : SUCCESS in 46m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c55b73c8fe7249508090630f0dc5cd04 : SUCCESS in 48m 02s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d519144ca2c240c18aff8f9682191ad0 : SUCCESS in 36m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2fd59e22139f44adab199c81a5514f1e : SUCCESS in 42m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/04d967b964f74478a102266fa934fc53 : SUCCESS in 50m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3773e02c31b949f4955dd3c8939ee82c : SUCCESS in 57m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/64f0391dc70f4d6cbfed2a7d315365b3 : SUCCESS in 38m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6a5cf849dee441519a99fd7e7f8413b9 : SUCCESS in 46m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb5b0337bbc942d88cd4d6b28c6774de : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fdf4081472c4846b50a6f70396fdd0e : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e19383ead8434028a5f3e8a1c0b9a099 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19550d87679a4863b8205363ba8e64c5 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f55e1778de9e4bce8b2e8bd258a7fc63 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbe423fb28964c449dfc376f7def098a : SUCCESS in 10m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ebb05cc46b074717a6b2ef9cb1bd67b3 : SUCCESS in 30m 48s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2b00133720504716a320e2efc0c11e8a : SUCCESS in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3767f065f47e889e84a6c9ada3202ecde30748a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 21:46:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7421184d88a5e436b46b7c713c7f76f2bd0d970","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 22:44:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30f74c079b334b598472a99015374645\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3f71a4ec8ab24b45a541c853beaad40a : SUCCESS in 36m 47s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/67184f62f871401ea5e4dff568c34851 : SUCCESS in 39m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2a3b104924c44b6db070138b2c8cacf4 : SUCCESS in 33m 03s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/61b1e330fcc640ae8036ac0199dfb2c7 : SUCCESS in 48m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b5d3f3d5e2a94597b6378ac1c205bb70 : SUCCESS in 34m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c853505899d645b387da7fe740ea15fc : SUCCESS in 52m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b436b4257ede40e493b6da621a839e20 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/101beb2d9de944f189223b96fdda5f4c : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e26dd9af2b334b308bc00350a17543ee : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87f461dd89cf4564a9aba5398cb8dc12 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4f49f4335e44e88b69a7774155bc853 : SUCCESS in 6m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1909ad5a8ab74dc5809f1cffc2e55a6b : SUCCESS in 29m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ceae20a1e585a0da2cb0ebf2ed0963c582a7622f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 22:44:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9248a60efa2b09bf76a32e4c28aac3c23de9e3ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 22:45:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7101dd89b264c0e92df573786e5d9df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9c0c27f635744afa7ac25f734c9920c : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2bbd67e547f0d02c47386a09daaff7756be8e5c","revisions":{"a2bbd67e547f0d02c47386a09daaff7756be8e5c":{"kind":"REWORK","_number":1,"created":"2023-06-05 06:04:48.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/09/884609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/884609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/884609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/884609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/884609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/884609/1"}}},"commit":{"parents":[{"commit":"aacea51715035690958afa79c5f69b4fea025994","subject":"Re-enable test jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aacea51715035690958afa79c5f69b4fea025994"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-06-05 06:04:41.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-06-05 06:04:41.000000000","tz":-420},"subject":"Add basic managed records coverage","message":"Add basic managed records coverage\n\nChange-Id: If67bd7121b98fddce77b3ae116dee48bd6464c63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a2bbd67e547f0d02c47386a09daaff7756be8e5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a2bbd67e547f0d02c47386a09daaff7756be8e5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
