)]}'
{"id":"openstack%2Fdesignate~884953","triplet_id":"openstack%2Fdesignate~stable%2F2023.1~Ia3ede57ac2249a1bcd49512aa36452b2b9ffd827","project":"openstack/designate","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-06-01 00:31:50.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-05-31 23:13:20.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ia3ede57ac2249a1bcd49512aa36452b2b9ffd827","subject":"Fix zone list/show for shared zones","status":"MERGED","created":"2023-05-31 22:44:12.000000000","updated":"2023-06-01 00:33:41.000000000","submitted":"2023-06-01 00:31:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884953","meta_rev_id":"7e31f5d062401e9f9a430554b3294e5ebc69749f","_number":884953,"virtual_id_number":884953,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-01 00:31:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-05-31 23:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-05-31 23:13:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-05-31 23:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-31 22:44:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-31 22:46:42.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"6338771f721d66815d085574a7f1b2aa0c973e3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-31 22:44:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"748c5277996f2f5c331fb13cfc403d849582bc13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 22:44:34.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/designate on branch stable/2023.1.  The error was:\n\n  expected a list for dictionary value @ data[\u0027gate\u0027][\u0027jobs\u0027]\n\nThe error appears in the following project-template stanza:\n\n  project-template:\n      name: designate-devstack-jobs\n      queue: designate\n      check:\n        jobs:\n  # Temporary disable to fix circular dependency\n  #        - designate-bind9\n          - designate-bind9-centos9stream-fips:\n              voting: false\n          - designate-bind9-centos-9-stream:\n              voting: false\n  #        - designate-bind9-scoped-tokens\n  #        - designate-pdns4\n  #        - designate-grenade-bind9\n  #        - designate-grenade-pdns4\n  #        - designate-ipv6-only-pdns4\n  #        - designate-ipv6-only-bind9\n      gate:\n        fail-fast: true\n        jobs:\n  #        - designate-bind9\n  #        - designate-bind9-scoped-tokens\n  #        - designate-pdns4\n  #        - designate-grenade-pdns4\n  #        - designate-ipv6-only-pdns4\n  #        - designate-ipv6-only-bind9\n\n  in \"openstack/designate/.zuul.yaml@stable/2023.1\", line 182, column 3","accounts_in_message":[],"_revision_number":1},{"id":"3357fd21643b5dbc34539f24af32d29f8cfbc006","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-31 22:46:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"54ba094c8ef85f5958926f976bdbb677b60f6d29","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-31 22:46:42.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"1727bdbae3eeb1553eba2e6f0d358df70b758681","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-31 22:47:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3966c3a1f19cfe8d8f83c180edc93a571b1b5e4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-31 23:08:17.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5d450785afb9f04c9a14086789a36aaa3a017ad0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-31 23:13:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1 Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"4487a790fa9640552a64fa3266f6c8b18d9a744d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 23:57:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ca8913ccc45419d887eb6db1317ba78\n\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/85efbb2d0a334573b59732d98c77b4d0 : FAILURE in 10m 31s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd0f307fb8ae40899c6325285726b9dc : FAILURE in 39m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52c823ebebf44ce6929a5bd3943e73c2 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89d59031711946aabcc309fd16973e9d : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b476030df3c045be85f23b701db46cc2 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4b0448b52ff4a95a85b11953f2d2571 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d56f085f8e854544a48bffd11b93a4f8 : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34438743c37542dd942bec9ebdc39d14 : SUCCESS in 4m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b3ef6e87088f47038406eb0cdfb08f5b : SUCCESS in 36m 11s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a997ca5a8d8c403b9a3cce15ee9618fa : SUCCESS in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b07d55db960a8fbabd9413f09ce710634150c58b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 23:57:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cf1bb8a00dedf94307671616acfba51721384d75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 00:31:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19a49ce350e44a06a7aa8dc54d4a58f0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7db10dfb525f4248947b798a5c680d9d : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/573476a01c53458a94ee061d81470afe : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e28fcf62790d4381a2eb2de4dee68bee : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e90896a0a24f4ebbb5b41e84f75129f3 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fce380d2ae84c2887dfcb95f3036032 : SUCCESS in 3m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/54405d6da15a434380a49a1a9c8b7df6 : SUCCESS in 33m 37s","accounts_in_message":[],"_revision_number":3},{"id":"d441ed170b4a3a26266606b47c932deb4fb25339","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 00:31:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7e31f5d062401e9f9a430554b3294e5ebc69749f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 00:33:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f95bf148eac849169631be3753f678a4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c758fb463b14cf9ad799090b869203a : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/beee88b1a5b543f5b170159929c05402 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"72de55dfea4af696e2b8b4d5b686e87c9113fe18","revisions":{"66e50ffbf60043fd04dbe368a09805e9cd845981":{"kind":"REWORK","_number":1,"created":"2023-05-31 22:44:12.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/53/884953/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/53/884953/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/884953/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/884953/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/884953/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/53/884953/1"}}},"commit":{"parents":[{"commit":"38b100a6fb775ab5f1b566294b2c2d72f889300f","subject":"Merge \"Use ids when removing ptr records\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/38b100a6fb775ab5f1b566294b2c2d72f889300f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-05-26 19:16:45.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-05-31 22:44:10.000000000","tz":0},"subject":"Fix zone list/show for shared zones","message":"Fix zone list/show for shared zones\n\nPreviously projects that had a zone shared with them were unable to view the\nshared zone. This patch corrects that issue.\n\nChange-Id: Ia3ede57ac2249a1bcd49512aa36452b2b9ffd827\n(cherry picked from commit 5b7ee8899dd041a351ed61a2a6378e75812bfd59)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/66e50ffbf60043fd04dbe368a09805e9cd845981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/66e50ffbf60043fd04dbe368a09805e9cd845981"}]},"branch":"refs/heads/stable/2023.1"},"855d7f9a6f0c196ae94f3010c768ac9addb8f0a0":{"kind":"REWORK","_number":2,"created":"2023-05-31 22:46:02.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/53/884953/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/53/884953/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/884953/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/884953/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/884953/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/53/884953/2"}}},"commit":{"parents":[{"commit":"38b100a6fb775ab5f1b566294b2c2d72f889300f","subject":"Merge \"Use ids when removing ptr records\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/38b100a6fb775ab5f1b566294b2c2d72f889300f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-05-26 19:16:45.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-05-31 22:45:56.000000000","tz":0},"subject":"Fix zone list/show for shared zones","message":"Fix zone list/show for shared zones\n\nPreviously projects that had a zone shared with them were unable to view the\nshared zone. This patch corrects that issue.\n\nChange-Id: Ia3ede57ac2249a1bcd49512aa36452b2b9ffd827\n(cherry picked from commit 5b7ee8899dd041a351ed61a2a6378e75812bfd59)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/855d7f9a6f0c196ae94f3010c768ac9addb8f0a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/855d7f9a6f0c196ae94f3010c768ac9addb8f0a0"}]},"branch":"refs/heads/stable/2023.1"},"72de55dfea4af696e2b8b4d5b686e87c9113fe18":{"kind":"REWORK","_number":3,"created":"2023-05-31 23:08:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/53/884953/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/53/884953/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/884953/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/884953/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/884953/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/53/884953/3"}}},"commit":{"parents":[{"commit":"38b100a6fb775ab5f1b566294b2c2d72f889300f","subject":"Merge \"Use ids when removing ptr records\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/38b100a6fb775ab5f1b566294b2c2d72f889300f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-05-26 19:16:45.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-05-31 23:08:10.000000000","tz":0},"subject":"Fix zone list/show for shared zones","message":"Fix zone list/show for shared zones\n\nPreviously projects that had a zone shared with them were unable to view the\nshared zone. This patch corrects that issue.\n\nChange-Id: Ia3ede57ac2249a1bcd49512aa36452b2b9ffd827\n(cherry picked from commit 5b7ee8899dd041a351ed61a2a6378e75812bfd59)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72de55dfea4af696e2b8b4d5b686e87c9113fe18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72de55dfea4af696e2b8b4d5b686e87c9113fe18"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
