)]}'
{"id":"openstack%2Fdesignate~885164","triplet_id":"openstack%2Fdesignate~master~I1be68b27dc052cc273fcfdf53940b38a7661e73b","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-06-05 16:12:50.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-06-05 17:21:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1be68b27dc052cc273fcfdf53940b38a7661e73b","subject":"Fix zone list with system scoped tokens","status":"MERGED","created":"2023-06-02 21:40:21.000000000","updated":"2023-06-05 17:23:16.000000000","submitted":"2023-06-05 17:21:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"885164","meta_rev_id":"9da0ccff7a34ee2fe0a7bc3e34c72bbfe2ee4f9c","_number":885164,"virtual_id_number":885164,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-05 17:21:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2023-06-05 12:22:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2023-06-05 16:12:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-03 04:13:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2023-06-05 16:12:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-02 23:05:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-02 23:21:05.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2023-06-05 12:22:14.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"ca71e014f1d3606abb90a386d7338888f8cddd56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-06-02 21:40:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bcab3c306f1d84dc36fdf99232ac94865432a34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 23:05:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/64db665da51d4732883eccc0342c92b7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3c46faa20aa34c7984ce9ea9d7d17842 : SUCCESS in 48m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/04cbb1d9e4b440a08a21c1bc3c8e39a6 : SUCCESS in 1h 06m 33s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9298a4bf85c2430c87d94ffd8e6d965a : SUCCESS in 56m 34s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8cd21fb1eb5343be886b8faa2ad1a223 : SUCCESS in 55m 59s (non-voting)\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/06a5dffbf45d4416bdf16d22885f7784 : FAILURE in 30m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/dd32f5c94f50409e9e666b6bfe138a35 : SUCCESS in 1h 16m 58s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/28ce39867ffe4881b6a0d7ddc63dd2e5 : SUCCESS in 50m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e2009877d8fd4caf8b496e235816d561 : SUCCESS in 47m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/178346fc157f4b62ac88f7c75260ca18 : SUCCESS in 49m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/482059dee906451d87e643e5f0d472d6 : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872e711ca633455983a0a2543467fb55 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17dde39131804079b85990fcd7840b9b : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b41745609cd48e1933510a8697ab1ce : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b6d55ba56164f0fa9ad4eb7d7ca2ef8 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6848c8ad698542db883b32fb8c70489a : SUCCESS in 6m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e727d01f3e5442e78766dc61037548b9 : SUCCESS in 39m 00s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/86047f7184534142aaf77bca95c1a634 : SUCCESS in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f374f3a9b588a8e950864423ee57073bc971ce6f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-06-02 23:15:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b8953c6efecae7cdb81a1cbfd7f8c9d54b9b2e7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-06-02 23:21:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5f69ebf38daed2659c61bd34618e9b0f7fbcbfca","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-06-02 23:22:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3a574f0d01304fcd1b044a03ec8cd04d016eb7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-03 00:35:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/516a4d9b857f48879a6b1f34515644d8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/def186406257462bad524aeb7280a499 : SUCCESS in 30m 00s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/28696a8e0f434d5c9657f09a6348e674 : SUCCESS in 1h 04m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/70901a4f01b34cc38fc64449df05abf4 : SUCCESS in 56m 28s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/cba82162b0af4cd9af5298fc293e3f2f : SUCCESS in 55m 00s (non-voting)\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/008067c170c247a3966f2c4748f08422 : SUCCESS in 47m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/86fead5cbdb94258a3e19266291000e8 : SUCCESS in 1h 07m 54s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a3d1f3dd5e2b4c80bdc5be181685097b : SUCCESS in 1h 14m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a0a8012c55b8491a87dd9b16e8516683 : SUCCESS in 48m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bd823be18ea44f8b96f6040c2d1f73ab : SUCCESS in 49m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d0ebebb43d14b6cb7be1d5887fa4337 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d44c5cf2be94f53849cb6a3b27e0637 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d528a93ebc8c4e91844b4617f7e16c16 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9514be42cd8f4adeb80a7e64bdee8ecb : SUCCESS in 6m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cdde66fa80b84d7d882d867f5d4c8155 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ff15b73607b4035b5c5a08bcd74d2d1 : SUCCESS in 6m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b6cfaca6284944fc8618fa64bcfab2f6 : SUCCESS in 34m 07s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/346e0301e96a49909c6b324331fe4580 : SUCCESS in 5m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b48352d3e230fcabfef6a48c844feae94e302f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-06-03 04:12:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f00ee5c35c954d751056013c44d7411074aaa0d3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-06-03 04:13:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d2b4be0ab8a0f2155049a6fa0eaf84bfc9cdca5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-03 05:12:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8918b637e6d449c6b0b25fe1255d076f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/23cebe33177c40b386ec679b0bff66d6 : SUCCESS in 36m 31s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7fef634d7fb142098d50197aa1dd6cfc : SUCCESS in 46m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cbd3da704d14ba3b1934e40b6805750 : SUCCESS in 41m 09s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6f3e03a5ccf744c29139bd2db57f1ce0 : SUCCESS in 33m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dbaf84955cb04c14ad83932e685d95fc : SUCCESS in 39m 01s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8ef73eb9f34f43d9b2c375e4cda5f53f : SUCCESS in 46m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/67e75e1385fb4fc0852a659a7325ff90 : SUCCESS in 52m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1d1a1796956049298e711d1764f80e0a : SUCCESS in 39m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2658a3586f624a949eb197f9fb377d15 : SUCCESS in 34m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c1acb809dd441b58508e7bd8298b1da : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8149910684b14cf68276a84418506567 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38648349e1744a48a87e27ba8ae9e7b6 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bdebd6ea50041da8d5e7e59e39e0a45 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92f84a8327aa4d72bd935c6fcfe61e96 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df02d28023894855be81726b011a147e : SUCCESS in 8m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6a38a82883d34426872acb3d54869ed0 : SUCCESS in 30m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e1a4ba0bfa574e4496a8958191bc3f8e : SUCCESS in 4m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2fe122dada8bca46d277044a68439d95abb387d5","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2023-06-05 12:22:14.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f6ce0f1790cdf1517602afdd235a429b091d2ff3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-06-05 16:12:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8adc4b73df966bc6f020b56cd1a223092c11622f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 16:13:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b5af0b3937a5788f45f97e42d7582b345f4c3a55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 17:21:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2acc44dc5c354e63b9bc879baa0a3a52\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cbee963bf63c4bb1a1b2496dd3d45b22 : SUCCESS in 49m 06s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4e4f8da5a44f4757accb11a66bf5a07b : SUCCESS in 51m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/83cfb05dd4ba4d7bbfa688904ed3bbcc : SUCCESS in 48m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8333b54b7be640cb87f4c552fd1537e8 : SUCCESS in 56m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/be8ae9892cae4ca7a76cef50df7b6aa5 : SUCCESS in 1h 01m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/994fa5463fed41749d26f342e2c5f789 : SUCCESS in 54m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0aa546b63244b94a79a9766c1c68468 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61ddd6fcde7d4183835c39ed538f9494 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ecda09462c84578a6ee6b17aee27f3b : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a3d1a64360549eca253cec6442a87bc : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab0ebd133c3744ee9d2d3a0621e0b357 : SUCCESS in 7m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d0213111b3a74763b9b3ab3578d369dd : SUCCESS in 31m 30s","accounts_in_message":[],"_revision_number":2},{"id":"b00e6caed4644896bc198f4b1fe3c4e1ac3df50f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 17:21:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9da0ccff7a34ee2fe0a7bc3e34c72bbfe2ee4f9c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 17:23:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec4492eccd15464083d3cf9f388a0b08\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7f250a847f746f0901ad6c707192d48 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9c46364483b43b326c64896e1ed46067334feff1","revisions":{"43fc0802928a6dcac7181b0d5a33387291c8a740":{"kind":"REWORK","_number":1,"created":"2023-06-02 21:40:21.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/64/885164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/64/885164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/885164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/885164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/885164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/64/885164/1"}}},"commit":{"parents":[{"commit":"aacea51715035690958afa79c5f69b4fea025994","subject":"Re-enable test jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aacea51715035690958afa79c5f69b4fea025994"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-06-02 21:38:27.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-06-02 21:40:19.000000000","tz":0},"subject":"Fix zone list with system scoped tokens","message":"Fix zone list with system scoped tokens\n\nIf you used a system scoped token to list zones and the zone is not shared,\nyou could see more zones than expected.\nThis patch corrects that edge case.\n\nChange-Id: I1be68b27dc052cc273fcfdf53940b38a7661e73b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/43fc0802928a6dcac7181b0d5a33387291c8a740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/43fc0802928a6dcac7181b0d5a33387291c8a740"}]},"branch":"refs/heads/master"},"9c46364483b43b326c64896e1ed46067334feff1":{"kind":"REWORK","_number":2,"created":"2023-06-03 04:12:11.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/64/885164/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/64/885164/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/885164/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/885164/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/885164/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/64/885164/2"}}},"commit":{"parents":[{"commit":"aacea51715035690958afa79c5f69b4fea025994","subject":"Re-enable test jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aacea51715035690958afa79c5f69b4fea025994"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-06-02 21:38:27.000000000","tz":0},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-06-03 04:11:59.000000000","tz":-420},"subject":"Fix zone list with system scoped tokens","message":"Fix zone list with system scoped tokens\n\nIf you used a system scoped token to list zones and the zone is not shared,\nyou could see more zones than expected.\nThis patch corrects that edge case.\n\nChange-Id: I1be68b27dc052cc273fcfdf53940b38a7661e73b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9c46364483b43b326c64896e1ed46067334feff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9c46364483b43b326c64896e1ed46067334feff1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
