)]}'
{"id":"openstack%2Fdesignate~885594","triplet_id":"openstack%2Fdesignate~master~Ic9e593d661f2eb1db59a5b235521e0746489f9bb","project":"openstack/designate","branch":"master","topic":"catalog-zone","attention_set":{},"removed_from_attention_set":{"33634":{"account":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"last_update":"2024-02-14 20:16:54.000000000","reason":"Change was submitted"},"32755":{"account":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"last_update":"2024-01-31 08:29:58.000000000","reason":"\u003cGERRIT_ACCOUNT_32755\u003e replied on the change","reason_account":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}},"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2024-02-14 20:16:54.000000000","reason":"Change was submitted"},"5572":{"account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"last_update":"2024-02-14 20:16:54.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-12-11 08:51:10.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2024-02-14 20:16:54.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-02-14 20:16:54.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-02-14 18:59:24.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"31933":{"account":{"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},"last_update":"2024-02-14 20:16:54.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-01-22 14:40:11.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ic9e593d661f2eb1db59a5b235521e0746489f9bb","subject":"Add support for catalog zones","status":"MERGED","created":"2023-06-08 16:04:24.000000000","updated":"2024-02-14 20:18:02.000000000","submitted":"2024-02-14 20:16:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":140,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"885594-catalog-zone","meta_rev_id":"2370c98371bb478cc12ef3307b78a2658b3171ee","_number":885594,"virtual_id_number":885594,"owner":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-14 20:16:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2024-01-24 08:13:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-14 18:59:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-01-22 14:40:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-14 18:59:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"}],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}],"CC":[{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":35037,"name":"Nico Keller","email":"nico.keller@inovex.de","username":"nicokeller"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-08 16:10:21.000000000","updated_by":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2023-06-08 16:10:21.000000000","updated_by":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"reviewer":{"_account_id":35037,"name":"Nico Keller","email":"nico.keller@inovex.de","username":"nicokeller"},"state":"CC"},{"updated":"2023-06-08 16:10:21.000000000","updated_by":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-06-08 17:07:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-26 16:20:18.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-06-26 16:20:18.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2023-06-26 16:20:18.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2023-10-25 05:04:02.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2024-01-24 08:13:46.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2024-01-31 08:29:58.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2024-01-31 08:29:58.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},"state":"REVIEWER"},{"updated":"2024-01-31 09:33:22.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"afcf3a4fcec52e8ae27af78d2429703958fec136","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-06-08 16:04:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbc58a46097ff8c8aad9ebc588539fd6cc8f84fa","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-06-08 16:10:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a4c97e83e34c8083f2a5b7551b5906f9457e5630","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 17:07:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e47cc68944d40c2a8ce99018430635c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3234af8dda6644d88ccc65c4a0fd6030 : FAILURE in 17m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/296eb147f8124b2b9b9f260d6a914438 : FAILURE in 23m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/abd3dc5771d644b08a3f3d46df85562e : FAILURE in 21m 25s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7fd23663e65a40caafbe1ea01c322c70 : FAILURE in 16m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8342780a9db44381bf5643bf46ceb260 : FAILURE in 28m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ef820230d22c4c46a34a98236a5e21a7 : FAILURE in 54m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0eb7af6040004c08993359e4a8bd9cdc : FAILURE in 57m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2f93e43c89c94f5b842747c6b2d23d80 : FAILURE in 31m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3079c0e549454a07bea82cc6cc883389 : FAILURE in 34m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7abe6a10287044ffa2aba87e8f26bc75 : FAILURE in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e6fc4cb000741668161341795a8103e : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d02f8f85f84848e587d9bccdb3cc302d : FAILURE in 8m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e6464a1b2114e67960245a01c0671f0 : FAILURE in 6m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0829c3d6cda4e13a544094c5955c233 : FAILURE in 9m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2903c2cc60a14d0bbec1a3ab2af79618 : SUCCESS in 6m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b9de3ecfbb10420c9e507c924ddff33b : FAILURE in 28m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/fd83d9dc6ee8497f990c0bf968711723 : FAILURE in 7m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6169ccab5c435a8051606d57100b72a45a6804a","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-06-19 07:57:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39f3bbfb68f36774a1e55fc2514547ccdafe036b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-06-19 12:39:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7059017984ace4f6fd4d5f3de9d796062d9837de","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-06-19 12:42:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13b7fde5bd218abc4828dbdba31906f0df7c9597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 14:05:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e49ed4e9c15848fba2d49b6c15f02492\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a795b536ca0e48e6a246f413a6ce6b70 : SUCCESS in 40m 44s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d5c68b6cf57f4a1487c782d5c8dbf6a7 : SUCCESS in 1h 01m 07s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8dbada490b554d909d68047070a7adf5 : SUCCESS in 45m 56s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/071bb0c0408f4836b6b5e1f1dc68036a : SUCCESS in 43m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b02b87bd0d8a41a99b22b3900818ca2f : SUCCESS in 53m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9fb70b1251a242b5923fb11213393ccd : SUCCESS in 52m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/aead1ea4e4fe436d8a1d8264aa96f807 : SUCCESS in 1h 17m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7e7a552252a54b34916120cfcd1800e4 : SUCCESS in 34m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/432c517d4e5a4bb6b62ce870d1713529 : SUCCESS in 53m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce0e2963c44748ffa6dfaa6918119cc5 : FAILURE in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8235248e07b848c394538e6401ff42b7 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/933b0df235e74d8489744464e2f46e1d : FAILURE in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d96aed3564b44b6fa53feb9d4a5858b7 : FAILURE in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6bd3f90aa9e0410a928165ff2c7f2e53 : FAILURE in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a497a963bc714c2e8f47aa7bcaf58c34 : SUCCESS in 7m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b875ff7171c4462fb905efec30ce6e29 : SUCCESS in 32m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d9df2cd9aea14f26ba789e24233f0636 : FAILURE in 7m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e95009a478a68a6f899f56ea55bbf6e9ca01c455","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-06-20 12:45:05.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3a7827cd2d098a0a3fcb2de9d2ab4660887244b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 13:56:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2cb3694dd6e473286a436bd41303b7b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b715b018481b48c5b456f08f9dd6a692 : SUCCESS in 1h 06m 15s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/418301a2b53848a39c0cee67c38712a5 : SUCCESS in 42m 40s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/364686484a5a404e943526177ceee3c1 : SUCCESS in 39m 11s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/571eac895d144782a802257a7486b3a4 : SUCCESS in 49m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/db090ef5fcac4f1faa10e07e7102bf16 : SUCCESS in 52m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/06246631c0ec468f831c8518b6551c50 : SUCCESS in 1h 00m 15s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1f82703de86f4582be0ccba4a74c8d78 : SUCCESS in 1h 07m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7f00592d0464455bb2412493394aec51 : SUCCESS in 42m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0a0031a42715451d93a211eb106d7eab : SUCCESS in 52m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d2a22c1f92543fe9f39cb08f8c6c4c1 : FAILURE in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03aece51882a4da8bee4c1df4d54a9eb : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8617050717fc46b198f6cb892fedf7cf : FAILURE in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd3687504b3c462689930d22a2f10165 : FAILURE in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6fada8ca66840debb21c2fbbf59bde8 : FAILURE in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21ec6a26ec7848a69e63dd9e3a593c68 : SUCCESS in 7m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aa3dab373d584e659eb5d1e799c84ff4 : SUCCESS in 33m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/dee9e0e0fcfe41cd885ff561ebc969f1 : FAILURE in 5m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f7a68bfc3d10f9f95af94be1db2ffcce2ba30947","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-06-20 14:37:03.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ed6ba7ea492945d380d03e9ce4b0465c2516c59b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 16:11:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d781efab4e8a41978dc7491a70a334fb\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1e264912d7d5498e83f59b80f7ef5740 : SUCCESS in 59m 04s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1d296d5cd6ac41629611e562493b8136 : TIMED_OUT in 1h 14m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/91f6da8665004848bd42dfd7cc9fc7aa : SUCCESS in 1h 11m 44s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8802ca2f82cd48d18bf7f1ce1b9520fa : SUCCESS in 52m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8f707c40f648413d9aa0f615a10af026 : SUCCESS in 1h 09m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/84721a511df544d1834e4027d89d5337 : SUCCESS in 1h 10m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7559c105f00a4e04a1fcf39a5229df72 : SUCCESS in 1h 21m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1f133defdf6e4c3c8171c93a82c740da : SUCCESS in 1h 21m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/db2e8ec2f3c8479892ea08f485395163 : SUCCESS in 1h 01m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d27b1e701ef84ed2b13f4586c93ba0ac : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f8d7bcf6daa4aeb86323a15deb64dd1 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/097ece81038c4781877137499aac1e2f : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d945e1fd284c472da8d8249557a6ae56 : SUCCESS in 7m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1e0ad22a4734e78917921a256862862 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b065e122d2844e1b9a87895a6ea074a2 : SUCCESS in 10m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/110be6b707f84011927f14dd442dbbb3 : SUCCESS in 32m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/86ef647331aa40d4912b2a7138400df1 : SUCCESS in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d8b7d7ce7120c0e1a93e85c2b0d216b146f60ae6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-07-11 11:41:09.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"f7cf6df64b0883fec967b97a3c4f0fa0dd1b2e4a","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-07-11 11:45:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"59b1ba2bae18dca1cc93f14c94d10efecc3d55fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 14:11:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ad74ea3e72d4a999be29adc25c1341f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ec053fd6e9474236b9613875c975253f : SUCCESS in 40m 42s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1cc77237113b48f6b5031020987e8e02 : SUCCESS in 42m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a17ef93eebc64a45bce4819178872471 : SUCCESS in 37m 47s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/cf90570e8d3a4a8a91b241672aa014b7 : SUCCESS in 51m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fa4ae940218445f6b6936d5bed74d1f7 : SUCCESS in 51m 45s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/17409b046bdb4b89bb95b59beec770ae : FAILURE in 2h 21m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6fca592dc2d24f7ca26ce8e2a2b2fe4f : SUCCESS in 48m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0bdf2b97fb4f4bcdb3b80c545ad59a0b : SUCCESS in 41m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8c355739ae6c4840b01d8a7747a8570e : SUCCESS in 37m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee39f5a806b44b6f93666bcf7de0e079 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9afa94f7bb14cffb3c9429cc7be6cf7 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/580f6a4416774e56ba244a3530c5121d : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6cdfc0243c2402fba6c27b24a72200b : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/850c6c26a3b640bf913f3f1fcd91669e : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebcc2ea350a34648ba86970c2d41ede5 : SUCCESS in 8m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1021189734a54a139fb098e0447d910b : SUCCESS in 32m 16s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6869a51a4c914316a06019f463763f13 : SUCCESS in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e3dccb4bd40c99b2a88db701433f56d09e321c8b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-12 18:05:29.000000000","message":"Patch Set 5:\n\n(7 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fdc3763caaa15acecf1397ad61d2d83396fe82ef","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-18 09:54:13.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0a16149750f1a8084a1325c173087a4675e94cde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-07-18 15:04:00.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"bb5a71dfe5377fc325ed1fdc3e3a799f9cbe4dfe","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-07-18 15:12:30.000000000","message":"Patch Set 6:\n\n(8 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4b28c02c94fee83d88b362e093ce62e1ab149295","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-07-18 15:22:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5451194c578a9df7f1e5c4eb7ebfe9dc72480601","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 16:23:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5075761e1e746da8c39b0ca38ba19c0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7ba0acd61c0843a1a2ce0fb184e5b378 : SUCCESS in 48m 12s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0f5ec10aba634a41a0c340f236a9289f : SUCCESS in 45m 14s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/51985ec59002443a856c16cb31e11162 : SUCCESS in 39m 19s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b40914e37d804aa6b7b2136955f2cb8b : SUCCESS in 41m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b6f3374d36d5460ebf5eef022751b926 : SUCCESS in 45m 37s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f32b8219a6b0471f8aefcac949733539 : SUCCESS in 53m 05s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7d0b9ad6561f49f882e9bf9d5fbbe557 : SUCCESS in 52m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1acd5edb4a9a455c96f5a239c9b3277a : SUCCESS in 36m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bfe1d912c0844ae28aaa7337c14e45bc : SUCCESS in 53m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65701f9749a440e89d5f092432195cd7 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36ce91b165524eb1bcaaceecae7bb534 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76f2c2d7f80f4ec0be4fc7ad495e4fd6 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ca86c8399b946e89bdf47dff4564c96 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56e62f3a41ed404f896d704d87565901 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dda915a8a65427ebaa3d2aa29085dac : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/216a4daf74a24235baf7221f45104558 : SUCCESS in 5m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/90c8278722e3452f8fb8bb2252bded2a : SUCCESS in 43m 48s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/08438a7231614087af59133784aa5773 : SUCCESS in 6m 14s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a76350edeee8ab539d8bea2808a2deb2c4e25302","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-18 23:58:25.000000000","message":"Patch Set 7: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":7},{"id":"7523bba2f3a2c5b0c465b3acc7719c664c707076","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-07-19 15:05:20.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"dfdb15ec812c405310fba32a9894bab4df50c46a","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-07-19 15:09:12.000000000","message":"Patch Set 8:\n\n(11 comments)","accounts_in_message":[],"_revision_number":8},{"id":"978f5d217222702c7a6f7b7b663501e681475935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 16:32:46.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c2da764e90c46f989a4590979e2de54\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f28ea1aa609a419fbf0b5bb8c610ad29 : SUCCESS in 51m 34s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/316899a71b394af79975c904eb199444 : TIMED_OUT in 1h 14m 33s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/727eabca28244171b25c75993c104d59 : SUCCESS in 1h 04m 54s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/bbbb1c75916f4a3a9c3b5ea8568e00ef : SUCCESS in 59m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/eca1a6cbbc694e62bc79ce933374d4c5 : SUCCESS in 52m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/de86390ae44b4c88bbd241109a8df2f2 : SUCCESS in 1h 19m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/759addda51e548129bc1ee3ddff50c53 : SUCCESS in 1h 14m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bc3331b7d5c54533990dd7e68545f79e : SUCCESS in 50m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e0f76ade595841268b09f936d5a6de94 : SUCCESS in 56m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ec9283041594a21bddc33936d5ba074 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/122df5282d874c199371a34ba5b1d8fd : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbe905e51b454666a9ef23a0067bb61e : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08935288884543b28675e3c40d359213 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79736c4e174b49ab966da0f72cf8f563 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca352db5a4534d509465a0e3e5b66d69 : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9074d82dfeaf44cf987921fa5e3fcc07 : SUCCESS in 4m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e72e9b1e126f4b509e6ee295e22ba534 : SUCCESS in 33m 57s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a7fcfc3f806242658c9333833a7c5397 : SUCCESS in 6m 52s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7904f0c0591847d38f08bfee5432b756 : SUCCESS in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8da46e6d418b8ef7ced600e9ba023be8004cffce","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-19 23:30:16.000000000","message":"Patch Set 8: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":8},{"id":"9d487541c37e1d3f81fd63f11a14fc8db6fb6fd9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-19 23:31:29.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"8a08cf6e0d8caffe71cb07c96b28fcd9c8d4004b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-21 22:09:47.000000000","message":"Patch Set 8:\n\n(14 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f1d6972fcbf2057131ba1498b97627481c1cb893","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-08-08 14:34:34.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"e61c6b464c4726f96187567692881dae0d3caf88","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-08-08 14:37:29.000000000","message":"Patch Set 9:\n\n(18 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d5bf1cd9578548f66f4f72d0c364487eb77e43dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 15:41:56.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2bd9a54b9864f2ebdc99ecbeb7af95b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4bd011fc78584fdeba637d39c52df60b : SUCCESS in 50m 19s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/985a2fa7cb3346d58855eea2a0130af4 : SUCCESS in 48m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbed114677f24727a05bb1adc195b5eb : SUCCESS in 58m 53s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/95cc3fdc2bfe406189e51cf58fc44198 : SUCCESS in 41m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1a1c4659ec2340b882f09388145b3110 : SUCCESS in 57m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/feb80d6683074eb98985a1bcee3fa01b : SUCCESS in 59m 22s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b6908e8810bb4905a443d737b2b0bf4f : SUCCESS in 57m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2c300ff2309e4a42bf070c79bc1a4f16 : SUCCESS in 36m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9ba45e82f0394c91bb684e551fa55f01 : SUCCESS in 35m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/631ff7a9a2ca460bb88cf1a07c666b99 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03a990bc6b234c52bf7506a193c5065c : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28ee00a613934b73b93c3b74326cbe94 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7b9c3b31aaf45cb8b494e834b3a33f7 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7223e4e79334da88e3be16a97fbaacc : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/359fd3cce1c04a069ec821b16d052dc4 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3aac2b0a96f54e748103f76e46e12fbd : SUCCESS in 5m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4b9b7e7251584920b56c07951fdf336c : SUCCESS in 33m 06s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/23a27b7124c44b4887efda42c43ca9b7 : SUCCESS in 5m 48s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/69db586bdfb9423ab70fc5aa21309014 : SUCCESS in 39m 25s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3c74e5ce01f6468b852237a5a9d28db1 : SUCCESS in 4m 28s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"22d1d80752a7d6e332923ff5aef92dcca9fc3c9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-08 18:49:52.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"ffb0a9a958c641bce3e67ae094d8d198f0979d1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 19:56:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7350d99c88b4eca82e4f50841f72f66\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b32cda4f4c2746ccaf52310be7b63f35 : SUCCESS in 32m 20s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1e96c01137c64519aba22585a1b18d48 : SUCCESS in 1h 04m 06s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/070ff22507594006a569eefa07f24e23 : SUCCESS in 59m 27s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a0a2e3d254a84a849927b8c671948229 : SUCCESS in 46m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dc1a32bb9a004e3591382b4616d2e6e9 : SUCCESS in 48m 27s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8b92018651fa4a3d92ed017b33fb933e : SUCCESS in 57m 45s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/15f694daf6cd40069912c0f61cab4ab1 : SUCCESS in 1h 03m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/20b2318a46b249fcaf47e04c76b89775 : SUCCESS in 45m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9ce5c56bad9c4b2688a581fa662456d1 : SUCCESS in 40m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d391eee42a894749a722efcc5e470207 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc99c543d5f84f3cae463dd5a2b66a75 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37d516fb8ea84314904faa656eb18428 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3f3beca4e01457ea2aeeae2c5e184a3 : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa5c9f5826c647d3826fee4240d2537d : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1a06f362ef14f1584ea58b37f130d4c : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70c16dc3b55547e5af3a4156958c492a : SUCCESS in 4m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/203b4b5015f545dda4f39ee8cbc29ca2 : SUCCESS in 33m 17s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/48296a30faef4db3a66866aefc2e0320 : SUCCESS in 5m 36s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/abf543e0f76b415e8b4c525a63f871a7 : SUCCESS in 46m 38s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/08c579377e6c414f9b0860ce279fc10c : SUCCESS in 5m 29s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6cac4bc98389d7465f4fb43332dbd844fec12cd0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-10 20:30:36.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1a825e8f69b53b7c6288bee401c7f3d1d5715f3c","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-08-16 07:33:04.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ca2c7588aee6085984aee68bf22e506f9b200859","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-17 20:27:35.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7cd22df9b5f27b8aec7700e3cb67907bdaf512f2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-23 14:41:17.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0dbd1665c919c93e4f0e56be03fc0e100138a2d6","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-08-24 07:48:46.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f9789605b765b565c33c28fc5d09e0cc30275aa0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-08-24 09:49:18.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"88fb9e4c5abf440762e02c0b7775be24a6ffa263","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 10:54:57.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/58fea77a94b94a69bfe6d39f6c7c5447\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b1dcff008b3b4f7c8582ffe9dd4cb2f5 : SUCCESS in 2m 13s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5ed549601de44c819497d054aaf00af6 : SUCCESS in 45m 11s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/03dd8d2e16514b91ad55b47710583885 : SUCCESS in 40m 04s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bad5a1abeeec4db4a77991e56cd88a7a : SUCCESS in 42m 29s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/bc94c31fdeaf42ea98e6a8cc3ca7fccb : SUCCESS in 39m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/145ae301f1d8476ca43b5b2519d53274 : SUCCESS in 40m 19s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/393fcc1164154de08bcc8c29ccd513b8 : SUCCESS in 58m 29s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1570f82b0d0941db9bd7a39b9470533b : SUCCESS in 54m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c561670d2ac841e6a2ed5d1190a0f98e : FAILURE in 40m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/606dc1c0522a46e58542f46557ae860a : SUCCESS in 39m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64bf0988101249259a2769740a6d7525 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bafb2b7d03d14e449f504ae193691a4d : FAILURE in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/932c372ca7e14b34b5c97956779d12d3 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9b1b5b1ab074c48ba752bc8210e9ffd : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef33af4d54524b0fbbff6cb11e530eda : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89780ccc18444afab26b4056c31315a4 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71fd3aa27a0a4abd84d16c96a59e7f08 : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e34bc734c0e470497aa37a2c34fc6a5 : SUCCESS in 4m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/16a9f4378c224d12ba6976e4552d4add : SUCCESS in 33m 48s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/22480daf69e2435eb17a9ad375073f1c : SUCCESS in 7m 14s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/85647b31d88f41a99eece96ff18014d3 : SUCCESS in 43m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9d15e18e1fa94dc1831716d83eb4006d : SUCCESS in 6m 18s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a828ca2a3d7cdf7fb2e45131bd8f32b42e5efe19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-08-24 12:48:41.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"6bc0069a1e7818e192c44f28ef72fd8abdf71c78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 14:02:55.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/634b2e5ecc0c4e25911ac285de8bc0f0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0758a174a0e9443bae540fa4adc35c12 : SUCCESS in 1m 52s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/46a7e9ae7dcc4788aaa59fb85832c7e0 : SUCCESS in 46m 34s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/cf4c97d26d674000bbb516bbc53e5d13 : SUCCESS in 40m 26s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a759871eca94b6b80634b365b56cf2f : SUCCESS in 49m 05s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/321a67602b674863af0b439243ae65c3 : SUCCESS in 39m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/444456848cf94c6580a074f615a4d69c : SUCCESS in 45m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d8498534e5114beaad68663cd7f9c4c3 : SUCCESS in 1h 08m 57s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3506e3b5a4e646b98a4a6f2a398f46df : SUCCESS in 50m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6076c2d6bc59417eaca115c7ac71689b : SUCCESS in 47m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/11a3eae022dc4026bdc79edb6d0a7bfa : SUCCESS in 32m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4239a4226a464f20a9c71ee8f5893bf9 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f08273055df64ed39fb6dff48187e823 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e3c6dd582574534892a4a90e0d10433 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbfd379433de443bad3d5266d5fb011c : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29ff0654434f4fb0adef59d4a43c57fa : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb0164e64e7349e8a7c7ee000f7bcbad : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0df0ec0e322466db3180890b4e60f28 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac5b26ac6cd74745a516e19c32dae24c : SUCCESS in 4m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aa8d4b0d2e6b43efaba91bbfd5435df7 : SUCCESS in 35m 12s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/05a76e06f6ab4ce7ab6c93686bff4f3d : SUCCESS in 6m 00s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9a9f163deae0449f9de765133a40f432 : SUCCESS in 47m 13s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6e59fb82eaf74a7fb0641a5d41a4e181 : SUCCESS in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"08364c2e867504cf91448377d069164d2c4c4230","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-08-24 14:18:48.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"9cd80ca36105c0f574b3130e41a67d2a3bf1f4e7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-24 21:25:52.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"3d14b0484616c003d801b9ab6b5799e70a13df7f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-28 17:55:15.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"428cfe99960ca4ab52f527f47fccd1e122163c4d","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-08-29 06:33:33.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"4190422f294a2d935a9b2061948b28c6485dd3c1","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-09-12 12:31:48.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"888b947244aad8d5051a6769dfe4c4659c36d4ad","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-10-23 11:35:24.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"330c54e79cad9eb4b62119262a4a535814c129d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-10-23 13:26:36.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"b9ceeac87a1810420d2b4b154928366efbffb206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 14:40:36.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd54b073473b43fda1facba127d6f2e1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/417da9dd6d4a40698d509b1dece459ac : SUCCESS in 2m 02s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/03dc689983bd4f79b55e6f952b34468a : SUCCESS in 32m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/100fc469abdb48d38aeaf09a1df9bfbb : SUCCESS in 42m 23s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3efadff4423b4dc792d6cb58d171eb1c : SUCCESS in 55m 11s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e71f309ff8524fa99b0b6c493d69a905 : SUCCESS in 31m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c9694af0a0d44a0fa3503326a5acb05e : SUCCESS in 52m 09s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7f475d78050645a79e82a3a6d6222c90 : SUCCESS in 47m 37s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6ab82a57e74e4b028c155fa9ff5f77a4 : SUCCESS in 1h 07m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f6b8b979b77d4cd8895afab0f7dc6d28 : SUCCESS in 50m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c0ca63db397e4cd9a0684fa3f771aab4 : SUCCESS in 33m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8eb647b7c2014c10a7c47d622146a1c1 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff1466c0d3a34b97a3835c0fd51314a2 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92dd2441d903448ead7ab718a8f5f759 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32651652adc843efa25b8e52f9afc2a5 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5838fabdc4a4f73b4480d0e49c0806c : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fc0fa4f78364d2fb901fe2ea0f3bde0 : SUCCESS in 5m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b3b81b2add234fcbb42119085383d25c : SUCCESS in 35m 47s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7a162227056d4a4caee15e1a4b705138 : SUCCESS in 5m 53s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d7b6ae5c10b146cc88b133df6063b304 : SUCCESS in 50m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f418a9dbea744eb7a9a71974852f317c : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ab3acaa8aa22cea7bfa4359fe7dd461c696ca1a9","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-10-30 15:43:01.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"3624ffa0290ec3d489445871e421155200a9cc7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-08 12:18:34.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"25d3445377a2b03e8864c4a1e61a35dd4b79beb0","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-08 12:22:38.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"513887f63b4b80270340dcbe77ca0b7a04ced55f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 13:23:15.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77da4656572f4c6388a5dc45d145cfcc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7eaa322994534ffeb9482c794f4f7b3c : SUCCESS in 1m 51s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ea01d449d9454b348fe4fffe899f17cb : SUCCESS in 53m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b8a1f9e4f11449709b0d55aa08d67ad7 : SUCCESS in 59m 31s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3c76dd642994b6ebb7d9248ace560a5 : SUCCESS in 56m 27s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a58154fb533243298da4f743e79582db : SUCCESS in 48m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3bdcb70eb2df498fa8624e8378dda72e : SUCCESS in 39m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e0c4edd14f01489eb19fe158a2da63d9 : SUCCESS in 58m 11s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f0a58717f87c445cb23fd570a5a1b375 : SUCCESS in 58m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9a9343158862457caf1227275fdd172d : SUCCESS in 39m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3cacb086f57f4850bab92fa3bebb0d14 : SUCCESS in 35m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e626e501592a4e5cb51cfd7cc0132fce : FAILURE in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2433b6c257144c5b6d1163e754c7073 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f572ee5f5064ce09765e305db074891 : FAILURE in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4c73b0fa8194f32ad55e9d01f88ebeb : FAILURE in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25cc37b24c954c24a059eaedb5c09379 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08e0e6c2670744bb8321b29a87f64756 : SUCCESS in 4m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/10599d63896f4913ba6056f449237b2b : SUCCESS in 35m 49s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5b67f2e3a4d74eccade8bc06ac3755c6 : FAILURE in 6m 24s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d1f0964b43af4e678789b734487640cc : SUCCESS in 53m 31s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9b4546a26d404c2aa7c458efbbedac72 : FAILURE in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"91bbe47c36ab00b41fb5a0741a3a0172323ec8a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-14 09:56:58.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"b3795face0fcc2e5958ba6ed9177898c8682f8b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 11:10:34.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e499abc679d4ac1989b0a3374d6740e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9f370b4548e94e3c81da90cd8183c378 : SUCCESS in 2m 05s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ff11d2cba3f54933ab05413424dfead0 : SUCCESS in 49m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a6752dd3f50146fd94f792722767bd5d : SUCCESS in 1h 04m 21s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/772a11e6c9c546839a20056ec5a6a0fd : SUCCESS in 38m 32s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1fd779dc550b4b96bec91f48dc7df73f : SUCCESS in 48m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/368939bf4f4f456f9150ee80bb98ea91 : SUCCESS in 38m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/510ed471012e4e21a6b4071c5ce1b49d : SUCCESS in 1h 07m 54s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ba49b3aff1e3484d927d1eab679a8a78 : SUCCESS in 51m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4d9bb6710ba04acb951d552df854bbed : SUCCESS in 33m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1259a45f57d84a7ab39b35577ccaebea : SUCCESS in 48m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c9558b748664860a0dad0665a6b1229 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a089a9e939884a52b4e4a4ea4e651d6f : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fc0273414ca498989e1e302e196dcdc : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1bda6d970ce54c5f833ff9c76fc48db3 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7f3bd33ee72474ab58dd3588e2d8c5d : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0bfedcc348f4590b4e3c7cdf03392f6 : SUCCESS in 4m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f8d3ff01bd584d88b090708cf4564ed1 : SUCCESS in 35m 28s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/93d8101f2a29434ea7ac27ce687b36f7 : SUCCESS in 6m 42s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1f490c50967740ac88f138d83d33ab1a : SUCCESS in 41m 46s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cf4138a15f344f13bb3169fb8f93c400 : SUCCESS in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"78b2640fba9019f0393991b4b23e7710fcf8e269","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-15 12:40:15.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"d38389f6cf9be3595cd7c6c8854fad8afa0566c6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-15 12:55:08.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"61d38fe289e4a6ad754e87beb3256d0c9f78158f","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-15 15:23:31.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"414e992700d8d3ca86c3e566a6ed011ac193bdbc","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-15 15:43:29.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"1fa534e751c4f0e540b3d177e1cdc96ef8b3b5e3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-15 15:47:33.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"89eac201eb521a54bf6a7dbb98d84c65a074625d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-15 16:22:36.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"361924bb2240b4bada0abd49a6ea200318761e6d","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-15 16:23:06.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"167b9b129d11b64091d09aec9c59f1a75c9b89f5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-15 16:26:59.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"dc50ee057f25a067dc9e6f61a44eff5ee93115cf","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-15 16:33:57.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"bdcaf99f30d483fb07c2d4a19cee7796e7bd8bdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-15 16:51:56.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"b5a06ab82edb2b0283631dce0b7173bbdfaefd86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 17:58:00.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2afbd1d1591475999fb68a40d06f38b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0083dbf53f2d458b92bb784f43d0f0be : SUCCESS in 2m 45s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/14fd042fa17d458c957dfb69191bcdf5 : SUCCESS in 53m 48s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8e62f6d76d4244e4989547315c29f47f : SUCCESS in 43m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/846881634fb544cb970c83c4d4b24083 : SUCCESS in 39m 11s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7204d63c1aa3463593a43a0b00f409ad : SUCCESS in 35m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/18b6aa3b82f043aea7d5a90a778b3e82 : SUCCESS in 35m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d7398978b2ca4187b6e8b49c521b803c : SUCCESS in 58m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/415dbe37337f4357935093e02c126c4b : SUCCESS in 53m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/402bb61c2132406d928bcba558ce4318 : SUCCESS in 38m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8bbdd2d99ada45deaf639ac27e31c861 : SUCCESS in 40m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/528a9ae9f8fd4e5586dbdd5b1c041678 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5748484dd4ce4735bf9ba7d1638378e2 : FAILURE in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ebd93b8f97c4a51b3a1cef8e904a03f : SUCCESS in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36fda551056e4d3da10398bb80de6a9d : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7661082416dc4491a39732f7cc5fbed5 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c15f077997e1409b93d9cb0eb20e2e41 : SUCCESS in 5m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/50ebacb658c14187860fc9e1d5562ff5 : SUCCESS in 34m 26s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9f7ce6f1a2464647bd263d2cc73e4572 : SUCCESS in 7m 27s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c6aefa2ae3a34048a091350f5e541584 : SUCCESS in 1h 00m 17s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/717ef451f2e24b6eba3f6399f6ed621e : SUCCESS in 5m 50s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"fefa0480f222273274f32a841513f2eb01d2b462","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-15 18:42:42.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"9b484f8688746cc705395ab4bb28ca8afe9f82bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 19:47:13.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37f76c88397f4b1c887e016dcb552065\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2d8cc766c67a4068bd4664d6571c7d23 : SUCCESS in 2m 10s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/362024868e24495db2c0c501313e787d : SUCCESS in 45m 16s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/fb8e9ec0bf31474eae88d82e4eb0e481 : FAILURE in 24m 16s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5627e9e9f244972a14acc42cf4c3ac8 : SUCCESS in 44m 10s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/30f31be8ede6455b8829b3437e640e4e : SUCCESS in 35m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/64916501917246c98dc04b0d5f8e462d : SUCCESS in 45m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1d93701075814e079ac2835cb63dec0f : SUCCESS in 1h 01m 09s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5110476bc0e0499696820744fd4d4eda : SUCCESS in 55m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/34dbe174e22548999d9a66f3248758f8 : SUCCESS in 35m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6a3a6b114a384126b342259833a9b067 : SUCCESS in 39m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89d17dff2ac24acc8ad3ab801f70712f : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b5318ef70f548b8be609d934f30bb18 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4842aa9024bd4a37964b12a5966c7c23 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10a02e3d27854e64b8193986df122d7a : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8807cfd8d48b4229ac87d972ed6043bb : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a26d0d1831843d7aca33dd90f4a068d : SUCCESS in 4m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f693ab3d445a4205b8087f3f33f9270b : SUCCESS in 31m 39s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d81d8df474d4415ea3666b75dfa73434 : SUCCESS in 8m 16s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2e91696a52dd4bcbb6ced7177dee5644 : SUCCESS in 51m 16s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3dbba3c1e4ca4a6eb0be5b0b771a1493 : SUCCESS in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"9fe2c54b984840408b6c4658355b49b1e1ed1866","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-15 22:50:30.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"4f6ea4e744c54b3aabe96a998c5a7e7a7038ade4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-15 23:04:31.000000000","message":"Patch Set 17:\n\n(6 comments)","accounts_in_message":[],"_revision_number":17},{"id":"b639af72515ca920a7e4fcb9de04d621e2899053","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-16 13:00:31.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"158710eaaf74201a9fbdeeb4fd522be381d87053","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-16 13:00:36.000000000","message":"Patch Set 17:\n\n(7 comments)","accounts_in_message":[],"_revision_number":17},{"id":"7a74c1aad607e0892e9c1e7f77e302d2e91c1df9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 14:18:04.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2aa973d3c3774479a4a995267e74a674\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0a0988bba0c849138d0c1a3184b768b4 : SUCCESS in 2m 17s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/220ebf37d800468d8440f4cf93120ecc : SUCCESS in 34m 59s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/45fd0398de5e4f8286b48a8fc4219d4d : SUCCESS in 1h 05m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f88ac1aba4c465e90fbddc216fd1ed3 : SUCCESS in 37m 51s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8e0893db19b74276964bfbb8bd356115 : SUCCESS in 44m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/688738c059fe44dbb483b8ea51053153 : SUCCESS in 37m 49s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/db2f34f92a5241f39eb465c6667123c6 : SUCCESS in 1h 09m 58s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fd25838cc59a4e4a8d1aa66e9c80a71c : SUCCESS in 1h 06m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/259e03f861eb4c76b52d7ffc087ad11a : SUCCESS in 35m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/886c8d1905b14b4f87ebd7b4f6b752b4 : SUCCESS in 48m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/352d7dfbf2c24fa9984f6391771aef25 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2fcd8317b6846ae9ceb8f53c274db43 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0923c1ca807f4e8f98a80c36f8d9ba8a : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ea26dc405474bc0b8f6518577d97ffd : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50ec94b8431a4ebbafac2fb42963a95e : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f53cb966179f4b559a0b3efcb04b0e15 : SUCCESS in 3m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/49a2ebe5a5184fd0879979e83eceaabf : SUCCESS in 33m 45s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9707b4cd4d6c464c85185ddbff2bec76 : SUCCESS in 7m 17s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/dd2a49961f534453bca5d00fb9520f9c : SUCCESS in 55m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c22374980b3c459ca1d0c46feb00bcf1 : SUCCESS in 4m 06s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"2a8f5afdce5165502f4ed10b44839f4acbb95286","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-17 12:16:35.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"4beb4832e47f019694149acf3893005b37ac7a97","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-17 12:17:37.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"96e0506e1a62ad8193968831c866d8f008f0e2e5","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-20 13:00:17.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"2e6e8253d3842c1597c9c1f675652ef6617a6147","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-20 13:02:52.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"a2af7f406478e352232e1d3193bc5b654809c74f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-20 15:27:20.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"1e1f85977349f7d97749ab93657162c158fae28e","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-20 15:27:39.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"8a7e442c616d70351f096b6d85969814fff7753a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 16:41:50.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6af4759c5dd45f0b9084a17d18fa6ad\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e5a63b70dbbf43e2a2cec9452976db6c : SUCCESS in 1m 57s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/faeafd9f8ce943ac9ce6d6e6b971d6c6 : SUCCESS in 47m 32s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/413f76a47fc04280844b766737cefb9c : SUCCESS in 1h 00m 22s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0626a37eacb745128be1048fd0767751 : SUCCESS in 47m 51s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/df107926acdc40f99a92efb6da26cf21 : SUCCESS in 47m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5cffe26745c341e78671e875760f70be : SUCCESS in 44m 37s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0fbe2ca548294ef28615096a6640435e : SUCCESS in 1h 08m 57s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bb9a23f5b0e84e0c85205d6d73b0c2cf : SUCCESS in 54m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/adfa517a0fdf406cbeb7e77ccd962aad : SUCCESS in 49m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cf1afdbd0afd4beeaaf1d38cecb4f69d : SUCCESS in 49m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c479fac21604abb88688ced8da27140 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6436e2e7ca8e46508abd5256fc57d491 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9381b5de10db450db1d9f2411431151f : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/859e0e8ed6a648b290305209ea38fb08 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/365508894424470f84b3fbf1663c31c4 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7ed12a90de94181bbc3fe50c987573a : SUCCESS in 5m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f481248b644c48c9b85a581fa7f9f30b : SUCCESS in 42m 08s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/160d9bbd40964a3985854c60514d9f81 : SUCCESS in 5m 59s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/86eededd30a6428eafcfb0518ada5b2a : SUCCESS in 38m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8e0a0ee088bc43c199d514b77a87f6e0 : SUCCESS in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"8352fabd7f4147f4912f2cd9493ff8d268f5be62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-30 13:14:03.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"e431193801c47f2261269b8aed88afcaa55fab07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 14:41:11.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dc3920dacb6468bb730fd5e5756bb44\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a6e806ec1e2a42ebb950b921ea6314b8 : SUCCESS in 2m 07s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3bd82b6672404a94a0cfb0ab2ad92e50 : SUCCESS in 49m 58s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1fc19f02bd2f45658f06250eebe008c5 : SUCCESS in 1h 05m 34s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1573bd3c2ef344198f6cf082dbbc7037 : SUCCESS in 43m 49s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7a02127f4b354833b34b5b68212cb4eb : SUCCESS in 55m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1b0a7fc18db54ecfa74122e0d0b6bb9f : SUCCESS in 43m 10s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/32e8af8f684e4f2db3dd082c26f03fd6 : SUCCESS in 1h 19m 35s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7be3cf601371444b808dde5d341653fc : SUCCESS in 57m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0152ccbf0d8142e6ab63ed2f79c7dd16 : SUCCESS in 40m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8d39ab284fd34a81b03c0bfffd832973 : SUCCESS in 46m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/908bbdfd06a648ebaebd0229d2ae0fbc : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c5186b40db24ef782bda076c921243f : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7986ccbc3258470081780fe209649c84 : SUCCESS in 6m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47587a3d2c454408b198649548284e56 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd66b8470b84ffba36a402fd27993d4 : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d6cddcc0ca148f3ab9bbef35fad40f5 : SUCCESS in 6m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cbe9fc9bd1584536bf4eefce33228ee3 : SUCCESS in 39m 15s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6a0c880caa184309a7b52a6d027625b8 : SUCCESS in 6m 57s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e2d3f8f0acc24b368a109e191ee0ffe0 : SUCCESS in 31m 35s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6285de15fbfa4b73ba0b45754ea68c2e : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"c0a60ab75e89bd0cb0b8fdcf34ba15de55c6ad70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-12-04 09:58:06.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"dcd929a5611b34ea04fa501dcb7cb38232861111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 11:14:56.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/709adfb98c4d4a8a9b0e87ecc82b2725\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6af1eff495a848848c4855ba773a7fca : SUCCESS in 44m 58s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/33d65e5351b44521b92c8ee8d276abe9 : SUCCESS in 1h 05m 00s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/19ac066cd47d4c459aafd2ee5a2197d8 : SUCCESS in 54m 40s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ec102203e2a847ae88f9e99e8362e65d : SUCCESS in 46m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7e1491d6db6d4ed081385e6fbb819ac6 : SUCCESS in 39m 24s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2d1f4a8e5f724aecb57e8b1a7bc9faaf : SUCCESS in 1h 03m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/44e1200bba3845a8a1611c4f46d0b402 : SUCCESS in 1h 11m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/850f47943ab1445f952af20a3d102138 : SUCCESS in 45m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cc380340ede84bec99b866b9e9b6aee0 : SUCCESS in 50m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19394546baec494d82540defc5547ddd : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcc9171b09ce48cfab85eef183398da4 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d831377c2e04f2ebd9d53ae154bda03 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ffb7761901bd45a39815f125d48c6d48 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f228aff3c83a4a39838c7c9c4f1bea97 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5ab1e7c147146e8bf165b72d9f85e5a : SUCCESS in 8m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cd43ea87f74e4179bf5b5a0d5330ab3b : SUCCESS in 47m 03s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/87e269e99eb646358eba34e1dfefc01c : SUCCESS in 7m 15s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f0e4a16d0fd14929a2c14899247c7467 : SUCCESS in 41m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/37959401a3af47e4a87f96ec59925b7f : SUCCESS in 5m 24s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"a5b6422fa41cd90563e57470e6b2233e8dfbd499","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-12-18 12:14:49.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"daff4891e6951d1370ebf94ca0ce17a0b6d6decc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 13:22:21.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c1f9fd88ceb4104903555401a26dd60\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f0cbd177f9b24fde8f18710a4e77ab28 : SUCCESS in 44m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e65d2451aa914dd2a129a6614dce27b9 : SUCCESS in 59m 52s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/92709b0d9491462fb01b653684a42012 : SUCCESS in 41m 57s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8688c29931dc42a99c25a990d277490b : SUCCESS in 35m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0ee4cd3f9a2f4e3ab6f29ceca2056571 : SUCCESS in 52m 35s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a24338f66187451c98d1fc56bbd49b34 : SUCCESS in 54m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7673af6a8c604af1a2ac8213c6b7ef43 : SUCCESS in 54m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/db6fdc3d22b54a4b85c71c86f78ea5cc : SUCCESS in 30m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3905937e5fa64787b05fe3b0a3275fad : SUCCESS in 53m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a25927e2269640e3b2cfc35da14741fb : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20ca373eccb346d79012b03a1280d6ed : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d2a5df73d9f4da6b91de7a3443b5fbb : SUCCESS in 6m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2ec6ce00179487a96a37544b48d49b1 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5595e537800f4f989f61da6e5521bea8 : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea0de6876e244e46b7b60d9ddef1aeb4 : SUCCESS in 5m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/da32cf8e65db4cf7aa453d2d03389cf8 : SUCCESS in 44m 39s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2ae40dad4a574f1a9511f351bb86653e : SUCCESS in 8m 12s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/92cc186854db45c19f84aeba58981b62 : SUCCESS in 43m 11s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a66b60de45164fbfbfb35190ffb3dc0d : SUCCESS in 5m 37s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"a43c8ef589a1706f04ea286dc9846d2eb309e5cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2024-01-09 10:51:33.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"8681ca22d5b7e567e4a877f4d410972672c60643","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 12:03:48.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c036e6b2ec1f4c2f97d98167e86136d5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8dd0867555a84fae84d2ea70bf79af66 : SUCCESS in 49m 34s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e9c595ae62204d1b8b31112e3c67d3b8 : SUCCESS in 43m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/84c6a9c69247413b8975715cfab99c81 : SUCCESS in 37m 08s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/97f30f70600c460e9eb949a9b218e330 : SUCCESS in 41m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/042d7be1415141cb93ab493a50e43e79 : SUCCESS in 35m 41s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/73d10c05fa4349a1a448caecb0159228 : SUCCESS in 1h 01m 10s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d3f8a90a57a24600ac811b01d5554d52 : SUCCESS in 58m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c3e1e1fdc3ff4dbd8b2590594baf9ea2 : SUCCESS in 42m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ecff08a5cc0c49768f82dd43be6a22f5 : SUCCESS in 39m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e56ee67bc9594ce7903fd913839a05be : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d211d2e703413ab51b469b421210ae : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/727fd49ac29d46c38c0507ce6ce3d39e : SUCCESS in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d7a4925c32324ec5a8b01528d47451ba : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/668f52e7ccce4afba578db8e7a1b522b : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/840c37d5b1294aec9773f99904b2c96e : SUCCESS in 6m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c157ecdf24ff481e82afa63368b9875e : SUCCESS in 32m 14s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/669936274d4942f4ab76004173b523a6 : SUCCESS in 7m 38s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f653592ab6b44ff0aefa2f9cb0dcf9ea : SUCCESS in 49m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/dd18820de32a48b9bbe1cca3338f457e : SUCCESS in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"065ede6db4f290fc72cd268ccd7fa50b668c1675","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2024-01-16 14:15:05.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"0d3090407b30ee62a7f5102e9ac6c8c2137b258b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 15:40:20.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a91bfc200efc41729c8eea909239ef5d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7cb0cbdd4f1e4db78c531e5cfb832b55 : SUCCESS in 51m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b5e9b914b7094695bdf17a92edbeb9bd : SUCCESS in 1h 03m 48s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8b5ae46c7284fb4ab5f768aa3fa2949 : SUCCESS in 1h 00m 41s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f8ce7ebb2737468da51aa18e3ee6722b : SUCCESS in 43m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b8a3b2a25f8742be9e4649146d434f8a : SUCCESS in 47m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f6fe8d2406c642ff9d8c8810047782fe : SUCCESS in 1h 15m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/51c8e5a5022c4507a24df498cc204ffc : SUCCESS in 1h 10m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3eebed550d964255a842a9bd82f1f16b : SUCCESS in 38m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/12ad43d53ffe4951873a4fd17642ff3d : SUCCESS in 46m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/639fad37272341b9bb3506170aec3819 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d09de6c30d8482d81f389aee94cfd57 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/900027b90ae54d778151759be602bfd2 : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02eb9dbe1bca4eee82962aaf230af9f2 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbe42622d7644f1ca8c0d1e6378b7606 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2bfe9990b3f49af985579bebd03e7f0 : SUCCESS in 6m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6779245137c3442cbc09a1d19ac04fad : SUCCESS in 48m 46s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7260d476b258447385b809311469ffde : SUCCESS in 9m 24s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ee5b4c099e0a4c5fb6719215f60876c2 : SUCCESS in 45m 38s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/790f74e194b240899a94102bf5080e61 : SUCCESS in 6m 26s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"abbc782f2bd00419e25f37879c1ee68e7be900a3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-21 20:11:37.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"b1d27d2a10e298e8642eb5d07a73ae8157f3c034","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2024-01-22 09:22:32.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"970fa83df0f73e3041cf84aae1ae6df4ac52df37","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-22 11:17:58.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"e5ce65c49cfac50aec4bd77bf7d955c060d33f3d","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2024-01-22 13:47:32.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"b37890a904d0bc91214c8578418e854f14e2752c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2024-01-22 14:23:42.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"33adbb5db2a9639525db67bb20b8a526834f1fca","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2024-01-22 14:25:17.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"6217e4a7c0a60113e16eb8a57b3acbe94f61c991","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-22 14:40:11.000000000","message":"Patch Set 25: Code-Review+2","accounts_in_message":[],"_revision_number":25},{"id":"2ed039740f844d8c38cb48b168dddda33e17f21e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 16:07:30.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d28927e5fddf4d789afdcde014b838ad\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/435b953a09e04f87a35e01517cbb8d80 : SUCCESS in 48m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ba0704db092341649ec1a078e3f1c7dd : SUCCESS in 50m 12s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9002be0bed64a7f82beaa872c3f63e3 : SUCCESS in 54m 25s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/efdb91ee394b46f1b30f3a595987c250 : SUCCESS in 43m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/07bd4ba9af474746a58d98e1d73a361d : FAILURE in 41m 01s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ebe6520076124be29504ff7e70ff9518 : SUCCESS in 1h 11m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2e95412e427144bab9c0f5c0174a9a60 : SUCCESS in 1h 13m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c0fba51fba284813844b5eff0c85749b : SUCCESS in 44m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/101e4022664f40298fb5eddab66b2e86 : SUCCESS in 50m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0de4554ae0134270aaf93a6f2f8ff9f0 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a0899b7273c4315b65334a973b9b20b : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc38c91fde504507b3a91904e6abef91 : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/43e8a014555444f88fa346eb19cd9f6c : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e5cb9cf98d49b781881e2c11171234 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58e71fab714b4428900df1a9c0655332 : SUCCESS in 5m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6c36e70169154d5086f0dba8c8b57217 : SUCCESS in 31m 58s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/92aeb60643474de886ac87973060a9e4 : SUCCESS in 7m 23s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bbc901454535437c81be9ce963e615b8 : SUCCESS in 50m 57s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4b81527a109c4bd6b7ee32cff852dcd6 : SUCCESS in 4m 50s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"450e92d427f4e57d22ce666030d41fadc0342ceb","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2024-01-22 16:27:11.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"ac2afd345b1efb3d3f706487e89d99a8817f9378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 17:58:14.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01ce74a2ac8a4289bc4fd8d9a43ce7cb\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/eb6b5b5235a2431fa14d42634a5f17fd : SUCCESS in 37m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/fc5bda9d85014762aa8ef0a39066d49a : SUCCESS in 1h 08m 57s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6f6f747c9284ab6ab06248e91cd5e0b : SUCCESS in 42m 47s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8ff6b2ed5d6d49439fc96b2cf3694b40 : SUCCESS in 58m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/04770c24c6a0449090fe44f1c65b67ab : SUCCESS in 50m 52s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b85aa677eef94998a1148f752d496f46 : FAILURE in 1h 21m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/96975ceb67614c4690d7333cfaed5d86 : SUCCESS in 1h 14m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1467733c35644c00ba462fa16fe76cc9 : SUCCESS in 45m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1036ddcbcf1e4b3699bab9721bfe0f40 : SUCCESS in 48m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2e0883d4c1f4881a7577fce4dfd7efb : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4a6e0892a5a448187343c4551f2aea0 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e6943803d95400faaf345a737cef3fe : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d26a05002af42f9a407e291da8c0439 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcaabe8047a74c0688afc3d4eaba16d7 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d751bd71e780458385057b438461255e : SUCCESS in 4m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9ebcce84b0b54580a313b45a1ff873c8 : SUCCESS in 31m 51s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9d687966753e418a9d704b00dd2a04ef : SUCCESS in 6m 51s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2ebfe8067d2146a4913f3395e781bce2 : SUCCESS in 36m 21s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/39bf238a04bf407d81013fc7cc41e21f : SUCCESS in 5m 31s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"aef02459eeb9971d79d791ccedf8978f43be05e2","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2024-01-24 08:13:46.000000000","message":"Patch Set 25: Code-Review+1","accounts_in_message":[],"_revision_number":25},{"id":"c54524eda08ec25bb4d8b7394ffa95639415d663","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2024-02-05 13:18:36.000000000","message":"Patch Set 25:\n\n(3 comments)","accounts_in_message":[],"_revision_number":25},{"id":"a65dc903156883eefe9aedf3cdbb3e2ded91f190","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-14 18:59:24.000000000","message":"Patch Set 25: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"8ab0d2dfd019da969140a8e5494fc8b37e622c89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 18:59:57.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"b1faefc60a3378056853622a5ade8748034addf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 20:16:54.000000000","message":"Patch Set 25: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c91b8438c93b41d5b4ee56af365f0fb1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3bb2ccee67934c4cb891962f8dbb27e9 : SUCCESS in 48m 01s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/dbf997c82ce7493fb6aaf36398f57d75 : SUCCESS in 55m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1911d4ca3184498dbd52c0ae92d72606 : SUCCESS in 50m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e7d28a660a0f4a47858684dc9ca5f13c : SUCCESS in 1h 16m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1ef81234b9d94139bc1a0a62a41eba64 : SUCCESS in 47m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a031ca55b65b4550ad39c8325c44f864 : SUCCESS in 46m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3361d72df1a74e0baf0a325a330451a5 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8208e45dd95f487f9bc17d1d11503934 : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41c0b41a8e9a4eeda7f2138cb693da19 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/881f915543d44d1b8eedcb8bd6166333 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8509cc8bde574b4d9e10530c9585c8f3 : SUCCESS in 4m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8d634ca53c0240bdb3b9a324b2362b5d : SUCCESS in 32m 23s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0e1ff465b08f44e1b5667bddab7ae241 : SUCCESS in 6m 18s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/044eccb8499a4fcf87f2b2f5a51a93a2 : SUCCESS in 42m 59s","accounts_in_message":[],"_revision_number":25},{"id":"2b646a9905998606f44dd8f48e0ee1275af99e17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 20:16:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":25},{"id":"2370c98371bb478cc12ef3307b78a2658b3171ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 20:18:02.000000000","message":"Patch Set 25:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a9016e200de42a8af4c292497074ca3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e06eea252c64c7bb9e3314c1fd293f9 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7306bab36ae94ce9886875c20b2311d6 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"60a6ceed09515842366e67de86b5aef783f1dcd2","revisions":{"41fb6ff3889a151dbb3dbdbf182014997bff342e":{"kind":"REWORK","_number":1,"created":"2023-06-08 16:04:24.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/1"}}},"commit":{"parents":[{"commit":"0805c874df7fa44a84d3ea99185265896f2a9fea","subject":"Merge \"Add basic managed records coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0805c874df7fa44a84d3ea99185265896f2a9fea"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-06-08 15:56:01.000000000","tz":120},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-06-08 16:02:37.000000000","tz":120},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/41fb6ff3889a151dbb3dbdbf182014997bff342e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/41fb6ff3889a151dbb3dbdbf182014997bff342e"}]},"branch":"refs/heads/master"},"ff6dc52cd690a825e902d61d714e60d6a509dc1b":{"kind":"REWORK","_number":2,"created":"2023-06-19 12:39:05.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/2"}}},"commit":{"parents":[{"commit":"3621b9d1b1123f9004ebf07b026ed00ca53a9b76","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3621b9d1b1123f9004ebf07b026ed00ca53a9b76"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-06-08 15:56:01.000000000","tz":120},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-06-19 12:30:22.000000000","tz":120},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ff6dc52cd690a825e902d61d714e60d6a509dc1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ff6dc52cd690a825e902d61d714e60d6a509dc1b"}]},"branch":"refs/heads/master"},"2352543deb0bdf57fecc41b82bbc31fed4268fb4":{"kind":"REWORK","_number":3,"created":"2023-06-20 12:45:05.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/3"}}},"commit":{"parents":[{"commit":"3621b9d1b1123f9004ebf07b026ed00ca53a9b76","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3621b9d1b1123f9004ebf07b026ed00ca53a9b76"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-06-08 15:56:01.000000000","tz":120},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-06-20 12:42:34.000000000","tz":120},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2352543deb0bdf57fecc41b82bbc31fed4268fb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2352543deb0bdf57fecc41b82bbc31fed4268fb4"}]},"branch":"refs/heads/master"},"2c1625185ea4d466ba24812133558652d9488f7b":{"kind":"REWORK","_number":4,"created":"2023-06-20 14:37:03.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/4"}}},"commit":{"parents":[{"commit":"3621b9d1b1123f9004ebf07b026ed00ca53a9b76","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3621b9d1b1123f9004ebf07b026ed00ca53a9b76"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-06-08 15:56:01.000000000","tz":120},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-06-20 14:35:49.000000000","tz":120},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c1625185ea4d466ba24812133558652d9488f7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c1625185ea4d466ba24812133558652d9488f7b"}]},"branch":"refs/heads/master"},"cfefee91549a1214cd2d0c6a72a247809d86115a":{"kind":"REWORK","_number":5,"created":"2023-07-11 11:41:09.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/5"}}},"commit":{"parents":[{"commit":"963751f2b403c3e9a73008ac62854ed3d3758247","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/963751f2b403c3e9a73008ac62854ed3d3758247"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-05-02 15:41:20.000000000","tz":120},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-07-11 11:38:02.000000000","tz":120},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfefee91549a1214cd2d0c6a72a247809d86115a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfefee91549a1214cd2d0c6a72a247809d86115a"}]},"branch":"refs/heads/master"},"df888dec25894c935e6df8858f6214a277b3ddec":{"kind":"REWORK","_number":6,"created":"2023-07-18 15:04:00.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/6"}}},"commit":{"parents":[{"commit":"7daa2578a51cad61fed8e645d9e5158d5c5f5c17","subject":"Fix a sqlalchemy 2.x compatibility issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7daa2578a51cad61fed8e645d9e5158d5c5f5c17"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-05-02 15:41:20.000000000","tz":120},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-07-18 14:59:29.000000000","tz":120},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/df888dec25894c935e6df8858f6214a277b3ddec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/df888dec25894c935e6df8858f6214a277b3ddec"}]},"branch":"refs/heads/master"},"07b50f9d4c4a7d3df2c7513f71e67b8692b24f4b":{"kind":"REWORK","_number":7,"created":"2023-07-18 15:22:13.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/7"}}},"commit":{"parents":[{"commit":"7daa2578a51cad61fed8e645d9e5158d5c5f5c17","subject":"Fix a sqlalchemy 2.x compatibility issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7daa2578a51cad61fed8e645d9e5158d5c5f5c17"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-05-02 15:41:20.000000000","tz":120},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-07-18 15:21:43.000000000","tz":120},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/07b50f9d4c4a7d3df2c7513f71e67b8692b24f4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/07b50f9d4c4a7d3df2c7513f71e67b8692b24f4b"}]},"branch":"refs/heads/master"},"4706c0f595201656145e122bf7c1104330da658f":{"kind":"REWORK","_number":8,"created":"2023-07-19 15:05:20.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/8"}}},"commit":{"parents":[{"commit":"b77934beb7d2c707b2a2e10e023a2e2542c6cfe8","subject":"Add job to test with SQLAlchemy master (2.x)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b77934beb7d2c707b2a2e10e023a2e2542c6cfe8"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-05-02 15:41:20.000000000","tz":120},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-07-19 15:04:16.000000000","tz":120},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4706c0f595201656145e122bf7c1104330da658f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4706c0f595201656145e122bf7c1104330da658f"}]},"branch":"refs/heads/master"},"0d1049ad2b7d14e23de2a27143676e47a84eb47e":{"kind":"REWORK","_number":9,"created":"2023-08-08 14:34:34.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/9"}}},"commit":{"parents":[{"commit":"b77934beb7d2c707b2a2e10e023a2e2542c6cfe8","subject":"Add job to test with SQLAlchemy master (2.x)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b77934beb7d2c707b2a2e10e023a2e2542c6cfe8"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-05-02 15:41:20.000000000","tz":120},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-08-08 14:17:56.000000000","tz":120},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0d1049ad2b7d14e23de2a27143676e47a84eb47e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0d1049ad2b7d14e23de2a27143676e47a84eb47e"}]},"branch":"refs/heads/master"},"cc91a10f3301624da684c3b579b832a0b026893a":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-08-08 18:49:52.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/94/885594/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/10"}}},"commit":{"parents":[{"commit":"5fa2dd8efc65723d68b2ae369e86e85f9ab2e48a","subject":"Merge \"Fix missing oslo.versionedobjects library option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5fa2dd8efc65723d68b2ae369e86e85f9ab2e48a"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-05-02 15:41:20.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-08 18:49:52.000000000","tz":0},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cc91a10f3301624da684c3b579b832a0b026893a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cc91a10f3301624da684c3b579b832a0b026893a"}]},"branch":"refs/heads/master","description":"Rebase"},"e6d6900d0b2030d0bb97ce36dbb2d5bb07ef0bb5":{"kind":"REWORK","_number":11,"created":"2023-08-24 09:49:18.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/11"}}},"commit":{"parents":[{"commit":"06482503881c89a709a1fe79c2c8d600244ad168","subject":"Fix DB upgrade with using SQLAlchemy 2.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06482503881c89a709a1fe79c2c8d600244ad168"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-08-24 09:39:36.000000000","tz":120},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-08-24 09:39:36.000000000","tz":120},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e6d6900d0b2030d0bb97ce36dbb2d5bb07ef0bb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e6d6900d0b2030d0bb97ce36dbb2d5bb07ef0bb5"}]},"branch":"refs/heads/master"},"394d09c3c533c4d765a0a01b259336c31a0cbaf7":{"kind":"REWORK","_number":12,"created":"2023-08-24 12:48:41.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/12"}}},"commit":{"parents":[{"commit":"06482503881c89a709a1fe79c2c8d600244ad168","subject":"Fix DB upgrade with using SQLAlchemy 2.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06482503881c89a709a1fe79c2c8d600244ad168"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-08-24 09:39:36.000000000","tz":120},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-08-24 12:45:42.000000000","tz":120},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/394d09c3c533c4d765a0a01b259336c31a0cbaf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/394d09c3c533c4d765a0a01b259336c31a0cbaf7"}]},"branch":"refs/heads/master"},"73ef8976ef50661533871be1aa6b37468e896caa":{"kind":"REWORK","_number":13,"created":"2023-10-23 13:26:36.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/13"}}},"commit":{"parents":[{"commit":"ac556c65b5b7254176485829fc5fff8cffadd9a4","subject":"Merge \"Remove deprecated designate-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ac556c65b5b7254176485829fc5fff8cffadd9a4"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-05-02 15:41:20.000000000","tz":120},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-10-23 13:16:39.000000000","tz":120},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/73ef8976ef50661533871be1aa6b37468e896caa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/73ef8976ef50661533871be1aa6b37468e896caa"}]},"branch":"refs/heads/master"},"af3a3db700fc787a9dcfca268e9f0633784ee1fc":{"kind":"REWORK","_number":14,"created":"2023-11-08 12:18:34.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/14"}}},"commit":{"parents":[{"commit":"30d8c7081405f67aa8586194a39da2defa566014","subject":"Disable tox sqlalchemy tests until fixed in oslo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/30d8c7081405f67aa8586194a39da2defa566014"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-08 12:03:46.000000000","tz":60},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-08 12:03:46.000000000","tz":60},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af3a3db700fc787a9dcfca268e9f0633784ee1fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af3a3db700fc787a9dcfca268e9f0633784ee1fc"}]},"branch":"refs/heads/master"},"9717f1503956ef422da758493019de5557ec5657":{"kind":"REWORK","_number":15,"created":"2023-11-14 09:56:58.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/15"}}},"commit":{"parents":[{"commit":"4e185d4e266299ce22376024977cd9471dcaa0ed","subject":"Merge \"Make sqlalchemy-2x tests voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e185d4e266299ce22376024977cd9471dcaa0ed"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-08 12:03:46.000000000","tz":60},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-14 09:53:26.000000000","tz":60},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9717f1503956ef422da758493019de5557ec5657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9717f1503956ef422da758493019de5557ec5657"}]},"branch":"refs/heads/master"},"ad8058b8ff1c2f9eec867d3c0f233857a6a23492":{"kind":"REWORK","_number":16,"created":"2023-11-15 16:51:56.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/16"}}},"commit":{"parents":[{"commit":"4e185d4e266299ce22376024977cd9471dcaa0ed","subject":"Merge \"Make sqlalchemy-2x tests voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e185d4e266299ce22376024977cd9471dcaa0ed"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-08 12:03:46.000000000","tz":60},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-15 16:51:01.000000000","tz":60},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ad8058b8ff1c2f9eec867d3c0f233857a6a23492"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ad8058b8ff1c2f9eec867d3c0f233857a6a23492"}]},"branch":"refs/heads/master"},"c6c3dc992ffb8d6edca77f4de338e2eb0e73ac14":{"kind":"REWORK","_number":17,"created":"2023-11-15 18:42:42.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/17"}}},"commit":{"parents":[{"commit":"4e185d4e266299ce22376024977cd9471dcaa0ed","subject":"Merge \"Make sqlalchemy-2x tests voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e185d4e266299ce22376024977cd9471dcaa0ed"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-08 12:03:46.000000000","tz":60},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-15 18:41:27.000000000","tz":60},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c6c3dc992ffb8d6edca77f4de338e2eb0e73ac14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c6c3dc992ffb8d6edca77f4de338e2eb0e73ac14"}]},"branch":"refs/heads/master"},"8a1117b204c2327c93cfda725a8f6110e32c79ab":{"kind":"REWORK","_number":18,"created":"2023-11-16 13:00:31.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/18"}}},"commit":{"parents":[{"commit":"aae83d09b88c90a5f0872daa206dbf5921aa1827","subject":"Fix bug with new zone default values being ignored","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aae83d09b88c90a5f0872daa206dbf5921aa1827"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-08 12:03:46.000000000","tz":60},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-16 12:57:25.000000000","tz":60},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8a1117b204c2327c93cfda725a8f6110e32c79ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8a1117b204c2327c93cfda725a8f6110e32c79ab"}]},"branch":"refs/heads/master"},"116fcde3bf011bc2bce02cbced13994f1bc0d6a6":{"kind":"REWORK","_number":19,"created":"2023-11-20 15:27:20.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/19"}}},"commit":{"parents":[{"commit":"3e5ad208565305972c380020adadb3529ff76e00","subject":"Merge \"Add charset and engine parameters to alembic create tables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3e5ad208565305972c380020adadb3529ff76e00"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-08 12:03:46.000000000","tz":60},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-20 15:24:22.000000000","tz":60},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/116fcde3bf011bc2bce02cbced13994f1bc0d6a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/116fcde3bf011bc2bce02cbced13994f1bc0d6a6"}]},"branch":"refs/heads/master"},"5ac83ea10b40a98eb423cf124d254204d04536ef":{"kind":"REWORK","_number":20,"created":"2023-11-30 13:14:03.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/20"}}},"commit":{"parents":[{"commit":"f9a2494b2db9a1ee63c9d8f72b54417846c6096e","subject":"Merge \"Remove executable flag from recordset.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f9a2494b2db9a1ee63c9d8f72b54417846c6096e"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-08 12:03:46.000000000","tz":60},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-30 13:10:39.000000000","tz":60},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5ac83ea10b40a98eb423cf124d254204d04536ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5ac83ea10b40a98eb423cf124d254204d04536ef"}]},"branch":"refs/heads/master"},"8e76f2f6caae51dc6a38d4ba5f4ecc70e0b204ab":{"kind":"REWORK","_number":21,"created":"2023-12-04 09:58:06.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/21"}}},"commit":{"parents":[{"commit":"f9a2494b2db9a1ee63c9d8f72b54417846c6096e","subject":"Merge \"Remove executable flag from recordset.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f9a2494b2db9a1ee63c9d8f72b54417846c6096e"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-08 12:03:46.000000000","tz":60},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-12-04 09:57:05.000000000","tz":60},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8e76f2f6caae51dc6a38d4ba5f4ecc70e0b204ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8e76f2f6caae51dc6a38d4ba5f4ecc70e0b204ab"}]},"branch":"refs/heads/master"},"81e879526528f96c5efd1a0c76dc628cc2ed7880":{"kind":"REWORK","_number":22,"created":"2023-12-18 12:14:49.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/22"}}},"commit":{"parents":[{"commit":"af940fb6e4431c9eb9799cc983a85465ec865011","subject":"Increase minimum test coverage to 96%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af940fb6e4431c9eb9799cc983a85465ec865011"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-12-18 12:13:26.000000000","tz":60},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-12-18 12:13:26.000000000","tz":60},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81e879526528f96c5efd1a0c76dc628cc2ed7880"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81e879526528f96c5efd1a0c76dc628cc2ed7880"}]},"branch":"refs/heads/master"},"f0c0ff4efdf63fb4b73bd77d678467e9aeac1619":{"kind":"REWORK","_number":23,"created":"2024-01-09 10:51:33.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/23"}}},"commit":{"parents":[{"commit":"9d96f5f955a48953cf15ce1ce40f163813265c07","subject":"Merge \"Improve Dyntext coverage and removed unusable code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9d96f5f955a48953cf15ce1ce40f163813265c07"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-12-18 12:13:26.000000000","tz":60},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2024-01-09 09:20:34.000000000","tz":60},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0c0ff4efdf63fb4b73bd77d678467e9aeac1619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0c0ff4efdf63fb4b73bd77d678467e9aeac1619"}]},"branch":"refs/heads/master"},"978049fa158f1d844edf69e4487d20f4c3e4bef5":{"kind":"REWORK","_number":24,"created":"2024-01-16 14:15:05.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/24"}}},"commit":{"parents":[{"commit":"002005a440e87198e05719d17c3d39a980a552e3","subject":"Cleanup Heartbeat Emitter abstraction code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/002005a440e87198e05719d17c3d39a980a552e3"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2024-01-16 14:12:28.000000000","tz":60},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2024-01-16 14:12:28.000000000","tz":60},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for providing catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/978049fa158f1d844edf69e4487d20f4c3e4bef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/978049fa158f1d844edf69e4487d20f4c3e4bef5"}]},"branch":"refs/heads/master"},"60a6ceed09515842366e67de86b5aef783f1dcd2":{"kind":"REWORK","_number":25,"created":"2024-01-22 14:23:42.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/94/885594/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/885594/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/885594/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/885594/25"}}},"commit":{"parents":[{"commit":"002005a440e87198e05719d17c3d39a980a552e3","subject":"Cleanup Heartbeat Emitter abstraction code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/002005a440e87198e05719d17c3d39a980a552e3"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2024-01-16 14:12:28.000000000","tz":60},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2024-01-22 14:22:13.000000000","tz":60},"subject":"Add support for catalog zones","message":"Add support for catalog zones\n\nThis implements the spec to introduce support for providing catalog zones.\n\nCo-Authored-By: Nico Keller \u003cnico.keller@inovex.de\u003e\nChange-Id: Ic9e593d661f2eb1db59a5b235521e0746489f9bb\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/60a6ceed09515842366e67de86b5aef783f1dcd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/60a6ceed09515842366e67de86b5aef783f1dcd2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
