)]}'
{"id":"openstack%2Fdesignate~887394","triplet_id":"openstack%2Fdesignate~master~Ic1d7c17fc2d7a563ff22f0b1dd42a58f87665cf1","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-07-26 16:24:53.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-07-26 17:30:10.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ic1d7c17fc2d7a563ff22f0b1dd42a58f87665cf1","subject":"Cleanup DNS Middleware","status":"MERGED","created":"2023-07-01 20:38:51.000000000","updated":"2023-07-26 19:01:12.000000000","submitted":"2023-07-26 19:00:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887394","meta_rev_id":"36ac6616b010d0c28eed6d2b8a220ab7ac29bac2","_number":887394,"virtual_id_number":887394,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-26 19:00:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-26 16:24:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-07-26 17:30:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-01 22:21:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-26 03:07:34.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"41a6dc57ccf2b676521aff4411c2164566f30dfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-01 20:38:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e390f26f768bf7ec6fb06cbf7c3807e8be1d929","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-01 20:40:02.000000000","message":"Topic cleanup removed","accounts_in_message":[],"_revision_number":1},{"id":"58876f4aa0f4be4bdaaada76a3dd05630c6253d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-01 22:21:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abd76ef60f95463db95d0b6d6f2e05bc\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7ed43d6083994030aedba943ff8a4b1d : SUCCESS in 37m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ade2ae60c36d4101a1f18b5d09ee68e0 : SUCCESS in 37m 24s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/27857b2b177d46f29c7df09bf3ce676c : SUCCESS in 55m 58s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a709b9a8f8c642a09b5e16dc8786c301 : SUCCESS in 32m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fa0965f3a31b4e0288cc25fbbda25344 : SUCCESS in 29m 38s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b2480ac8462f461abf23b08c7626db7e : SUCCESS in 1h 11m 27s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4682bb22326d4fdfa1f7d88454f2c65f : SUCCESS in 1h 25m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2eae13d2e35c45f7b63b20a2290a7e55 : SUCCESS in 31m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a347e793c66a4f4e94918b1d3a650387 : SUCCESS in 33m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ee4c3fccb504dd982cfb52fe5840318 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/779496e12df840c3bd5952166ea53731 : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e50490a2041343fc848e1f7f78eadde8 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd3c43f2c2094c71bb2d2ec9fb3ef4de : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d7c06ef7f1542b1bccd9ba17b814486 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b1646c22bd74174b05e5e20846b13d5 : SUCCESS in 6m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/910c90a6044b40f08aaefb9fa620ef1c : SUCCESS in 34m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d35a21b4251349b4a612eaa09efa7c11 : SUCCESS in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b71bee27e1ff2879a1d99473aaa3f447ce37c228","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-02 10:03:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"07920a9fc22350701de3be44b8f509b70e7dbdf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-02 10:19:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"12de3c73748cd1a4e81407dc87b9382d147eb5eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-02 11:30:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/616eb87f3cdd455c96a84aa4263b7555\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fdf7c2d7769b4d3bae298a05a7df7ad7 : SUCCESS in 38m 15s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7846ae8a8d884bad8fa125d8380e6f59 : SUCCESS in 44m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/780f475773c24a6f94f860c6398ff130 : SUCCESS in 36m 13s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8d4589e6533c47449561ec5e29209660 : SUCCESS in 39m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9015844812f54afa8f35f9c0466edc26 : SUCCESS in 31m 30s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/eded5da11ee84a09b04144e61abd183b : FAILURE in 26m 19s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9391ffd071b44c4f91817ed09d827b61 : SUCCESS in 48m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c2b41725150d4cacbf4b25fd5bcc708f : SUCCESS in 36m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d671455c52104c99b21369c44f70bc8e : SUCCESS in 36m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e293da865e384f73aac5ce700e3ccb92 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b81712a1f714c17993eea6128963b85 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e21c2786e9364975a2f85f0759deb539 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da791e5e38354769a040a533c897fc80 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ead394ad0a044f6b9ff93db2db7b4f6 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b715a79fa24df7ada3b7180b271373 : SUCCESS in 7m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c95bdda0eabf4ae288da518737d453ee : SUCCESS in 34m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3b9d8fa2547649ad949d556613445aa1 : SUCCESS in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"35534a3385c0c3f2cb91c21fdb506afe834f69b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-02 20:14:56.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"229b3779a4e96863c7669e9970a8894ca08f6b46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-02 21:05:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70c808a193164d3a890e3818e3490ec7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d9bc78240fc441439570050051e9c83e : SUCCESS in 31m 46s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/87a8c1d89a7a49ff9984b98c97cfe4bb : SUCCESS in 36m 55s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/fec4fd8695a941988f69848722e583b9 : SUCCESS in 37m 01s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b1cafd66655642d7b89041cd923342d0 : SUCCESS in 33m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c523defa11db472bb9bfb12f7a08cb45 : SUCCESS in 33m 37s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/56564b36ad9d4a43b9fe293ddbb477ac : SUCCESS in 49m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/01b691d756334cf8b7ce441738e2ab5d : SUCCESS in 43m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8d9e6f7b11094e2491e6ea68ad182806 : SUCCESS in 37m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e8091cbaf063419687aa971bbcc38e65 : SUCCESS in 31m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52252c9dd1554eca84382c85eaa43ea1 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faccd40a43f4419ebabfd9f3c74a432d : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2517a04966cd47d5ab54748d1d2f09d6 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/038ee4223dc54362bb1c2b047e153fa9 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d8b72bde68c46b89443e01a9c505ea4 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4a6d59a2f1547e89078096e12a39978 : SUCCESS in 7m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/497498dc4d16408e885a8c6dbf8afa4b : SUCCESS in 32m 25s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/72cd57d4cec242a6a79eaee68850e50a : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"47cf535e1b6eed9ad3125b11400257dce3cb7270","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-02 22:32:33.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"31dc9876c481510f24dfa1c1dae3e49ca47180ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-02 23:46:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aca6eadf364145d2b9a142984cde4ea7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6b211935995b447da5fa099bb65aa344 : SUCCESS in 33m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a93c6c67f2d949d48eafcb4b610b4ad5 : SUCCESS in 1h 06m 50s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a623b1da649c40c993600d53490dbd0b : SUCCESS in 52m 44s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8b58c036e06c4162a43a2b220055b865 : SUCCESS in 33m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/67797de6fba6408592d0768b4aa77de7 : SUCCESS in 46m 32s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0be39499588244b6aadd35337a9c8816 : SUCCESS in 1h 08m 01s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e651883816154f79bc4789531f81e820 : SUCCESS in 45m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2565c32adcc3448c9d8a370570a0b0ca : SUCCESS in 43m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c43a2336047e475bb65f4389d5299b79 : SUCCESS in 35m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07d98a49ea3d45aa8018f4a1d24bbaa6 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fe46e13874c42f08720c96b1084ae45 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f229f9f47f314c6ea2ad2f00a6b4dda7 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d80d3ac5af1341439a9e168c525bb5d8 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd3b992b6c0c4c3e87040de9de0693bb : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b95833a114d429292a4bba00b2d2dc6 : SUCCESS in 6m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b7ceb43f321744e39a5ffbc1d5e47563 : SUCCESS in 33m 44s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/30876db1b4084acd941e2763fdc7be60 : SUCCESS in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e29505d4a2f5d9dfbd76843a0f4f1e25bcbb6ccf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-12 07:22:47.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"b9a7fc3ce2c8e7f26a2d994fb41836167eaa29dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 08:37:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99f701025aa642868dc6a17cb60cd971\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6d8b1197270c4b92995946a0a8c08d9b : SUCCESS in 56m 35s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e54e727ce91349fab22733b16cd98eed : SUCCESS in 1h 09m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/555b6df8975c40b5be2a66227d3cf115 : SUCCESS in 1h 00m 57s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e5fe2241d7384db888f27f5cdf6ec7d4 : SUCCESS in 31m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/967f8dac8de8447a89f095cee7c05722 : SUCCESS in 39m 38s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/22dbd75148024d2ea59d17f56e4e811f : SUCCESS in 52m 03s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4ee463cf2b0546f49fca1e6fddd50efe : SUCCESS in 56m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/64aa5bfdd21d44faba6921679e2f0d75 : SUCCESS in 31m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/776513fddaab40f584caceccbfe78d1e : SUCCESS in 40m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bee6e23c738a47abba4fd58beed28ef8 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15cfe6d51f924717932accfcf1460edb : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfa89dda25f746f1876eadf125b3d14a : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb90bbc8ee64424d8223b00f0798cf87 : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c289a95ef7834cdd855f0dc89b10633d : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/317156ff27b64dc6a22f28b1a0d863a5 : SUCCESS in 7m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b7b91c6b5cc9471c8722f35dc7ff33c5 : SUCCESS in 31m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b01e80bef13444fe83d05136f4c32a4d : SUCCESS in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7ad1c32f219c98f2cb509ace60c77d957440ac56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-18 10:22:21.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"1f6229b134a0e272855d704bad5d632ed0d45d9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 11:51:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/013cdb6fb8084f3abefd88b414a504ca\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7e0c0a0bddd14ebfae662754275ecbd9 : SUCCESS in 45m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5f7440bf58634038ac4ed5d3fee4c13d : SUCCESS in 52m 30s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/27e41b3120b24cacb5066a6fdbab7e09 : SUCCESS in 43m 38s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/98f8abe295f14f88b28eea6241fbc35d : SUCCESS in 42m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/028ed98619484783a0a07b0b6b450dc4 : SUCCESS in 56m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/caba24ec74a246b7a26091b108b6b2f9 : SUCCESS in 1h 27m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6437b116660147a58079b4ce076794c1 : SUCCESS in 48m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2fac1986d95944fc87e563743c29ee3a : SUCCESS in 38m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b8a21ace4f2c403f8d0b7633fc096113 : SUCCESS in 37m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c833fbbf57c4d7cb589dd50691820ca : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db20fe8f1d2a40b2be82e83ffcb7c71e : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0f6caedc73e4d9da688d72032043406 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cf75050cda445f8bf6a52fb54d37f87 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d724651f0044fbeab305dbb52bc43c1 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c65758b94d3448cbd9a9ee4e1707450 : SUCCESS in 10m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ec1db9785c94a9386f40c394e4298ee : SUCCESS in 32m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cefb4ec7b2454006b77c782db9b90d63 : SUCCESS in 4m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6cf68cf971f5a837d75b45124222f623289f642d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-25 09:04:51.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"65b33c7598e29ebdf1bd81fb4d8fb247ca96b02f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 10:36:57.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/66baa893b9964f788e111d1c9464dcd8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e1f89449039c4515a207a620b3e364c4 : SUCCESS in 51m 30s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ed72625ec47c4d23b82d96b67b23a18c : SUCCESS in 1h 11m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/891f18c9b99a480e8f3d1c6402e29ae0 : FAILURE in 38m 57s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/15fc0235483949bdaae668976cc5e8c1 : SUCCESS in 51m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d103d18b48b846a18c8b213718dbd364 : SUCCESS in 53m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/489ba92b05374b28aba55c60e9ec344e : SUCCESS in 1h 23m 36s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/def4a70d612546ae83ae2c74df898f05 : SUCCESS in 1h 04m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/af6d7e30c85340fdb7e2869f340c50d6 : FAILURE in 42m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c60f0d8d098a446a9b71c2feaec6cca9 : SUCCESS in 56m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/721e540979b54ed7b36afee1c26deb81 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5488fed039574ca789745e86c36d8cce : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b17599c81d54230be8567b5e6d0c661 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0699b1980ac4a06a97885a87b1f28d0 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9dd3d53dabfc4e45b9f2a33ad0633aa6 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fff608d0ebc64ec5bd5e96356cd8ed8a : SUCCESS in 7m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3d8efeffb44e4db3bec8f66a3363acfc : SUCCESS in 33m 34s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/88c193d48178478988eac8265b437648 : SUCCESS in 6m 52s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/70a308b6a3a44bc5ab528ca7e46013ce : SUCCESS in 55m 44s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3592f1a424cd4eec937d73a1ec3b7bc3 : SUCCESS in 5m 16s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"daec3d94a0cae94e14dc07da88f00c3cc0ee9ace","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-25 11:37:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"26c6dce453da3a1cf06ca686f0fbb26a1a07a7e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 12:54:29.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a2587269b6e4203be8bbb18b9526683\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/745c91ba0ab44c3ab7c189354fb20082 : SUCCESS in 52m 09s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e5d3e417cccc41d88bbb2fa59ebbcaae : SUCCESS in 1h 03m 03s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a044eb03086c4cc3aa7a55757d1ead6e : SUCCESS in 1h 04m 43s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/373d78fa0dd04b039a17c3016261979b : FAILURE in 53m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cb1c2ace45f9489e9b81262ad004025c : SUCCESS in 50m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ca14783c195c409e84a9a3e6a4ab0c5c : SUCCESS in 1h 11m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/eb6d91332b8146abb8e9e6ce8435f6a5 : SUCCESS in 1h 12m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/99104bdd99724d6bbdbb2a4189b7dc96 : SUCCESS in 48m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fb00471fddb34c3bb5f9bab07f8242a1 : SUCCESS in 52m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbdeea51bdf5469482ff8b4c37102104 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44cc5e410b0a46b593097a6731b66f2a : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdf5c4cf63c745f9bfec6f180a9932e6 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb4ebdeb6bc041519d20e042d8762429 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7493a6ea35b14f008d7f890b3378dcfe : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aab5465a921d4146b08c2bcf9b99a906 : SUCCESS in 7m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/65453d24a21d43b4bfbbc9662ffcf83f : SUCCESS in 31m 11s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bee5ea7f2b684d7a9dde057a7d7dc2d0 : SUCCESS in 6m 25s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/44ba9bfce0104bdfaeadfb2b2c4c28b8 : SUCCESS in 44m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/32bfbeef7f56400d8621d204404e4501 : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e0405a4276d8798d202a5cedaa116d383e816589","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-26 03:07:34.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"78048a8d2b47d1b02824924f785cac966e26cc69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 05:09:32.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/39dd80e431834bc8b6744c6f9d135738\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0fc12f7c8bfe4893842ecab4e0dc06e1 : TIMED_OUT in 1h 15m 00s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/29455f047c8942c0b62b602417b0c732 : TIMED_OUT in 1h 13m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a78d210fe61455b996ccddc5518ca33 : TIMED_OUT in 1h 14m 26s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1beec6fc28eb4d66a1977d130c37046c : TIMED_OUT in 1h 13m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f7f9c4f1cda548578ddf487ab2204bb3 : TIMED_OUT in 1h 13m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b812c29756184d709469bcd2c784a5c7 : SUCCESS in 53m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dea96d78b8d14c3895f342bb0d442c54 : SUCCESS in 1h 52m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f9baf6ea842d441483f6e86c27665988 : SUCCESS in 1h 26m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0b0454d83942434995ae382a250a5991 : SUCCESS in 1h 22m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d996a15210474082a44d2b7998589f54 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bc64199abc444e4863cd17665e8a549 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/124c7d74053642d5a5d77263bfd51262 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bf9f84d9aa245b6a04623130f6f8a25 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4c5d0a5555540c3a51f98294cf96aab : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd5424b44faa4092b79ae1f275e07400 : SUCCESS in 8m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8b4825a7a17e4e6ea56c2bc6325267d3 : SUCCESS in 31m 58s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9644cf48fcd64a94b4dcd0f65a4cf798 : SUCCESS in 6m 35s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6a2b98d507e14f31b9a67c072bc0d923 : SUCCESS in 43m 07s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cd8ce22c3bc74f92ae1a03330966429c : SUCCESS in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b64a746fbc08bd923fef98e2b451b70bfff5d7eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-26 07:25:25.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"c24a4b620b7a18874d33877587b54ef31a1a1ef7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-26 07:39:01.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b5eb968f5fe4179a73f80a5cdcfc3e27601fc227","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 08:32:02.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c2a73f05c6c4c8aa5cd7816afe8c0c8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f8937ae561e64ce898251e9f8039e021 : SUCCESS in 44m 58s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0d850bc2f18449069652c4c40b8a2056 : SUCCESS in 40m 54s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c0af17b7d9346fb9588b84cbbdbe9d7 : SUCCESS in 44m 20s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b680406847324635bad7b96ecbebd3f2 : SUCCESS in 38m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/58fd58a861de47f8aa36145f0e5d643a : SUCCESS in 51m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/559b2597bc00496fbf7e36a11f54f7f2 : SUCCESS in 1h 03m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b952d86f3e5542acbf71034e68340b45 : SUCCESS in 51m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/39185f9630864d80a8b24b7081b12d71 : SUCCESS in 49m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dce068e1640b469ba1e80b54e29a7f98 : SUCCESS in 51m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48d49c7cbfeb459a85ec54e5fb470690 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/543b7bf94c034cf596c341bdde1a5ccd : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba1e8d3d941f49e697e5ccd94cf65c39 : FAILURE in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3701ed87a29249519c37080ce640d370 : SUCCESS in 6m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64c88cca3fe349118627afd364471c25 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea1be31a47b14e8db0c98903f6c4fcaa : SUCCESS in 8m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ad87ee9fc718461bb3d7e114d29af67d : SUCCESS in 38m 22s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c1a0402c2c6a47fcae0678137b4725a4 : SUCCESS in 6m 14s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f01db0fd66b641a8ad29c9a954f3ffe0 : SUCCESS in 35m 52s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4e85798f6e2f46d796d083fff736535b : SUCCESS in 6m 18s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"dc767e35b53dc8764d890233bf534bbc2307bf6c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-26 09:06:14.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"50fe0be5571e27bb4b1418b30edf1d790142b59a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 10:19:05.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/054777ea6c884f0186f2bb34a1cf02e2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e149cc1e4d6c4746b13e533b6d5f7e7f : SUCCESS in 41m 27s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1e19d0d3501b408db6135d4f7623886e : SUCCESS in 46m 07s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6dc76d1d04bf45cab97d9dce8b47a0d6 : SUCCESS in 51m 36s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2abbe0c65de14729b2f2b2a7483f970f : SUCCESS in 49m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/259a72deb268476b8517ef15cf9797cd : SUCCESS in 53m 44s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a6016f789b5b42749cf3f125b647a017 : SUCCESS in 1h 11m 03s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3cae6d8a7fdd4344907c8bdc9447893c : SUCCESS in 1h 01m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/26b347db7b7049d7a944de241d630bae : SUCCESS in 57m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0ce36c56338c4c0aaad07dc7cceb62cd : SUCCESS in 48m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71e3637ddaa243ad9910fcdc03d17bab : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/903a7db383124bf3a926f6674433b904 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c7daf6e92db4540bc9d1bfcfab56fe5 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c5a65332f2b49bd8c8d7216be904723 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d106e84997bc485ca0ff67c8babb7650 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd470c23ed414d79b8356957e29986e6 : SUCCESS in 8m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aabf01dd0fd84a8d8fff20aab4739d8e : SUCCESS in 47m 21s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/badb86dc80e14f47aaad746a61d985a4 : SUCCESS in 6m 58s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/65c5c8667280409799b25a88d8720ecf : SUCCESS in 36m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/19922bc6b69b4024b5e72335ae8c94bb : SUCCESS in 11m 45s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"672bd3344b9113a44b42dd65aea1671c54c6b706","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-26 16:24:53.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c856ade051619a791265b3a9a61583cea5a4a680","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-26 17:30:10.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"82906fae60071de23f9d111724e38f6a0f54a6b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 17:30:47.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"d9feb0028e20e3cc028d719335b197d31da84e63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 19:00:06.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2d586c7c2ed47698ea5807a71d85810\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c23aeef63a0e402582e769e992aa7720 : SUCCESS in 49m 40s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8105aae9fc154d48ab1797ac279deef2 : SUCCESS in 40m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/35743a9739494c70b7aaa40ef0063f5c : SUCCESS in 51m 04s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5773b6e49ef94583b533c5f65bed7349 : SUCCESS in 1h 23m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/09085e082ab4472eb455987aa8ff6140 : SUCCESS in 38m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4456a63c42c24abcac12026ddedc69da : SUCCESS in 49m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2832f91b429e4213b195db9709b2e249 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9a303e3573c41b8951ae0f76a263ace : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3daaa5a220f45e69aa89fd508a9484d : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e9409def1614b82a98f29cff0ea04a5 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b24e77bd1b584b86bd5f7efb66e44cb3 : SUCCESS in 9m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3941050090b74cc8b53c042424fa02ee : SUCCESS in 43m 25s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2e8a0a44a76a4cfa88dc8286e3109b51 : SUCCESS in 6m 46s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/52c02c0a2302406c83ac306c2e023888 : SUCCESS in 43m 57s","accounts_in_message":[],"_revision_number":9},{"id":"aa55352ea28b4f7371e77c6cbf8baaf227ebdc84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 19:00:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"36ac6616b010d0c28eed6d2b8a220ab7ac29bac2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 19:01:12.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aba4ecc1d3844dad84d5d3fd5b6d2558\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c7e5b1bb7f8428e8e2bb00caba44642 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"fa95018c919957ce730e825a11073e840a6af616","revisions":{"cee16b0858ae79c2203fc63364ae28fd2d6e368f":{"kind":"REWORK","_number":1,"created":"2023-07-01 20:38:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/94/887394/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/887394/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/887394/1"}}},"commit":{"parents":[{"commit":"c28b3b12f1e520460a8b6bfc10c6459c5502ea88","subject":"Merge \"Improve NAPTR support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c28b3b12f1e520460a8b6bfc10c6459c5502ea88"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-01 20:38:24.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-01 20:38:24.000000000","tz":120},"subject":"Cleanup DNS Middleware","message":"Cleanup DNS Middleware\n\nChange-Id: Ic1d7c17fc2d7a563ff22f0b1dd42a58f87665cf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cee16b0858ae79c2203fc63364ae28fd2d6e368f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cee16b0858ae79c2203fc63364ae28fd2d6e368f"}]},"branch":"refs/heads/master"},"20ab2d52afc74ff97afff48231b717b62dd881da":{"kind":"REWORK","_number":2,"created":"2023-07-02 10:03:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/94/887394/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/887394/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/887394/2"}}},"commit":{"parents":[{"commit":"c28b3b12f1e520460a8b6bfc10c6459c5502ea88","subject":"Merge \"Improve NAPTR support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c28b3b12f1e520460a8b6bfc10c6459c5502ea88"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-01 20:38:24.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-02 10:03:14.000000000","tz":120},"subject":"Cleanup DNS Middleware","message":"Cleanup DNS Middleware\n\nChange-Id: Ic1d7c17fc2d7a563ff22f0b1dd42a58f87665cf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20ab2d52afc74ff97afff48231b717b62dd881da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20ab2d52afc74ff97afff48231b717b62dd881da"}]},"branch":"refs/heads/master"},"19824479680e89bd3641d90e10d133438999e546":{"kind":"REWORK","_number":3,"created":"2023-07-02 10:19:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/94/887394/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/887394/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/887394/3"}}},"commit":{"parents":[{"commit":"c28b3b12f1e520460a8b6bfc10c6459c5502ea88","subject":"Merge \"Improve NAPTR support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c28b3b12f1e520460a8b6bfc10c6459c5502ea88"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-01 20:38:24.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-02 10:19:43.000000000","tz":120},"subject":"Cleanup DNS Middleware","message":"Cleanup DNS Middleware\n\nChange-Id: Ic1d7c17fc2d7a563ff22f0b1dd42a58f87665cf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19824479680e89bd3641d90e10d133438999e546"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19824479680e89bd3641d90e10d133438999e546"}]},"branch":"refs/heads/master"},"cb8740ba0416d6c07a7108eff6a1f904b66e8ac8":{"kind":"REWORK","_number":4,"created":"2023-07-02 20:14:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/94/887394/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/887394/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/887394/4"}}},"commit":{"parents":[{"commit":"c28b3b12f1e520460a8b6bfc10c6459c5502ea88","subject":"Merge \"Improve NAPTR support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c28b3b12f1e520460a8b6bfc10c6459c5502ea88"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-01 20:38:24.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-02 20:14:21.000000000","tz":120},"subject":"Cleanup DNS Middleware","message":"Cleanup DNS Middleware\n\n- Moved DNS Middleware to separate file.\n- Cleaned up logic.\n- Fixed non-lazy loaded logging.\n\nChange-Id: Ic1d7c17fc2d7a563ff22f0b1dd42a58f87665cf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cb8740ba0416d6c07a7108eff6a1f904b66e8ac8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cb8740ba0416d6c07a7108eff6a1f904b66e8ac8"}]},"branch":"refs/heads/master"},"79132f384d68df318dc17404a6fde3b23fc268a1":{"kind":"REWORK","_number":5,"created":"2023-07-02 22:32:33.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/94/887394/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/887394/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/887394/5"}}},"commit":{"parents":[{"commit":"c28b3b12f1e520460a8b6bfc10c6459c5502ea88","subject":"Merge \"Improve NAPTR support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c28b3b12f1e520460a8b6bfc10c6459c5502ea88"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-01 20:38:24.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-02 22:32:22.000000000","tz":120},"subject":"Cleanup DNS Middleware","message":"Cleanup DNS Middleware\n\n- Moved DNS Middleware to separate file.\n- Cleaned up logic.\n- Fixed non-lazy loaded logging.\n\nChange-Id: Ic1d7c17fc2d7a563ff22f0b1dd42a58f87665cf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79132f384d68df318dc17404a6fde3b23fc268a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79132f384d68df318dc17404a6fde3b23fc268a1"}]},"branch":"refs/heads/master"},"e40f30730ff52bf2c2a6ecc0e4ca23a5db5ad994":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-07-12 07:22:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/94/887394/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/887394/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/887394/6"}}},"commit":{"parents":[{"commit":"963751f2b403c3e9a73008ac62854ed3d3758247","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/963751f2b403c3e9a73008ac62854ed3d3758247"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-01 20:38:24.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-12 07:22:47.000000000","tz":0},"subject":"Cleanup DNS Middleware","message":"Cleanup DNS Middleware\n\n- Moved DNS Middleware to separate file.\n- Cleaned up logic.\n- Fixed non-lazy loaded logging.\n\nChange-Id: Ic1d7c17fc2d7a563ff22f0b1dd42a58f87665cf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e40f30730ff52bf2c2a6ecc0e4ca23a5db5ad994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e40f30730ff52bf2c2a6ecc0e4ca23a5db5ad994"}]},"branch":"refs/heads/master","description":"Rebase"},"0a20500888b14ea9befd7be3d685c75511b23fba":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-07-18 10:22:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/94/887394/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/887394/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/887394/7"}}},"commit":{"parents":[{"commit":"7daa2578a51cad61fed8e645d9e5158d5c5f5c17","subject":"Fix a sqlalchemy 2.x compatibility issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7daa2578a51cad61fed8e645d9e5158d5c5f5c17"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-01 20:38:24.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-18 10:22:21.000000000","tz":0},"subject":"Cleanup DNS Middleware","message":"Cleanup DNS Middleware\n\n- Moved DNS Middleware to separate file.\n- Cleaned up logic.\n- Fixed non-lazy loaded logging.\n\nChange-Id: Ic1d7c17fc2d7a563ff22f0b1dd42a58f87665cf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a20500888b14ea9befd7be3d685c75511b23fba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a20500888b14ea9befd7be3d685c75511b23fba"}]},"branch":"refs/heads/master","description":"Rebase"},"586577cab0d4c64d0e19cc0ae24835235c305227":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-07-25 09:04:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/94/887394/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/887394/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/887394/8"}}},"commit":{"parents":[{"commit":"8cc87bce039f1379403e83edf68a2910e44dc463","subject":"Merge \"Removed unused code path in sql base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8cc87bce039f1379403e83edf68a2910e44dc463"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-01 20:38:24.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-25 09:04:51.000000000","tz":0},"subject":"Cleanup DNS Middleware","message":"Cleanup DNS Middleware\n\n- Moved DNS Middleware to separate file.\n- Cleaned up logic.\n- Fixed non-lazy loaded logging.\n\nChange-Id: Ic1d7c17fc2d7a563ff22f0b1dd42a58f87665cf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/586577cab0d4c64d0e19cc0ae24835235c305227"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/586577cab0d4c64d0e19cc0ae24835235c305227"}]},"branch":"refs/heads/master","description":"Rebase"},"fa95018c919957ce730e825a11073e840a6af616":{"kind":"REWORK","_number":9,"created":"2023-07-26 07:25:25.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/94/887394/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/887394/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/887394/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/887394/9"}}},"commit":{"parents":[{"commit":"8cc87bce039f1379403e83edf68a2910e44dc463","subject":"Merge \"Removed unused code path in sql base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8cc87bce039f1379403e83edf68a2910e44dc463"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-01 20:38:24.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-26 07:25:16.000000000","tz":120},"subject":"Cleanup DNS Middleware","message":"Cleanup DNS Middleware\n\n- Moved DNS Middleware to separate file.\n- Cleaned up logic.\n- Fixed non-lazy loaded logging.\n\nChange-Id: Ic1d7c17fc2d7a563ff22f0b1dd42a58f87665cf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa95018c919957ce730e825a11073e840a6af616"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa95018c919957ce730e825a11073e840a6af616"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
