)]}'
{"id":"openstack%2Fdesignate~887425","triplet_id":"openstack%2Fdesignate~master~I8af9b5cf8c1473bbf7db71a1fb848fb64509db84","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-07-05 15:16:55.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I8af9b5cf8c1473bbf7db71a1fb848fb64509db84","subject":"Fix list zones if shared with multiple projects","status":"MERGED","created":"2023-06-30 18:53:42.000000000","updated":"2023-07-05 16:34:01.000000000","submitted":"2023-07-05 16:32:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887425","meta_rev_id":"712c6f647b530783aea7795dced294ce4aeaf7b9","_number":887425,"virtual_id_number":887425,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-05 16:32:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-07-05 15:16:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-01 07:50:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-07-05 15:16:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-30 20:35:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-01 07:50:22.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"ec74e33530fef325e5d3180ac56614cd0ea97137","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-06-30 18:53:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0c53f28b3bb2a0ca58d7df6f48389ba25e457b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 20:35:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d96341cbf7ed47448b97c2f9de5f613a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6a58902bb3a442d3b6bd6b3883f86d28 : SUCCESS in 49m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/dbd51e1063d044d186f1aa2cf78f9f45 : TIMED_OUT in 1h 14m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d6fa1b223224ee6afc4adc05657d6df : SUCCESS in 1h 03m 34s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/55a4a3121f904fbca059a277df1b4d31 : SUCCESS in 55m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a93df159e30440e8bc35dc8ba500b708 : SUCCESS in 1h 00m 05s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a5e67d9ac770457bbf1dbcccf8992742 : SUCCESS in 51m 46s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/838815a824dd4d708f7c1b00b331d3b0 : SUCCESS in 1h 36m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a39269782dfe4773a01c5124382e8c25 : SUCCESS in 44m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f99ed2b8280a4363ac667e17ca0fe2fa : SUCCESS in 47m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6735c2533a454f2fab8b07dbcac94f4f : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1ef04188fe14b45964c29c357ad23e5 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6efc9f0aac8d420f9fca902492eef2cf : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b95bec54dee4ff2ac9708d4c5254f20 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/146f244eb9d84386ae849befd8e69a33 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8c568f65fb74e3e92e453a07fcf7509 : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ffb8f12d0344fb292b1914a85bc3740 : SUCCESS in 4m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6ef28d6578df4320a9a510e8b014a5dd : SUCCESS in 42m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ed9fbe54e62342178ac2b8441f539a69 : SUCCESS in 3m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e548144384c27a151dcf27023f376eb1133c63b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-01 07:50:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aa28d8781854bebe89ce8a25a11743da32f7382e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-05 15:16:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca253dad8f72a84e9bbfbc0f98b9648fbdc1b38c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 15:17:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5ca7a05c9a77b111ff94b65a98c71bb411a0f6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 16:32:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43c4e4f3c4554445aaf034db1a720ce5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d623fb943f164ecbbcb9ebea590ecf8c : SUCCESS in 50m 20s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8dada5c128a343b6a984d5cc33bb6f29 : SUCCESS in 52m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4a29373fa8e34ce48363fb209ec90b0f : SUCCESS in 53m 43s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b36d41755fe5461d9667b6a4a4629ff1 : SUCCESS in 1h 08m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b8282c4c6760459a884e5d25e95420e7 : SUCCESS in 54m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1ed451a3c66943179cb738a083d78f61 : SUCCESS in 45m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3019cdddfae4e1c9a6469e135976368 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52a749d658894c47a6df17d8764ebc28 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b917f6f8959949c79dc2178398fbb116 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc92f4643d1540f387c7e822e3b780fd : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fce8e4d62335488696e56daf2d12cf02 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22974b2923684e9e94070d085232698f : SUCCESS in 4m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5fe0c61f397244af9d6cffb85349f4ca : SUCCESS in 34m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7ed3d48e411ec3862bab26f8b0a358f7a499a08c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 16:32:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"712c6f647b530783aea7795dced294ce4aeaf7b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 16:34:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1a3b90f66fd4efe9a30b1a7c60e886d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a61f8e2f5c24f9d98b593070af161c1 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57f1783225f04cca8becdbf76b4f0763 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"011ebe2e7cd0df0c7f0869f0c7abbce79434821a","revisions":{"011ebe2e7cd0df0c7f0869f0c7abbce79434821a":{"kind":"REWORK","_number":1,"created":"2023-06-30 18:53:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/25/887425/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/887425/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/887425/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/887425/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/887425/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/887425/1"}}},"commit":{"parents":[{"commit":"3621b9d1b1123f9004ebf07b026ed00ca53a9b76","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3621b9d1b1123f9004ebf07b026ed00ca53a9b76"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-06-30 18:51:17.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-06-30 18:53:40.000000000","tz":0},"subject":"Fix list zones if shared with multiple projects","message":"Fix list zones if shared with multiple projects\n\nThis patch fixes a bug when listing zones or updating recordsets in\nzones that are shared with more than one project.\n\nCloses-Bug: #2025295\nChange-Id: I8af9b5cf8c1473bbf7db71a1fb848fb64509db84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/011ebe2e7cd0df0c7f0869f0c7abbce79434821a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/011ebe2e7cd0df0c7f0869f0c7abbce79434821a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
