)]}'
{"id":"openstack%2Fdesignate~888138","triplet_id":"openstack%2Fdesignate~stable%2F2023.1~Ie9e86199ce9d6b3746637156a47b46b686e12437","project":"openstack/designate","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-07-12 08:28:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie9e86199ce9d6b3746637156a47b46b686e12437","subject":"Remove designate-tox-dnspython-latest from stable","status":"MERGED","created":"2023-07-11 16:03:19.000000000","updated":"2023-07-12 08:31:41.000000000","submitted":"2023-07-12 08:28:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888138","meta_rev_id":"37d5c703a33057570f869010723731ee312208d0","_number":888138,"virtual_id_number":888138,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-12 08:28:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-07-11 16:40:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-12 07:21:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-12 07:21:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-07-11 16:40:24.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-11 16:40:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-07-11 17:00:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-12 07:21:25.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"4f1fb9c8c2395d3f652495262ba0ce778401ea11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-11 16:03:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f94a23c5819241115a692abe9f7e8c3ea765b2d6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-07-11 16:40:24.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"9356db4b0e3da4d77ced02bd84b2c26f9b759e8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 17:00:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8fb27c026884e0fad2218803e6cc96c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c1d721cbbaf9434180b5cb795fd29f77 : SUCCESS in 41m 34s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3f5639245e744883b2b00014d53fb5ed : FAILURE in 11m 55s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/941e6cf096f149a59e29647465b0697b : SUCCESS in 38m 43s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/acb11b137bed48aaa15187bca00b79a8 : FAILURE in 38m 12s (non-voting)\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/75d69b46203b4b8ca98bd73520d774a7 : SUCCESS in 41m 35s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0f98a34cab2d4888a53731aa673e056a : SUCCESS in 51m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d254bf97fa3d4344b3df8f9b3cb9534b : SUCCESS in 56m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b69fd8a014e74649b588c5219e91bfe4 : SUCCESS in 46m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a53a3846eb37455b8a51fde7db6eecad : SUCCESS in 44m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8105235b454c43359c39ba1b279566a5 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d935d7234054e4092e88e8f735883fc : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cee20b97027a4fa7822e6a4405b26496 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a3837f36a2c40e9b5d9ec1f75f62ba3 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b57f86114f164cc19f13cb41f2a33fb7 : SUCCESS in 7m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6c70da61464f43fcb5f078620acbf294 : SUCCESS in 35m 34s","accounts_in_message":[],"_revision_number":1},{"id":"dd3c6cd85de317e3b6d401795a2b18d1faccc2c2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-12 07:21:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ed81768dc171e735d3c41a84aac8841a1555e6cd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-12 07:21:47.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e492e9c176123f7301712738a56dcce29f1a8008","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 07:22:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cd90346cb0f740bf93d91ec89ec4ee27beb40bc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 08:28:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fbd3224217548d48b45312df8e4752d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2b4e7895705d41b082467aaea579f459 : SUCCESS in 47m 28s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/303ed4d67c354ddfb9bc09977515afe5 : FAILURE in 49m 32s (non-voting)\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/255eb767a7a849b8a0d1046bb8a9014b : SUCCESS in 35m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/80f5c155e7ed4ccf987cc345acd713eb : SUCCESS in 1h 01m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/027aa53f256e4e4986f50386806b56b3 : SUCCESS in 31m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/eae8af702e6d45619c368e3bbc49c404 : SUCCESS in 54m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74935950283645629a9b326b4c0fb077 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ae9200be9454514ac6e9b16867a32d8 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7052308c3fa14be88f50221d608e794f : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb95a631011545f498e67951e3a4dae0 : SUCCESS in 7m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/22329dcd66f3482a9741263fed5ac79b : SUCCESS in 31m 20s","accounts_in_message":[],"_revision_number":1},{"id":"03bd6119b767634922a06caebc5e128289f1a5a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 08:28:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"37d5c703a33057570f869010723731ee312208d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 08:31:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a947ffb977a94f8089205f2dfe55fa05\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ea341dd3a7347fb80a43c75bf96ae81 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b570518c379a5b3dd0b3116a4ce9555219c6b22","revisions":{"1b570518c379a5b3dd0b3116a4ce9555219c6b22":{"kind":"REWORK","_number":1,"created":"2023-07-11 16:03:19.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/38/888138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/888138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/888138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/888138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/888138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/888138/1"}}},"commit":{"parents":[{"commit":"d3b6f0c3e914aabe34ab76df0fe4473e832c8a89","subject":"Merge \"Fix list zones if shared with multiple projects\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d3b6f0c3e914aabe34ab76df0fe4473e832c8a89"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-11 16:01:15.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-11 16:03:18.000000000","tz":0},"subject":"Remove designate-tox-dnspython-latest from stable","message":"Remove designate-tox-dnspython-latest from stable\n\nThis patch removes the non-voting designate-tox-dnspython-latest from the\nAntelope branch of designate. It is really just useful for the master branch\nas the stable branches have upper-constraints restrictions.\n\nChange-Id: Ie9e86199ce9d6b3746637156a47b46b686e12437\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b570518c379a5b3dd0b3116a4ce9555219c6b22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b570518c379a5b3dd0b3116a4ce9555219c6b22"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
