)]}'
{"id":"openstack%2Fdesignate~888156","triplet_id":"openstack%2Fdesignate~stable%2Fzed~Ie9e86199ce9d6b3746637156a47b46b686e12437","project":"openstack/designate","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-07-12 08:28:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie9e86199ce9d6b3746637156a47b46b686e12437","subject":"Remove designate-tox-dnspython-latest from stable","status":"MERGED","created":"2023-07-11 16:04:01.000000000","updated":"2023-07-12 08:32:26.000000000","submitted":"2023-07-12 08:28:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888156","cherry_pick_of_change":888138,"cherry_pick_of_patch_set":1,"meta_rev_id":"a80f423f8f3dc0e09e2308421f8420ca2a61a1c4","_number":888156,"virtual_id_number":888156,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-12 08:28:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-07-11 16:42:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-12 07:21:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-12 07:21:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-07-11 16:42:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-11 16:42:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-07-11 17:07:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-12 07:21:39.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"5d37a7dca5a1cc81f54dcd9bb000d41de7d5749c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-11 16:04:01.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"220a6ab322b4c074e951158328eab9535d6ceb47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-11 16:04:25.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"adcbcc88efd92b59e37c858a52c097e8063619bd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-07-11 16:42:18.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"d950e8b4b325347fb974e6159b641f14c2d47bd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 17:07:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b57c44978a546b4addf19927f47c9f9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0a89d993d51a4fe183c12e9c65a8c657 : SUCCESS in 43m 59s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/26af94915e6e4513b3ff3b2d8adcc310 : FAILURE in 12m 03s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/117c4575525149acb4bebbb09753bbd5 : SUCCESS in 47m 25s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/81ba78ec5e3a4b268545f3f870fefeca : SUCCESS in 37m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6f5b2fea005c49f7b4dca4f8828d8030 : SUCCESS in 40m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2ccb2a9a046d4813acdab28d5bf5780a : SUCCESS in 57m 36s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8de60f3d25784ddc8a138dc4be2de02b : SUCCESS in 1h 01m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9e1cfdf247e946208e6726c34ad3dbaa : SUCCESS in 50m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ec149c34704a46deba74fa20be0a90aa : SUCCESS in 55m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a67ae93baf64a30ac81f57718584218 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35775cae156f46dca011c59cb934ff21 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c328ac223114fdfba937b760b31384b : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/303b59549051473b99924d695f67c923 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfbed27504fd4ff5be03c3e387f68fc2 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46712ca2d3b442969816541690920d02 : SUCCESS in 7m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fd1839f7594649deabd8deb57a9b0f2d : SUCCESS in 34m 24s","accounts_in_message":[],"_revision_number":2},{"id":"0f70d0bd6ece9614e4ba95edea049bc8597fe289","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-12 07:21:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b043ecd0d752a875179d9a70007059e050966feb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-12 07:21:49.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"da7fe360cbfd71216bc68cf2c21360037d71d2c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 07:22:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1832f296f353cfca2a5fa06474ea8228ab2ff375","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 08:28:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcb7a51725b9427fa33562c35b5f0c2d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e1a7b790ca5c42e38c3365a6d103d12c : SUCCESS in 32m 13s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d62638a7835a4c0699320a255e078be8 : SUCCESS in 36m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/424520932372499ea4c2de33513aba19 : SUCCESS in 33m 04s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f07c6bba0f5748b5a8d609c5514982a4 : SUCCESS in 52m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/86a789011fbb47d1b44be0d1737e1fbc : SUCCESS in 35m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6820584c37764f80ac427b0ab0d16196 : SUCCESS in 51m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7f9c75d89a04f8991bcf0659432fa05 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f940c5942184c3f9993595a2c65ac8d : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb9af6e6605644a6b4a63b5170a645dd : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d58d99f325947c29b093771dee9f952 : SUCCESS in 6m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e779858e7d114628a00750e21bedd697 : SUCCESS in 43m 47s","accounts_in_message":[],"_revision_number":2},{"id":"a42fb9bd52bed3cda8354acef2cd3720ea23cad5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 08:28:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a80f423f8f3dc0e09e2308421f8420ca2a61a1c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 08:32:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d74132c22cf143a3970ca71fc5e6d2d1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeca7fe2d0d343a2aab09ef84d3b1596 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2537aedfb5ca68aae5fba210fe6d3d9e3767eb90","revisions":{"5abc680d8563605e6ae20efe63f9eaf0950e4388":{"kind":"REWORK","_number":1,"created":"2023-07-11 16:04:01.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/888156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/56/888156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/888156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/888156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/888156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/56/888156/1"}}},"commit":{"parents":[{"commit":"a99367cff117268ec265a12d1f46fdf0376b40cc","subject":"Fix TsigKeyring issues with dnspython 2.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a99367cff117268ec265a12d1f46fdf0376b40cc"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-11 16:01:15.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-11 16:04:01.000000000","tz":0},"subject":"Remove designate-tox-dnspython-latest from stable","message":"Remove designate-tox-dnspython-latest from stable\n\nThis patch removes the non-voting designate-tox-dnspython-latest from the\nAntelope branch of designate. It is really just useful for the master branch\nas the stable branches have upper-constraints restrictions.\n\nChange-Id: Ie9e86199ce9d6b3746637156a47b46b686e12437\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5abc680d8563605e6ae20efe63f9eaf0950e4388"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5abc680d8563605e6ae20efe63f9eaf0950e4388"}]},"branch":"refs/heads/stable/zed"},"2537aedfb5ca68aae5fba210fe6d3d9e3767eb90":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-07-11 16:04:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/888156/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/56/888156/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/888156/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/888156/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/56/888156/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/56/888156/2"}}},"commit":{"parents":[{"commit":"a99367cff117268ec265a12d1f46fdf0376b40cc","subject":"Fix TsigKeyring issues with dnspython 2.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a99367cff117268ec265a12d1f46fdf0376b40cc"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-11 16:01:15.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-11 16:04:25.000000000","tz":0},"subject":"Remove designate-tox-dnspython-latest from stable","message":"Remove designate-tox-dnspython-latest from stable\n\nThis patch removes the non-voting designate-tox-dnspython-latest from the\nZed branch of designate. It is really just useful for the master branch\nas the stable branches have upper-constraints restrictions.\n\nChange-Id: Ie9e86199ce9d6b3746637156a47b46b686e12437\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2537aedfb5ca68aae5fba210fe6d3d9e3767eb90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2537aedfb5ca68aae5fba210fe6d3d9e3767eb90"}]},"branch":"refs/heads/stable/zed","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
